)]}'
{"id":"openstack%2Fkolla~796279","triplet_id":"openstack%2Fkolla~master~I4ce8244a3fb034493a75f4096de2640e11b48931","project":"openstack/kolla","branch":"master","topic":"bug/1931867","hashtags":[],"change_id":"I4ce8244a3fb034493a75f4096de2640e11b48931","subject":"Fix missing templating block for kolla-toolbox.","status":"MERGED","created":"2021-06-14 13:46:51.000000000","updated":"2021-07-26 10:38:30.000000000","submitted":"2021-07-26 10:36:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"796279-bug/1931867","meta_rev_id":"2f0895e90032965513165f5594c54424a2bc3bc2","_number":796279,"virtual_id_number":796279,"owner":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-26 10:36:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-07-26 09:24:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-07-26 09:14:18.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-07-26 09:24:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-14 13:50:42.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2021-06-14 15:55:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-15 08:40:32.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-06-17 06:31:07.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"bb0344d09dbd10cb3cb24dc559feb26b85bde17c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"date":"2021-06-14 13:46:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6d2a44d8c5ba0e4ea92cabac4889ce12a9e96cdb","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-06-14 13:50:42.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)\n\nWould it be perhaps better to let the user customise the URL instead of forcing them to customise the whole block?","accounts_in_message":[],"_revision_number":1},{"id":"dc3f76b0dc6b334ab21f29c4afb5cf84e2fae092","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"date":"2021-06-14 14:32:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6384ff48497274682c0e63341e59d32e0f4726a5","author":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"date":"2021-06-14 14:35:19.000000000","message":"Patch Set 2:\n\n(3 comments)\n\n\u003e Patch Set 1: Code-Review-1\n\u003e \n\u003e (3 comments)\n\u003e \n\u003e Would it be perhaps better to let the user customise the URL instead of forcing them to customise the whole block?\n\nYes, it was my first thought but I wanted to do it the \"standard\" way first and then create a wider patch set with all roles that I encountered covered.\n\nDo you prefer we goes with the extend patch set way?","accounts_in_message":[],"_revision_number":2},{"id":"7d3334e3b23699f20e8bc0561a6f45e4a9b47d4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 15:55:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/a06f4d4137df4a56b629afb1e6710e29 : SUCCESS in 28m 14s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/5cf5f2abcfab4d70890adc53bb2f0ae8 : SUCCESS in 48m 29s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/7054af0d03224dd88ae977d1a149fc4b : SUCCESS in 1h 05m 00s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/d7728eec282a4654a5dd6c5214f6d25c : SUCCESS in 1h 17m 42s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/c294d028cf334ffd8f41163294b74ba1 : SUCCESS in 1h 11m 42s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/7fbf50cfa93e4a9f923f556b9351b7be : SUCCESS in 38m 56s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/978dbb0613704fcaa37e022ed7e6be32 : SUCCESS in 18m 38s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/cacdff3cd86f4efd9b9596ccea018d76 : SUCCESS in 56m 18s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/012c6c34539e4d50b179ea4b62908669 : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/26e4ad8993e44265994a8ea2b1dc0920 : SUCCESS in 6m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/354c2724e01e418a8f53f8594b995352 : SUCCESS in 5m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/09e12bbb086243d2ab7f913253cb1416 : SUCCESS in 6m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a9db46fcb3241bc81b8fedb209fb500 : SUCCESS in 13m 15s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f15503a5b8b94008a9f6fd6ea88723fe : SUCCESS in 22m 30s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0b70b78ac9c24a7dbd487980c0a8dc76 : SUCCESS in 46m 39s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/24c4d45ec57343c29dafea7f3717d839 : SUCCESS in 51m 55s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ed1d7f2a1e79429781b19e1b28ed0881 : SUCCESS in 44m 57s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c7be2ff4fa1a4b0e999a040c9ed6d8d2 : SUCCESS in 1h 02m 54s","accounts_in_message":[],"_revision_number":2},{"id":"6644519d53ecf812a3b8b1f303316e628ac7e90e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 16:19:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/e4fe2d9838744bb28d754ce93d5d4d67 : SUCCESS in 1h 15m 31s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f145632267184eb5aff0ffd833e2eaf1 : SUCCESS in 1h 06m 05s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a26c060b496e41f6804fa864c099f3a2 : SUCCESS in 1h 40m 10s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/645b4e1c4c094a7b91b2b1f2459fe744 : SUCCESS in 5m 41s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/06a99064b40543488a4a664978985031 : SUCCESS in 7m 11s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/3a10c9bed5c34abdaec280d4bc63e1e9 : SUCCESS in 1h 12m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c20ec053e3cb517caa79d349d09d34237506ca07","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-15 08:40:32.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"6f6f77220182fc659446823d4e012514a91eb5a6","author":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"date":"2021-06-15 12:03:57.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"3adc1a01e39b1b8bde853e9d03c40b9a94e7125e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-15 13:00:36.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"48277d7b86427919c9ef2b931e8c7141cf1054ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"date":"2021-06-15 13:15:14.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"db41693691d46a3acd2b90db0bfb56979c6ca8e9","author":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"date":"2021-06-15 13:17:42.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"15a41b89fd45d5abd672466a26ebbca8de6b8834","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 14:48:11.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/be736d489c6846a1b22aa63ee01a4a34 : SUCCESS in 1h 02m 20s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/7a1bdc35142343508a71466aca813bf0 : SUCCESS in 1h 16m 34s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/f95a3049ac124fa68430be0f3c3436aa : SUCCESS in 1h 15m 04s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/260d4c4c0d6d471e9deaa4dcb4ceca3c : SUCCESS in 1h 19m 41s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/f60688dce3fc4612ad6ca9af1a76d3d5 : SUCCESS in 1h 31m 41s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/30c4292dcdf44d54bb380105e3adc9ba : SUCCESS in 40m 17s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/b6b3f226c7e444a8a6d30a391fbbf539 : SUCCESS in 26m 48s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/0f36ab0f0e554af795ffaefccebcdae1 : SUCCESS in 1h 06m 33s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba84c8ba7453432082e8f316d75e39f2 : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ea521f6e34604615be99efeab72c9e53 : SUCCESS in 6m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ed1eb738783f412a8614b8bab5c54f43 : SUCCESS in 4m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f2c2bd54bf7f4462a0d2886a6a062b0f : SUCCESS in 5m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/569a122154754a188ed72b4a9f16c7a6 : SUCCESS in 8m 03s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e6bbaf3d07ae4e52b873d4f7f7ccdde6 : SUCCESS in 28m 16s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ccaa865ebf83475997f1378f86d134b6 : SUCCESS in 39m 07s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/498eefa35d254994ac04447b7b62b993 : SUCCESS in 52m 54s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/28d36132bc884db5a7beb95eaabc1bcf : SUCCESS in 47m 46s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b70ed61f143d4e199bb63d6f322ba8ac : SUCCESS in 1h 06m 39s","accounts_in_message":[],"_revision_number":3},{"id":"102aa1c0f102b4d745de690553657fdc069befdb","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-15 16:00:38.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"e836d03ac3394e4da6d6fbcef3cdf8b9e8152489","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 17:22:06.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/f6ef2127f61946d6b02a781b59acf650 : SUCCESS in 1h 23m 23s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0c26c6a8f1224288a8d71c33a98a3492 : SUCCESS in 1h 10m 30s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/29536289cda449e9aecb1eba634c993e : SUCCESS in 1h 32m 44s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a616ff251ef047b0b00e15f21c503311 : SUCCESS in 6m 25s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cf16a0c290e3419bb0348f1d751cf4ad : SUCCESS in 7m 12s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/a0224810f84d4129b79365e4f1039a95 : SUCCESS in 1h 00m 59s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0c05d135e5ad3fbb2e7ced2999c642e8933e42f3","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-06-17 06:31:07.000000000","message":"Patch Set 3: Code-Review-1\n\nmissing reno","accounts_in_message":[],"_revision_number":3},{"id":"a12e908070eb2a8f5914a50d7ef5fca79a0f2e1b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"date":"2021-07-23 09:46:50.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5ab89739639405a12706b8796d5fdd1859c8259b","author":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"date":"2021-07-23 09:47:18.000000000","message":"Patch Set 4:\n\n\u003e Patch Set 3: Code-Review-1\n\u003e \n\u003e missing reno\n\nThanks, I\u0027ve added it.","accounts_in_message":[],"_revision_number":4},{"id":"12cc8d3370d809d1c15614472bf8e16d2749f7b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"date":"2021-07-23 09:48:49.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"e7fd3af117f3eb4b94f894938f8c6d7ab1ff3964","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-23 10:54:05.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/56160346d7524429a9fa2e60bb393b7d : SUCCESS in 24m 00s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/d2c12d3a92c5483e8c2b8e6d3721e6bc : SUCCESS in 26m 10s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/e7985e681a63479a84d8e4df8f7b426d : SUCCESS in 1h 00m 10s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/d7789e2e85934b76968c89028e5b8b3a : SUCCESS in 1h 04m 20s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/ce4d6f1bb000425fa7862b7359d45896 : SUCCESS in 1h 00m 26s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/03a36e91e13042a3b1eaef7ceef69e28 : SUCCESS in 27m 53s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/6d934195f17a47ca916ea62cbc886eb6 : SUCCESS in 17m 02s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ce99bd8ed6604aa2b3f4bf573f43527c : SUCCESS in 51m 32s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f50b8437b77b4c80be129c0a0820c2e3 : FAILURE in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f8274da79ed14f8e83990246463b6ad5 : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0455ee404b4441c6b9d7c4dc063baf9e : SUCCESS in 4m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f2acc1a028c24dee8821b21f0e103727 : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65319f507a9245beae406037c32b43b0 : SUCCESS in 6m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1898a854f8fc42adb8e08f28b015d5ca : SUCCESS in 4m 06s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b27e0f669abb4fda9f9f031b4939dade : FAILURE in 19m 24s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/dabd3467dd5b4feab0d3ab10d55a6892 : FAILURE in 32m 42s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/eac561dfb138416b9ebe464ece9e43a3 : SUCCESS in 54m 58s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c312121868024ff6a6fac8ec878eb51d : SUCCESS in 48m 00s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/496855cdfebb4feaa5e1b1c4e2e318c9 : SUCCESS in 1h 03m 04s","accounts_in_message":[],"_revision_number":5},{"id":"9ebf102ae44b68a07dd91397f349c9f2dc5d1913","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-23 11:25:10.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/df14b922f8eb4c788d1635f796b358f8 : SUCCESS in 51m 14s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4bdc30cf85b3422da4dc393d15931741 : SUCCESS in 46m 47s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/09d217d653d14dae867823a05dd99bd5 : SUCCESS in 1h 24m 45s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/359dc2d133e24fc19f37594d62720946 : SUCCESS in 5m 14s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/78c9d5eb67934623b8479fc4d07a32f6 : SUCCESS in 6m 12s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/f0c8781b98374a418176110f91bb495f : SUCCESS in 1h 23m 50s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"70e00510c2b5e2a914e5cec9167bdfb248906ffa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"date":"2021-07-23 14:05:13.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"9fc35b8045d0a1e9475639188c4111a149a942cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-23 15:23:56.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/71bb7ebc50ca46c08471b781769b5c1f : SUCCESS in 23m 02s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/9db70d9ddf6d4abb897489fef4957f4f : SUCCESS in 36m 07s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/020aaf8b9c0a42f492614e78117d6ffa : SUCCESS in 1h 08m 31s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/3bc955437557485a8a531d59b882017d : SUCCESS in 1h 16m 31s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/4da0f70b8b6e4e9098eb5301321fa440 : SUCCESS in 1h 16m 07s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/c12e9f9df5bb4496ae97423aede95c4c : SUCCESS in 30m 31s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/c586f6c40c76438181d7376c87faa487 : SUCCESS in 21m 43s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/de34f0efef36415f8b831a0aa0917a29 : SUCCESS in 54m 27s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0d34618b6004a45bf9d6bbac91b210f : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7e599eefc578463b81fd7cc5361abb17 : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/62f3402fac4d4535b079f560033b7dbb : SUCCESS in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/965035be3b4d488792d9ab55d35d36d7 : SUCCESS in 4m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6971d2fd1ffb4e35a4e6ae5abd2eae5c : SUCCESS in 7m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5534bd1a65b04647acd1603560505f16 : SUCCESS in 4m 47s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/daa7200625374c4380cc85b43b738b69 : SUCCESS in 20m 45s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6653f0b641a24a5fb4c322fd3bd619fc : SUCCESS in 33m 08s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/076ea757bf0541da9b0a7d2ada68e416 : SUCCESS in 49m 15s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/948e03e4424b4790bc4fbcbc8f80d605 : SUCCESS in 52m 29s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6ab7b8a58e33450494dc7ebb50463123 : SUCCESS in 56m 42s","accounts_in_message":[],"_revision_number":6},{"id":"ce2cbc7928a66c9c3ca0dbb355c81b1a864b4272","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-23 15:55:37.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/1c3ad84f2485456690c10032d547ac39 : SUCCESS in 1h 40m 23s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3f5ec3d920b349d9bebe88d97fcd082c : SUCCESS in 51m 20s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/dc2b97525bfe4e32a173de5c35ea9af9 : SUCCESS in 1h 38m 53s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0aacb60ff74f426d92e325e894ffb2e3 : SUCCESS in 4m 47s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/51d191956dba42358d6c96dd59a0ef0a : SUCCESS in 5m 37s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/112ece132c6a47e6af2aebca1c0811c9 : SUCCESS in 55m 58s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"e798615592683db97f0820a833dd87d2f2d3d29b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-07-26 09:14:18.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"073cbb5ff94a8506a7863c43f0d4f323366338e2","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-07-26 09:24:59.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"b95e178109cd7067464035543457e6b7220d78e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-26 09:25:18.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"81d3ec15e1b4fbc9c3398c64b7b5199f852a64c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-26 10:36:57.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/7fbaf2d1a4d943eba9d9ea1be61cacbd : SUCCESS in 36m 12s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/21824bb288ff4cd9be0b486934244c5d : SUCCESS in 56m 52s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/db939925a4844f5b9d47a6486497ef67 : SUCCESS in 1h 07m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b266f5f043284335bf90f8553b24df24 : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/16c710c370454ab094acb0dcc040c9c0 : SUCCESS in 6m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9599a9d0f8424d1294e75479788fb69b : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/601c84b6cbdb4ecbbbf93abb7e26c414 : SUCCESS in 5m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/215203fc85604b2db3d50f22c9e3239c : SUCCESS in 4m 45s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f0f674a82ef64f7eb1bfcc5330a30340 : SUCCESS in 32m 45s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f3041dedd2b249c8888703318db16462 : SUCCESS in 48m 26s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b1e1062616f1477d8b12fd1d50232836 : SUCCESS in 55m 14s","accounts_in_message":[],"_revision_number":6},{"id":"d119d4554007d61807f627e9de4bdae694e8085f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-26 10:36:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"2f0895e90032965513165f5594c54424a2bc3bc2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-26 10:38:30.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6ff754903a647258f99f36f6efec182 : SUCCESS in 1m 12s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/47a63af0fc1444b2973a04324ba07264 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"fb218ff9bf3803188991ae0f1a8f66ffa5176872","revisions":{"b7a7bca6489896d44d80b06afd80fd4b1fb1f491":{"kind":"REWORK","_number":1,"created":"2021-06-14 13:46:51.000000000","uploader":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"ref":"refs/changes/79/796279/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/796279/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/796279/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/796279/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/796279/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/796279/1"}}},"commit":{"parents":[{"commit":"4e4dcf0eae4ab0338ca3480a3524066c1fdd6218","subject":"Merge \"Remove panko\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4e4dcf0eae4ab0338ca3480a3524066c1fdd6218"}]}],"author":{"name":"Gaël THEROND (Fl1nt)","email":"gael.therond@bitswalk.com","date":"2021-06-14 13:45:09.000000000","tz":120},"committer":{"name":"Gaël THEROND (Fl1nt)","email":"gael.therond@bitswalk.com","date":"2021-06-14 13:45:09.000000000","tz":120},"subject":"Fix missing templating block for kolla-toolbox.","message":"Fix missing templating block for kolla-toolbox.\n\nChange-Id: I4ce8244a3fb034493a75f4096de2640e11b48931\nCloses-bug: #1931867\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b7a7bca6489896d44d80b06afd80fd4b1fb1f491"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b7a7bca6489896d44d80b06afd80fd4b1fb1f491"}]},"branch":"refs/heads/master"},"ed14ae323dec16e4309c591e5b0f668293119dd6":{"kind":"REWORK","_number":2,"created":"2021-06-14 14:32:02.000000000","uploader":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"ref":"refs/changes/79/796279/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/796279/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/796279/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/796279/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/796279/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/796279/2"}}},"commit":{"parents":[{"commit":"4e4dcf0eae4ab0338ca3480a3524066c1fdd6218","subject":"Merge \"Remove panko\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4e4dcf0eae4ab0338ca3480a3524066c1fdd6218"}]}],"author":{"name":"Gaël THEROND (Fl1nt)","email":"gael.therond@bitswalk.com","date":"2021-06-14 14:31:49.000000000","tz":120},"committer":{"name":"Gaël THEROND (Fl1nt)","email":"gael.therond@bitswalk.com","date":"2021-06-14 14:31:49.000000000","tz":120},"subject":"Fix missing templating block for kolla-toolbox.   * Fix improper URL.","message":"Fix missing templating block for kolla-toolbox.\n  * Fix improper URL.\n\nChange-Id: I4ce8244a3fb034493a75f4096de2640e11b48931\nBug: #1931867\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ed14ae323dec16e4309c591e5b0f668293119dd6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ed14ae323dec16e4309c591e5b0f668293119dd6"}]},"branch":"refs/heads/master"},"62a24eb8727f25c79198ae4ecc9ad44e5f0b2a25":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-06-15 13:15:14.000000000","uploader":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"ref":"refs/changes/79/796279/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/796279/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/796279/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/796279/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/796279/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/796279/3"}}},"commit":{"parents":[{"commit":"4e4dcf0eae4ab0338ca3480a3524066c1fdd6218","subject":"Merge \"Remove panko\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4e4dcf0eae4ab0338ca3480a3524066c1fdd6218"}]}],"author":{"name":"Gaël THEROND (Fl1nt)","email":"gael.therond@bitswalk.com","date":"2021-06-15 13:14:58.000000000","tz":120},"committer":{"name":"Gaël THEROND (Fl1nt)","email":"gael.therond@bitswalk.com","date":"2021-06-15 13:14:58.000000000","tz":120},"subject":"Fix missing templating block for kolla-toolbox.","message":"Fix missing templating block for kolla-toolbox.\n\nChange-Id: I4ce8244a3fb034493a75f4096de2640e11b48931\nCloses-Bug: #1931867\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/62a24eb8727f25c79198ae4ecc9ad44e5f0b2a25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/62a24eb8727f25c79198ae4ecc9ad44e5f0b2a25"}]},"branch":"refs/heads/master"},"46c66479da591814a147bc42ba79cc1050a51570":{"kind":"REWORK","_number":4,"created":"2021-07-23 09:46:50.000000000","uploader":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"ref":"refs/changes/79/796279/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/796279/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/796279/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/796279/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/796279/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/796279/4"}}},"commit":{"parents":[{"commit":"4e4dcf0eae4ab0338ca3480a3524066c1fdd6218","subject":"Merge \"Remove panko\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4e4dcf0eae4ab0338ca3480a3524066c1fdd6218"}]}],"author":{"name":"Gaël THEROND (Fl1nt)","email":"gael.therond@bitswalk.com","date":"2021-07-23 09:46:37.000000000","tz":120},"committer":{"name":"Gaël THEROND (Fl1nt)","email":"gael.therond@bitswalk.com","date":"2021-07-23 09:46:37.000000000","tz":120},"subject":"Fix missing templating block for kolla-toolbox.","message":"Fix missing templating block for kolla-toolbox.\n\nChange-Id: I4ce8244a3fb034493a75f4096de2640e11b48931\nCloses-Bug: #1931867\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/46c66479da591814a147bc42ba79cc1050a51570"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/46c66479da591814a147bc42ba79cc1050a51570"}]},"branch":"refs/heads/master"},"c7dc6dac01551862ee84526c4993be27c9731116":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-07-23 09:48:49.000000000","uploader":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"ref":"refs/changes/79/796279/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/796279/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/796279/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/796279/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/796279/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/796279/5"}}},"commit":{"parents":[{"commit":"a1c6ef413c96b5b39a5ef55fbe2c93bf964c2c9c","subject":"Merge \"docs: Update release management for Xena process\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a1c6ef413c96b5b39a5ef55fbe2c93bf964c2c9c"}]}],"author":{"name":"Gaël THEROND (Fl1nt)","email":"gael.therond@bitswalk.com","date":"2021-07-23 09:48:39.000000000","tz":120},"committer":{"name":"Gaël THEROND (Fl1nt)","email":"gael.therond@bitswalk.com","date":"2021-07-23 09:48:39.000000000","tz":120},"subject":"Fix missing templating block for kolla-toolbox.","message":"Fix missing templating block for kolla-toolbox.\n\nChange-Id: I4ce8244a3fb034493a75f4096de2640e11b48931\nCloses-Bug: #1931867\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c7dc6dac01551862ee84526c4993be27c9731116"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c7dc6dac01551862ee84526c4993be27c9731116"}]},"branch":"refs/heads/master"},"fb218ff9bf3803188991ae0f1a8f66ffa5176872":{"kind":"REWORK","_number":6,"created":"2021-07-23 14:05:13.000000000","uploader":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"ref":"refs/changes/79/796279/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/796279/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/796279/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/796279/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/796279/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/796279/6"}}},"commit":{"parents":[{"commit":"a1c6ef413c96b5b39a5ef55fbe2c93bf964c2c9c","subject":"Merge \"docs: Update release management for Xena process\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a1c6ef413c96b5b39a5ef55fbe2c93bf964c2c9c"}]}],"author":{"name":"Gaël THEROND (Fl1nt)","email":"gael.therond@bitswalk.com","date":"2021-07-23 14:04:52.000000000","tz":120},"committer":{"name":"Gaël THEROND (Fl1nt)","email":"gael.therond@bitswalk.com","date":"2021-07-23 14:04:52.000000000","tz":120},"subject":"Fix missing templating block for kolla-toolbox.","message":"Fix missing templating block for kolla-toolbox.\n\nChange-Id: I4ce8244a3fb034493a75f4096de2640e11b48931\nCloses-Bug: #1931867\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fb218ff9bf3803188991ae0f1a8f66ffa5176872"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fb218ff9bf3803188991ae0f1a8f66ffa5176872"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
