)]}'
{"id":"openstack%2Fkolla~796387","triplet_id":"openstack%2Fkolla~stable%2Ftrain~If69536387513ca221e03a80878016532a41aef87","project":"openstack/kolla","branch":"stable/train","hashtags":[],"change_id":"If69536387513ca221e03a80878016532a41aef87","subject":"Fix build of masakari-monitors image","status":"MERGED","created":"2021-06-15 07:42:45.000000000","updated":"2021-06-16 23:32:51.000000000","submitted":"2021-06-16 23:24:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"796387","meta_rev_id":"5d111b1dca80152fc456a12f23cb38d0705c2a69","_number":796387,"virtual_id_number":796387,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28295,"name":"Buddhika Sanjeewa","email":"bsanjeewa@kln.ac.lk","username":"bsanjeewa"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-16 23:24:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28295,"name":"Buddhika Sanjeewa","email":"bsanjeewa@kln.ac.lk","username":"bsanjeewa"},{"value":2,"date":"2021-06-16 17:10:24.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2021-06-16 15:15:45.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28295,"name":"Buddhika Sanjeewa","email":"bsanjeewa@kln.ac.lk","username":"bsanjeewa"},{"value":1,"date":"2021-06-16 17:10:24.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2021-06-16 15:14:47.000000000","_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":28295,"name":"Buddhika Sanjeewa","email":"bsanjeewa@kln.ac.lk","username":"bsanjeewa"},"all":[{"value":1,"date":"2021-06-16 23:32:51.000000000","post_submit":true,"_account_id":28295,"name":"Buddhika Sanjeewa","email":"bsanjeewa@kln.ac.lk","username":"bsanjeewa"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28295,"name":"Buddhika Sanjeewa","email":"bsanjeewa@kln.ac.lk","username":"bsanjeewa"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28295,"name":"Buddhika Sanjeewa","email":"bsanjeewa@kln.ac.lk","username":"bsanjeewa"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-15 09:23:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-16 15:13:49.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"CC"},{"updated":"2021-06-16 15:15:45.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2021-06-16 17:10:24.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-06-16 23:32:51.000000000","updated_by":{"_account_id":28295,"name":"Buddhika Sanjeewa","email":"bsanjeewa@kln.ac.lk","username":"bsanjeewa"},"reviewer":{"_account_id":28295,"name":"Buddhika Sanjeewa","email":"bsanjeewa@kln.ac.lk","username":"bsanjeewa"},"state":"REVIEWER"}],"messages":[{"id":"0edd799a1812068d81d9c7c7641a1b80687a2fc7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-06-15 07:42:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d8780e25a5c4e1c1e3bcc105b76100a9a7addcb4","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-06-15 07:49:54.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"421d0e96bba1983ba0eee448444df55e975fb95f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 09:23:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/76757e9bc2624a17b4b1f5314f7b0994 : SUCCESS in 50m 24s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/90e2eb79d05e4561b23cea9683752cc4 : SUCCESS in 53m 51s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/cb5a8664240145e491b244c6e313d7a0 : SUCCESS in 29m 51s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/5511c82287f845a58e7f72a66e7764b5 : SUCCESS in 37m 53s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/98625796b6594764a0ddb344127968f3 : SUCCESS in 1h 12m 18s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/0ad189c2ce194c21a7c60bf6a8483678 : SUCCESS in 1h 03m 21s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/58ad6d42503248eda9f52a28513073bf : SUCCESS in 1h 17m 20s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/28a9f9a7fe6645b899a2df46bfd9bf13 : SUCCESS in 1h 03m 45s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/83180e20ec1148cbbf644c7db28018aa : SUCCESS in 1h 26m 45s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/04c05fc6b3e64b2b889398a06728a95f : SUCCESS in 52m 41s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/0b662ec157ad47ca8d0135179c356d19 : FAILURE in 6m 47s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/0f7662200da0453dacd3a4110777d938 : FAILURE in 6m 55s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/521f12367e104865a159b209f0b038d1 : FAILURE in 10m 24s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffb37ae1b7fb49d3988ba63f2ae9cdee : POST_FAILURE in 5m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f1a786ebd8d34c079b2b16a2e4c7cce8 : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6e687cc6354c43e28aa9098a3148bd22 : SUCCESS in 5m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c2aa7ca002d74c20b86fb3d49835f51a : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b875d880a5a64855a3ab3f06c73f78f1 : SUCCESS in 6m 47s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/24d034e6f1d048c3bb5b087188f88988 : SUCCESS in 24m 25s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d7c330d87dda4c14befed32cbb7db105 : SUCCESS in 27m 14s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3e3c1e60fd2948e0905e8242852671aa : SUCCESS in 49m 08s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/024d290b954c4c1f9a24e53df737611d : SUCCESS in 1h 18m 54s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/dd3308f6c8884a55b0fae01b5152170b : SUCCESS in 1h 09m 52s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"99043d1668f59c348311be647669295ba0515b20","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-06-16 15:13:49.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a86e9fd5ff906dd221c0392965d4b62755ca284b","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-06-16 15:14:47.000000000","message":"Patch Set 1: -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"c30edab3a58abdf9923cbfa75315b9c65e8c8c07","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-06-16 15:15:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f84960a93c868a27bc3f2e7075c7a789b19c16f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 16:44:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/d3e3d3b2a30341569c18659369e50652 : SUCCESS in 47m 34s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/99c8e9203dfa43a5b2859be38b29b4b6 : SUCCESS in 58m 03s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/1bc217ca0e0b4b2198df06fb3ac54e01 : SUCCESS in 37m 38s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/688b35c9c19f4673879213cbeda87b79 : SUCCESS in 45m 19s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/aa4b1305f02d4ea588b24d99ce388288 : SUCCESS in 1h 19m 49s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/7fe06ee8b33847ab8a4509a8b8d70794 : SUCCESS in 1h 06m 01s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/8182ddf1d6fe4314b6f10aabc2d5810f : SUCCESS in 1h 13m 24s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/ff392a5225444b1fae51e70cb9abbdf1 : SUCCESS in 54m 58s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/23b2dc08f11f45d18bf69a5fed688cfb : SUCCESS in 1h 25m 41s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/b707173dae834fe6a24be3a2c2f26e5d : SUCCESS in 43m 42s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/33447bea49014c448d1d742e2d7bc68b : FAILURE in 7m 15s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/0d913963da8f4eb09ae68c83373f4730 : FAILURE in 8m 01s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/1c8d0ef6e5b9473a8e322beb33d3216b : FAILURE in 10m 41s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/debf730cf4584f8c8a722bdaa3b2f87c : SUCCESS in 5m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7ec300000b2545478f47607dd085a23a : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/33ace391c9034d33980b8be188160a87 : SUCCESS in 4m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3269a3adc525459ca6f5c06cb6e9086b : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a70d6e49ed854a6aa02e4fef350ed0e5 : SUCCESS in 6m 59s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6783a04932ef49e8b517cd40cb53959f : SUCCESS in 31m 47s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8ef84d41c3b84184b31ad4962bc83507 : SUCCESS in 44m 23s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/269fe6e15a1a4a29b35aa2cc29487d7f : SUCCESS in 51m 31s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/baf1ab3795ae4b258596ba95d0a3387f : SUCCESS in 1h 21m 34s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/169c5988861c4295bd9b93189c696e75 : SUCCESS in 1h 02m 07s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"e3b827ba7ae125393e5669cb4e80e3347f33c702","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-16 17:10:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"034b4b84164ce874398d5aabfc79928daf07ad50","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 17:10:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"29a8c0db34caccf104840d307b14821f1d5dcc2a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 20:17:53.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/7b0007932d1c4ec6b7a490302e6f7eed : TIMED_OUT in 3h 03m 03s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/e06a3406c28f42b6af08135e93a85cb3 : SUCCESS in 32m 15s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/5fc8ce3fe2884250952fdbf24e6166bf : SUCCESS in 1h 08m 41s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/5f04528ec58a43e3a669b6a2ed2b395a : SUCCESS in 56m 10s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/06611712779645a59c659e9e4659a961 : SUCCESS in 1h 15m 57s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/3c0008d34f764a81bb44e684516d896f : SUCCESS in 51m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0ded27d607945d097b496f8ed42f806 : SUCCESS in 5m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c87e0f38e0c94b0f9c259af9e243c714 : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/96291cee74634bf38f5afb6f1bc42310 : SUCCESS in 4m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d675babcf1a44538966be71123f290b2 : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76fd52b705b3497db6fed6a7d65c1fba : SUCCESS in 5m 58s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d9eb9247ba5140899f4758b4ab650027 : SUCCESS in 28m 31s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/749bac599a934d3fa4cd3f549eca5543 : SUCCESS in 48m 54s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1539f7d1171a483595e1176d66331adb : SUCCESS in 1h 04m 13s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"967a43d7253adee659b56e28c62c696e3bbf545b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-16 20:39:45.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"82294834e63a9ca6c009a2cadb2a27f976fb9a38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 21:59:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos-binary https://zuul.opendev.org/t/openstack/build/6b149cfde4ed4e2f84965a9da5a6b153 : SUCCESS in 40m 49s (non-voting)\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/1a8fc25c69ca4712998915e8b82e6269 : SUCCESS in 47m 33s\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/c0ff1cb8c29d4c2e8e9fafbd32c2b04f : SUCCESS in 25m 39s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/2d0ac281839b4cdaadb98f5283ddbec0 : SUCCESS in 33m 29s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/2a52a1d0cd1d44e4847851ae37a7d3e6 : SUCCESS in 1h 06m 40s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/d6814ba6bc19497ca7d44a2a04b32a31 : SUCCESS in 57m 30s\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/eaeb2b5541884365905cbe523e095a48 : SUCCESS in 1h 06m 39s (non-voting)\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/bc8863592d1d4989a3bd0e9587a4d8e5 : SUCCESS in 58m 12s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/133d8be988384acdb8b2d0dc96b31588 : SUCCESS in 1h 17m 12s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/f64c40bb16cd4720b978564f14a3d4c3 : SUCCESS in 43m 07s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/3f79aac544914718a9ed3b9db579a301 : FAILURE in 8m 20s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/25d9e0fb32f2427380c685f00c6e790d : FAILURE in 8m 30s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/00fc77e610af4c5e83e3a7a45465727d : FAILURE in 12m 31s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab03ac5dfb1f43ff9d10d361e604c29e : SUCCESS in 7m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1c37b70f15a944b190a21624d75ba8cd : SUCCESS in 7m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/852aa3f3de874f719f3a5811e25e619e : SUCCESS in 5m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4df94f2813454b23867c7bd8fef59596 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b26846b2aa0f469f9804b0704b808e47 : SUCCESS in 7m 57s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8995626e888442b8a1edecd6f8bc1f4e : SUCCESS in 46m 03s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6b8b1b1565034ba0a8e239eafee8617c : SUCCESS in 45m 43s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7fba54b9cfc0408f8dbc5cae6799bc26 : SUCCESS in 47m 40s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/be5057ecc8c14a0cb81ef94438ab5486 : SUCCESS in 51m 20s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/726bb478c68c400ca57ffbb301710931 : SUCCESS in 58m 14s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"e1bd640da2ce5d139eb4d036279ada79863170ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 21:59:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"aaa1306ea65a5a1433b62f475a6f12d1b07c62a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 23:24:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos-source https://zuul.opendev.org/t/openstack/build/3e880e552410450ea165209e5b4cdd64 : SUCCESS in 46m 08s\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/8c7c9bd12f3c4b3c831ce5ba11dd283c : SUCCESS in 31m 01s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/07c66193a50f473ca9080457fe0f3aba : SUCCESS in 1h 06m 06s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/a91b3b822bdb4473b37c52079e4f20fa : SUCCESS in 52m 44s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/dd45c80303a746cba263e9c721742923 : SUCCESS in 1h 24m 24s\n- tripleo-build-containers-centos-7 https://zuul.opendev.org/t/openstack/build/8491a3adc08c4f09b1605aecad3c3798 : SUCCESS in 41m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e624596815b48f29a2090797823181f : SUCCESS in 5m 16s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a1f1b8591e554b27a2c5f864a39036c5 : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d86b3c2a363f404281dd650b682cb109 : SUCCESS in 5m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/49cb6e323ac1427bafa220ad6e270ba4 : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6a1a24568ad4cbca7e77d7e1fac1d17 : SUCCESS in 5m 42s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7b0320b6a41f407b9ee2ae18380bc444 : SUCCESS in 21m 26s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/044294fe54af461c8b845c31a1b182ed : SUCCESS in 1h 09m 28s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6e2586c1a5264aca9b3a293cc831f8ca : SUCCESS in 1h 22m 18s\n\nWarning:\n  Comments left for invalid file docker/ceph/ceph-osd/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"8d96ca5b76ea5af673eaf9e58daf440f2603b107","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 23:24:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2161b3b49d181223c3122945100a65b2be9903d2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 23:25:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c52f2cfe89b41ec9d2eb8ce866c098e : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1},{"id":"5d111b1dca80152fc456a12f23cb38d0705c2a69","author":{"_account_id":28295,"name":"Buddhika Sanjeewa","email":"bsanjeewa@kln.ac.lk","username":"bsanjeewa"},"date":"2021-06-16 23:32:51.000000000","message":"Patch Set 1: Backport-Candidate+1\n\nNeed for Victoria too","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"715d799ab997cbfe9e444b3f2a72c5dbbbf6728e","revisions":{"715d799ab997cbfe9e444b3f2a72c5dbbbf6728e":{"kind":"REWORK","_number":1,"created":"2021-06-15 07:42:45.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/87/796387/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/87/796387/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/796387/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/796387/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/796387/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/87/796387/1"}}},"commit":{"parents":[{"commit":"326201932f93fc18371c2c5385fbf55ceb5d882d","subject":"CI: drop bashism from template overrides","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/326201932f93fc18371c2c5385fbf55ceb5d882d"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-06-15 07:42:32.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-06-15 07:42:32.000000000","tz":120},"subject":"Fix build of masakari-monitors image","message":"Fix build of masakari-monitors image\n\nThe CentOS 8 Advanced Virtualization repository recently added libvirt\n7.0.0. The current version of libvirt-python fails to build against it:\n\n    running build\n    /usr/bin/pkg-config --print-errors --atleast-version\u003d0.9.11 libvirt\n    /var/lib/kolla/venv/bin/python3.6 generator.py libvirt /usr/share/libvirt/api/libvirt-api.xml\n    Found 480 functions in /usr/share/libvirt/api/libvirt-api.xml\n    Found 0 functions in libvirt-override-api.xml\n    Generated 389 wrapper functions\n    Missing type converters:\n    char ** *:1\n    char **:1\n    ERROR: failed virDomainAuthorizedSSHKeysGet\n    ERROR: failed virDomainAuthorizedSSHKeysSet\n    error: command \u0027/var/lib/kolla/venv/bin/python3.6\u0027 failed with exit status 1\n\nWe need at least libvirt-python v6.10.0.\n\nChange-Id: If69536387513ca221e03a80878016532a41aef87\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/715d799ab997cbfe9e444b3f2a72c5dbbbf6728e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/715d799ab997cbfe9e444b3f2a72c5dbbbf6728e"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
