)]}'
{"id":"openstack%2Fkolla~796438","triplet_id":"openstack%2Fkolla~stable%2Fwallaby~Iefcdd3100b7e3c5320bc5f1286a18251bdeab885","project":"openstack/kolla","branch":"stable/wallaby","hashtags":[],"change_id":"Iefcdd3100b7e3c5320bc5f1286a18251bdeab885","subject":"Pin td-agent to 4.0.* to fix missing logs","status":"MERGED","created":"2021-06-15 14:29:32.000000000","updated":"2021-06-15 19:38:25.000000000","submitted":"2021-06-15 17:23:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"796438","cherry_pick_of_change":795086,"cherry_pick_of_patch_set":4,"meta_rev_id":"e8be243af9bab1e5dd4bd150b7bb3aac3d7eee76","_number":796438,"virtual_id_number":796438,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-15 17:23:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-06-15 15:58:18.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2021-06-15 14:30:37.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2021-06-15 15:58:18.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-15 14:29:32.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2021-06-15 15:58:18.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-06-15 16:01:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"56f0c66921604a4ad199df8068d7165ae02675c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-06-15 14:29:32.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"9927996b4515828fd31634b93b9357f8b6ac1349","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-06-15 14:30:37.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"15c977bafad87e279dc43d69f0ce7195a75a9b45","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-15 15:58:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7b589fea13a2d7e4e452ce0d3985d730caab27a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 16:01:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/11e25c87deed41b68cf0c99ff5b9641e : SUCCESS in 22m 23s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/f990b976e0274d5eafe8116730cdc9ba : SUCCESS in 44m 22s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/1f6d443e53cb4065b9696ed043a76563 : SUCCESS in 1h 07m 19s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/18003195f1454450945c8a901389143b : SUCCESS in 1h 07m 49s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/ba81926a4b0247eda5c5a7f4257a9f29 : SUCCESS in 1h 20m 23s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/42801958ab75407aa740a226ffeed0f8 : SUCCESS in 34m 07s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/527d475d64d84c3883f651a93fcdf08a : SUCCESS in 23m 54s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/0e15d4cf2ea941a28a58cc4a2d1f78f6 : SUCCESS in 1h 02m 29s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7fef48fccd89401d8685fb5a89b047d6 : SUCCESS in 6m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/431b3d3858fd478fa11f365c2122540b : SUCCESS in 6m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/33e90159b2e94fca8f30739df2207355 : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/80c74737b140443f8eb451d3b115e0f7 : SUCCESS in 5m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a46320243c264c43bf5da3f76a2bfee7 : SUCCESS in 7m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/286cc8191b174c5e8792f5ddfe9d6aef : SUCCESS in 6m 40s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/242c067f3ba843abae344a4c80ea3d7c : SUCCESS in 22m 38s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/79248730770a41e4b6639990a7b90481 : SUCCESS in 24m 12s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2f4e8b3da232490683a74400b14fc385 : SUCCESS in 54m 12s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9f10c649a7c740679d8a28790b2b7d2d : SUCCESS in 52m 00s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/fc67a5f76b14414fa36950512ac0b6bf : SUCCESS in 1h 20m 57s","accounts_in_message":[],"_revision_number":1},{"id":"e72e5b991cec8d02939a1a03cf802f28b7551f1c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 16:01:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6b0e37200bb646262d73fe339ab33c4bf94a652e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 17:23:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/2465acf1756e4747855c62a2c3f6c463 : SUCCESS in 39m 23s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/4a004ea952c949e6b7ffc0f051de5d44 : SUCCESS in 1h 05m 57s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/57a787fec8f2494083422d4a923d801c : SUCCESS in 1h 20m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/164ed310eb8b4abba3bd719863a071db : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/99b8187f10354145a574694acee22ab7 : SUCCESS in 5m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/138e302a68214b2790139911ce90b252 : SUCCESS in 5m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d1fdce56fb34e73b3deb7b7a51a63ad : SUCCESS in 8m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/806b26e69a8948e296ec8acfcefc58cc : SUCCESS in 5m 26s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/54c0461ba7c64c9da5a1d1e98ef95cc4 : SUCCESS in 33m 55s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c8aeaa5c8f7646f88b83d5bb21a9ef2a : SUCCESS in 53m 27s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/de7c439f8a7e4c6ea97f69c6936c2a2d : SUCCESS in 1h 02m 43s","accounts_in_message":[],"_revision_number":1},{"id":"cfea1f10ccd62e22e180b8fb1eff6e4e29623f72","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 17:23:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e29d50aaf7bbd083f56bad56102525e320488312","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 17:25:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03dcea058f14423ba18165e2cc1a240c : SUCCESS in 1m 05s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/571fc583b92f417faff584cd20ef668a : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":1},{"id":"e8be243af9bab1e5dd4bd150b7bb3aac3d7eee76","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-15 19:38:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/3ca9f1d0429447b99166674a7177fef7 : SUCCESS in 1h 00m 44s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b850972de34c4ef69b231999a5f83fd5 : SUCCESS in 1h 12m 43s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/cf6c68e427914625b0ca012b5541221d : SUCCESS in 1h 58m 34s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/252db936f41b41f68caee68ccfff454b : SUCCESS in 1h 03m 03s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"11d63e07c5168d9479f3fb542566c1fc6e4af9a3","revisions":{"11d63e07c5168d9479f3fb542566c1fc6e4af9a3":{"kind":"REWORK","_number":1,"created":"2021-06-15 14:29:32.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/38/796438/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/796438/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/796438/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/796438/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/796438/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/796438/1"}}},"commit":{"parents":[{"commit":"b5ba6792d8c765a9f34e80acade1e7f60a7cd084","subject":"[docs] Fix Debian release name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b5ba6792d8c765a9f34e80acade1e7f60a7cd084"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2021-06-07 10:14:55.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-06-15 14:29:32.000000000","tz":0},"subject":"Pin td-agent to 4.0.* to fix missing logs","message":"Pin td-agent to 4.0.* to fix missing logs\n\nThe recent inclusion of Fluentd 1.12.x in Centos/Ubuntu packages\ncauses a significant number of logs to fail to appear in the\nElasticsearch or Monasca output plugins. The logs appear to\nsit in a local buffer, and are only flushed when Fluentd is\nrestarted. The same issue persists in the most recent release\n(1.13.0), so for now we pin the td-agent package to 4.0.* which\nshould restrict Fluentd to the 1.11.x release where we don\u0027t\nsee this issue.\n\nBackport to Victoria.\n\nChange-Id: Iefcdd3100b7e3c5320bc5f1286a18251bdeab885\nCloses-Bug: #1930867\n(cherry picked from commit 6d0882ef799bb531d007c3e6839ef318fd512d5c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/11d63e07c5168d9479f3fb542566c1fc6e4af9a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/11d63e07c5168d9479f3fb542566c1fc6e4af9a3"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
