)]}'
{"id":"openstack%2Fkolla~796868","triplet_id":"openstack%2Fkolla~stable%2Fwallaby~I0b3eb6744a34a601266179fb358510a5c500e6ba","project":"openstack/kolla","branch":"stable/wallaby","topic":"pip_local-stable/wallaby","hashtags":[],"change_id":"I0b3eb6744a34a601266179fb358510a5c500e6ba","subject":"Fix _get_images_dir when pip-installed with --user","status":"MERGED","created":"2021-06-18 07:38:42.000000000","updated":"2021-06-21 12:49:58.000000000","submitted":"2021-06-21 12:46:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"796868-pip_local-stable/wallaby","cherry_pick_of_change":794153,"cherry_pick_of_patch_set":7,"meta_rev_id":"c1af778ff90fb3505facb43c407583ef159db692","_number":796868,"virtual_id_number":796868,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-21 12:46:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":2,"date":"2021-06-18 07:42:25.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2021-06-21 09:29:22.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":1,"date":"2021-06-21 09:29:22.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11928,"name":"jpic","email":"jamespic@gmail.com","username":"jpic"}],"reviewers":{"REVIEWER":[{"_account_id":11928,"name":"jpic","email":"jamespic@gmail.com","username":"jpic"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-18 07:38:42.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":11928,"name":"jpic","email":"jamespic@gmail.com","username":"jpic"},"state":"REVIEWER"},{"updated":"2021-06-18 08:59:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-21 09:29:22.000000000","updated_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"reviewer":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"state":"REVIEWER"}],"messages":[{"id":"cf12c9f1dc0043684ddc198cd39bc5f38972df37","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-06-18 07:38:42.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"a000dbeb9f75a98e6691b54944365f9584d64c98","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-06-18 07:42:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"387cb000ce2bbfbc258f589a31bbaa6d007b8c90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-18 08:59:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/00743d7903154f8ca7168649d2f51caf : SUCCESS in 27m 45s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/4424495239be4782aeb71f0a59ef150b : SUCCESS in 36m 25s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/6c48dcef20e1431ea8fa90d8f9f54b7b : SUCCESS in 1h 03m 23s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/4b7b35412f4746e199a4e4226517e08a : SUCCESS in 1h 00m 22s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/d32959f4864a426f93bd41537a323628 : SUCCESS in 1h 12m 01s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/28f6fd62282f42b79b0444d5b96f15ff : SUCCESS in 38m 52s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/2852b333b8c745fcbac0d07448bd4f3a : SUCCESS in 19m 22s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e1d8e956456d4dc6804fe179331adbfd : SUCCESS in 1h 15m 17s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13363fba6ac34187a88b2fd9edcb2378 : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6d3963eda62a4314baf1768251c72d3e : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b066ba9f91ae4ed2985f6fdd8bb18ea3 : SUCCESS in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/19bc39353abe4d758497493c49dd5c8d : SUCCESS in 5m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72b32bae62574550bef6c0691c9dab86 : SUCCESS in 8m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a1e68ac68ce242b7a3783850c45234fa : SUCCESS in 5m 03s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/daca527f08da49638ffb2f02634c9d36 : SUCCESS in 20m 06s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/5212db1f521547c2af749dd17e209bb4 : SUCCESS in 38m 18s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/012928194d7c406faf9633a4d04989e4 : SUCCESS in 55m 16s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b0f20118ecb5440389ffc7427c3d4d50 : SUCCESS in 47m 04s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/fc3b8ebbb0184bb3886320839ed1c6fe : SUCCESS in 59m 42s","accounts_in_message":[],"_revision_number":1},{"id":"9750840c984dd4e5dd9f928e83f6ec44df1da7b2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-18 10:41:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/65e04a7050784619bf737040b9654383 : SUCCESS in 2h 35m 53s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/51e7038a175d452a87b76b7dd73cf72b : SUCCESS in 2h 24m 29s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c91167865339441ba2a6f31a3f675018 : SUCCESS in 2h 52m 49s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/5adb5a10f4f54876a2a4ab1f72bfb679 : SUCCESS in 1h 45m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a19dd23dadb7cbc108366b681b8cf3e578dc5abd","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2021-06-21 09:29:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3094ca4469a25fede3bf38110c633e8be08fa2bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 09:29:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4312dd849391cd359078a6c2ccd1f7a9290598ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 12:46:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/949a1fd178c34ec4a75592d6873ec989 : SUCCESS in 1h 14m 45s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/f9db2fc3e16a496c9c62524d43b391be : SUCCESS in 1h 48m 23s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/1c39d896a6e246a2934c0d0741d5ef1f : SUCCESS in 1h 16m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c579517e93634b709c163a838f2374a0 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f46e4135d12a477fae1b263b2972fd85 : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1c74674610cf48c5b0b736238d651639 : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93322b931a5a4c0299102bd6b56db544 : SUCCESS in 7m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/25234c7586914242a3c4b29f9039f880 : SUCCESS in 4m 53s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/85bf110b76844957944d993d3f5290c3 : SUCCESS in 32m 13s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/93ba0662f63d4f4ea80d4d46da53496d : SUCCESS in 1h 03m 52s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/34786d99f4954218b5c01753f736eab1 : SUCCESS in 1h 12m 42s","accounts_in_message":[],"_revision_number":1},{"id":"802bab1c0a5f94a230f701da4b1cf33836076037","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 12:46:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c1af778ff90fb3505facb43c407583ef159db692","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 12:49:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d130b3d3f2184f698d08599c948f67cb : SUCCESS in 1m 31s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d29be3f7a7104f149e3934d76e369cdf : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8cd42eccc0d125c1496e110d276f17583526424c","revisions":{"8cd42eccc0d125c1496e110d276f17583526424c":{"kind":"REWORK","_number":1,"created":"2021-06-18 07:38:42.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/68/796868/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/68/796868/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/796868/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/796868/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/68/796868/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/68/796868/1"}}},"commit":{"parents":[{"commit":"11d63e07c5168d9479f3fb542566c1fc6e4af9a3","subject":"Pin td-agent to 4.0.* to fix missing logs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/11d63e07c5168d9479f3fb542566c1fc6e4af9a3"}]}],"author":{"name":"James Pic","email":"jpic@yourlabs.org","date":"2021-06-02 07:42:35.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-06-18 07:38:42.000000000","tz":0},"subject":"Fix _get_images_dir when pip-installed with --user","message":"Fix _get_images_dir when pip-installed with --user\n\nThere are two cases where pip installs kolla share into ~/.local/share/kolla:\n\n- with `pip install --user kolla`\n- with `pip install kolla` as non-root\n\nThis patch helps kolla find the share dir in these cases.\n\nCloses-Bug: #1930544\n\nChange-Id: I0b3eb6744a34a601266179fb358510a5c500e6ba\n(cherry picked from commit db78657c231663f3e8ebbc18de43c06ddaa94c97)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8cd42eccc0d125c1496e110d276f17583526424c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8cd42eccc0d125c1496e110d276f17583526424c"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
