)]}'
{"id":"openstack%2Fkolla~797102","triplet_id":"openstack%2Fkolla~stable%2Fvictoria~If69536387513ca221e03a80878016532a41aef87","project":"openstack/kolla","branch":"stable/victoria","hashtags":[],"change_id":"If69536387513ca221e03a80878016532a41aef87","subject":"Fix build of masakari-monitors image","status":"MERGED","created":"2021-06-18 17:36:42.000000000","updated":"2021-06-21 12:34:02.000000000","submitted":"2021-06-21 12:32:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"797102","cherry_pick_of_change":796387,"cherry_pick_of_patch_set":1,"meta_rev_id":"525591f7f8cb2f81437ee18d3719496af15d91dd","_number":797102,"virtual_id_number":797102,"owner":{"_account_id":28295,"name":"Buddhika Sanjeewa","email":"bsanjeewa@kln.ac.lk","username":"bsanjeewa"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-21 12:32:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-06-21 09:20:45.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2021-06-20 07:52:34.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2021-06-21 09:20:45.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-18 17:36:42.000000000","updated_by":{"_account_id":28295,"name":"Buddhika Sanjeewa","email":"bsanjeewa@kln.ac.lk","username":"bsanjeewa"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2021-06-18 19:00:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-20 07:52:34.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2021-06-21 09:20:45.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"b879fe019deb1312eb6d6eef45a63965ee25da0e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28295,"name":"Buddhika Sanjeewa","email":"bsanjeewa@kln.ac.lk","username":"bsanjeewa"},"date":"2021-06-18 17:36:42.000000000","message":"Patch Set 1: Cherry Picked from branch stable/train.\n\nThe following files contain Git conflicts:\n* docker/masakari/masakari-monitors/Dockerfile.j2","accounts_in_message":[],"_revision_number":1},{"id":"9b5750b5a5a23c534b6091787d5d8003f5508363","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28295,"name":"Buddhika Sanjeewa","email":"bsanjeewa@kln.ac.lk","username":"bsanjeewa"},"date":"2021-06-18 17:39:34.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"a9c6e6bc1e485c6d59a5201b1ef93a575e188e8a","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":28295,"name":"Buddhika Sanjeewa","email":"bsanjeewa@kln.ac.lk","username":"bsanjeewa"},"date":"2021-06-18 17:41:22.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"2e7764f6a2bbf04d3fe47f1a3f763c0b8be1cfc3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28295,"name":"Buddhika Sanjeewa","email":"bsanjeewa@kln.ac.lk","username":"bsanjeewa"},"date":"2021-06-18 17:53:50.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"94964941ab0e7edfe01da0ad808febdd45a3e6b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-18 19:00:53.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/ee2f76594b1f4d1fa520d24a3604bc02 : SUCCESS in 23m 19s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/6edb4f2939b840ca90c85e4d75808400 : SUCCESS in 23m 57s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/7415408529d14147b84344184defdc76 : SUCCESS in 1h 02m 30s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/c3150c7dbe86466da202781f8568c252 : SUCCESS in 1h 01m 54s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/91d478be939c45dfad9b7f2911b7517e : SUCCESS in 1h 02m 29s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/1aecb18dde284dbc8130a4a6351a91b2 : SUCCESS in 22m 57s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/b1b835f43a6f417d9d4e3603802ef747 : SUCCESS in 33m 37s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/daabd855e59141739a08e59fc527a039 : SUCCESS in 54m 42s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb35f71965134f3e847800dc01cd1dd2 : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/711ed4c40e78434c8b73f3ce3157b3ae : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e92ff9216a264fd98cd288b2427c1a33 : SUCCESS in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4795bf83448b452487913276423c2b28 : SUCCESS in 6m 53s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1dddd10202e34516a2f0306a838284cd : SUCCESS in 31m 07s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/fa5accfb43dc4537b55cf68cda0fb12c : SUCCESS in 24m 17s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/bf2d1847c0c24a43ac2076bbfd2b7a30 : SUCCESS in 57m 03s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b2020108291643a5ae6409f3488f0b7b : SUCCESS in 59m 32s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/bd4a5c02f8b0424ebd09125adbbb2a17 : SUCCESS in 1h 04m 56s","accounts_in_message":[],"_revision_number":3},{"id":"3f73b1417e9026fd4083dad3846c107d57c22254","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-18 19:59:55.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/ea9d7956bd4e429ca7adb99653ab4cf1 : SUCCESS in 35m 54s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3c7f8ddbf395478689664b9a026a9364 : SUCCESS in 1h 29m 04s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f193b442952842efa4d3f4bf6be52fbc : SUCCESS in 2h 04m 12s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/5c76cede1c124123b84299b0db96f8f6 : SUCCESS in 1h 28m 42s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2feb2586d985fe9a1fef45162c9ef9156e02f3c2","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-06-20 07:52:34.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"3b01b97c5ae59048f509f4c993e8f2b40b076b4b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-21 09:20:45.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"9f363b0fb240c4386ba79d6d89827655c2e5405d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 09:21:11.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"062c6c2e0869ce81f9ace6b88076012ee93dc9be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 12:32:35.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/1eb6f18beee74008adae621881bced52 : SUCCESS in 35m 08s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/3c9fdda330e842ff98103744613ca878 : SUCCESS in 1h 15m 25s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/cb4c7ce635fa44cbb50f8bf1f19f8dd4 : SUCCESS in 1h 23m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9000edf7a20245038ff8eac9b2e16b61 : SUCCESS in 6m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e9a5646a0b5f48eaa10330cabb848056 : SUCCESS in 7m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8baba612faed4eed98732538b1cc6678 : SUCCESS in 7m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7406c82ce5364c8e9871150562f0115f : SUCCESS in 18m 06s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c38060e3639d49a09d03c3a5a40b3d49 : SUCCESS in 1h 00m 46s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/27d71bb723db4dca9c0279f2eece2dea : SUCCESS in 1h 19m 52s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d6b9f63045f84ef28a58fd566da8cd86 : SUCCESS in 1h 34m 05s","accounts_in_message":[],"_revision_number":3},{"id":"d8f8f2cd68e3111376112184344e8c2d75fc3c50","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 12:32:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"525591f7f8cb2f81437ee18d3719496af15d91dd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 12:34:02.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acf40e5ddb7849dba0185b76b04375ee : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"29048206ed67323b318f35f5756d41e8e83f5b8d","revisions":{"70e54792885505db97e422aa7f353fdde769aa96":{"kind":"REWORK","_number":1,"created":"2021-06-18 17:36:42.000000000","uploader":{"_account_id":28295,"name":"Buddhika Sanjeewa","email":"bsanjeewa@kln.ac.lk","username":"bsanjeewa"},"ref":"refs/changes/02/797102/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/02/797102/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/797102/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/797102/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/797102/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/02/797102/1"}}},"commit":{"parents":[{"commit":"d52bb8e7b94e532a3b8c7e9c5ce4560a8824a26d","subject":"Pin td-agent to 4.0.* to fix missing logs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d52bb8e7b94e532a3b8c7e9c5ce4560a8824a26d"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-06-15 07:42:32.000000000","tz":120},"committer":{"name":"Buddhika Sanjeewa","email":"bsanjeewa@kln.ac.lk","date":"2021-06-18 17:36:42.000000000","tz":0},"subject":"Fix build of masakari-monitors image","message":"Fix build of masakari-monitors image\n\nThe CentOS 8 Advanced Virtualization repository recently added libvirt\n7.0.0. The current version of libvirt-python fails to build against it:\n\n    running build\n    /usr/bin/pkg-config --print-errors --atleast-version\u003d0.9.11 libvirt\n    /var/lib/kolla/venv/bin/python3.6 generator.py libvirt /usr/share/libvirt/api/libvirt-api.xml\n    Found 480 functions in /usr/share/libvirt/api/libvirt-api.xml\n    Found 0 functions in libvirt-override-api.xml\n    Generated 389 wrapper functions\n    Missing type converters:\n    char ** *:1\n    char **:1\n    ERROR: failed virDomainAuthorizedSSHKeysGet\n    ERROR: failed virDomainAuthorizedSSHKeysSet\n    error: command \u0027/var/lib/kolla/venv/bin/python3.6\u0027 failed with exit status 1\n\nWe need at least libvirt-python v6.10.0.\n\nChange-Id: If69536387513ca221e03a80878016532a41aef87\n(cherry picked from commit 715d799ab997cbfe9e444b3f2a72c5dbbbf6728e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/70e54792885505db97e422aa7f353fdde769aa96"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/70e54792885505db97e422aa7f353fdde769aa96"}]},"branch":"refs/heads/stable/victoria"},"21078e4a97478d1d172ad3189f3bdbe7fa1b5f4b":{"kind":"REWORK","_number":2,"created":"2021-06-18 17:39:34.000000000","uploader":{"_account_id":28295,"name":"Buddhika Sanjeewa","email":"bsanjeewa@kln.ac.lk","username":"bsanjeewa"},"ref":"refs/changes/02/797102/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/02/797102/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/797102/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/797102/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/797102/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/02/797102/2"}}},"commit":{"parents":[{"commit":"d52bb8e7b94e532a3b8c7e9c5ce4560a8824a26d","subject":"Pin td-agent to 4.0.* to fix missing logs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d52bb8e7b94e532a3b8c7e9c5ce4560a8824a26d"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-06-15 07:42:32.000000000","tz":120},"committer":{"name":"Buddhika Sanjeewa","email":"bsanjeewa@kln.ac.lk","date":"2021-06-18 17:37:45.000000000","tz":0},"subject":"Fix build of masakari-monitors image","message":"Fix build of masakari-monitors image\n\nThe CentOS 8 Advanced Virtualization repository recently added libvirt\n7.0.0. The current version of libvirt-python fails to build against it:\n\n    running build\n    /usr/bin/pkg-config --print-errors --atleast-version\u003d0.9.11 libvirt\n    /var/lib/kolla/venv/bin/python3.6 generator.py libvirt /usr/share/libvirt/api/libvirt-api.xml\n    Found 480 functions in /usr/share/libvirt/api/libvirt-api.xml\n    Found 0 functions in libvirt-override-api.xml\n    Generated 389 wrapper functions\n    Missing type converters:\n    char ** *:1\n    char **:1\n    ERROR: failed virDomainAuthorizedSSHKeysGet\n    ERROR: failed virDomainAuthorizedSSHKeysSet\n    error: command \u0027/var/lib/kolla/venv/bin/python3.6\u0027 failed with exit status 1\n\nWe need at least libvirt-python v6.10.0.\n\nChange-Id: If69536387513ca221e03a80878016532a41aef87\n(cherry picked from commit 715d799ab997cbfe9e444b3f2a72c5dbbbf6728e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/21078e4a97478d1d172ad3189f3bdbe7fa1b5f4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/21078e4a97478d1d172ad3189f3bdbe7fa1b5f4b"}]},"branch":"refs/heads/stable/victoria"},"29048206ed67323b318f35f5756d41e8e83f5b8d":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-06-18 17:53:50.000000000","uploader":{"_account_id":28295,"name":"Buddhika Sanjeewa","email":"bsanjeewa@kln.ac.lk","username":"bsanjeewa"},"ref":"refs/changes/02/797102/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/02/797102/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/797102/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/797102/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/797102/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/02/797102/3"}}},"commit":{"parents":[{"commit":"d52bb8e7b94e532a3b8c7e9c5ce4560a8824a26d","subject":"Pin td-agent to 4.0.* to fix missing logs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d52bb8e7b94e532a3b8c7e9c5ce4560a8824a26d"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-06-15 07:42:32.000000000","tz":120},"committer":{"name":"Buddhika Sanjeewa","email":"bsanjeewa@kln.ac.lk","date":"2021-06-18 17:52:35.000000000","tz":0},"subject":"Fix build of masakari-monitors image","message":"Fix build of masakari-monitors image\n\nThe CentOS 8 Advanced Virtualization repository recently added libvirt\n7.0.0. The current version of libvirt-python fails to build against it:\n\n    running build\n    /usr/bin/pkg-config --print-errors --atleast-version\u003d0.9.11 libvirt\n    /var/lib/kolla/venv/bin/python3.6 generator.py libvirt /usr/share/libvirt/api/libvirt-api.xml\n    Found 480 functions in /usr/share/libvirt/api/libvirt-api.xml\n    Found 0 functions in libvirt-override-api.xml\n    Generated 389 wrapper functions\n    Missing type converters:\n    char ** *:1\n    char **:1\n    ERROR: failed virDomainAuthorizedSSHKeysGet\n    ERROR: failed virDomainAuthorizedSSHKeysSet\n    error: command \u0027/var/lib/kolla/venv/bin/python3.6\u0027 failed with exit status 1\n\nWe need at least libvirt-python v6.10.0.\n\nCloses-Bug: #1931817\nChange-Id: If69536387513ca221e03a80878016532a41aef87\n(cherry picked from commit 715d799ab997cbfe9e444b3f2a72c5dbbbf6728e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/29048206ed67323b318f35f5756d41e8e83f5b8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/29048206ed67323b318f35f5756d41e8e83f5b8d"}]},"branch":"refs/heads/stable/victoria","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
