)]}'
{"id":"openstack%2Fkolla~797109","triplet_id":"openstack%2Fkolla~stable%2Fussuri~If69536387513ca221e03a80878016532a41aef87","project":"openstack/kolla","branch":"stable/ussuri","hashtags":[],"change_id":"If69536387513ca221e03a80878016532a41aef87","subject":"Fix build of masakari-monitors image","status":"MERGED","created":"2021-06-20 07:52:43.000000000","updated":"2021-06-21 12:09:25.000000000","submitted":"2021-06-21 12:07:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"797109","cherry_pick_of_change":797102,"cherry_pick_of_patch_set":3,"meta_rev_id":"7d95fa13b89290c89146a2b1bc8dcf3650a40fcc","_number":797109,"virtual_id_number":797109,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30449,"name":"jinyuanliu","email":"liujinyuan@inspur.com","username":"jinyuanliu"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-21 12:07:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":30449,"name":"jinyuanliu","email":"liujinyuan@inspur.com","username":"jinyuanliu"},"all":[{"value":1,"date":"2021-06-20 13:32:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30449,"name":"jinyuanliu","email":"liujinyuan@inspur.com","username":"jinyuanliu"},{"value":2,"date":"2021-06-21 09:20:16.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2021-06-20 07:53:32.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":30449,"name":"jinyuanliu","email":"liujinyuan@inspur.com","username":"jinyuanliu"},{"value":1,"date":"2021-06-21 09:20:16.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":30449,"name":"jinyuanliu","email":"liujinyuan@inspur.com","username":"jinyuanliu"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":30449,"name":"jinyuanliu","email":"liujinyuan@inspur.com","username":"jinyuanliu"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30449,"name":"jinyuanliu","email":"liujinyuan@inspur.com","username":"jinyuanliu"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-20 07:52:43.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2021-06-20 09:00:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-20 13:32:09.000000000","updated_by":{"_account_id":30449,"name":"jinyuanliu","email":"liujinyuan@inspur.com","username":"jinyuanliu"},"reviewer":{"_account_id":30449,"name":"jinyuanliu","email":"liujinyuan@inspur.com","username":"jinyuanliu"},"state":"REVIEWER"},{"updated":"2021-06-21 09:20:16.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"d39e8c6f4fba5b4e515873cc94f598fea20da712","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-06-20 07:52:43.000000000","message":"Patch Set 1: Cherry Picked from branch stable/victoria.","accounts_in_message":[],"_revision_number":1},{"id":"3f0b890625a02d442a7b361680ef7566fcdeaca4","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-06-20 07:53:32.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a8d652c6a3da081154e0a61fac6114778de5077b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-20 09:00:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/5bd576e46fd647e2883f74f8ded6a90e : SUCCESS in 22m 43s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/2765f07e515a44f8b2e2059dcaf23b01 : SUCCESS in 44m 26s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/f5b5c8ab7acb42e887ddfbc75233134f : SUCCESS in 1h 02m 53s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/f511f657d0f7417b990646aaeb1baedf : SUCCESS in 50m 50s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/c3328849045c4a5f8be739630cb4e8eb : SUCCESS in 1h 00m 49s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/60e2f8be09e84f34a6258d2956d2e947 : SUCCESS in 23m 02s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/500e6a39868f4e54b358f9536b898a6a : SUCCESS in 37m 57s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/0846cc92e0ae478085a2d72102b81a94 : SUCCESS in 53m 00s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e11df022ff6242738a0fa836dc6a1ecb : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6dc86e42740e4489bc945fa00a1c9e97 : SUCCESS in 4m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/edd1f5088c5047c9a9111292236921d4 : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/074ddf63e86c40f3abc5d955da161bf7 : SUCCESS in 8m 45s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cdc32a9d83d7495390bb4bfd20845484 : SUCCESS in 20m 42s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/543e424de1474e77a3d574f408f99651 : SUCCESS in 48m 56s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5722856a9e404334b612fad122cb67d6 : SUCCESS in 1h 00m 37s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6a0bb9129be44d7cb0903819aa30e0d2 : SUCCESS in 57m 11s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f60f5068c8a344dbabc3750b2fb47e2b : SUCCESS in 1h 00m 01s","accounts_in_message":[],"_revision_number":1},{"id":"0f8db148000d3ae762e760a948518766e92ac865","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-20 09:57:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/1d3a711412be4d9e9c8345aafe410d75 : SUCCESS in 1h 22m 26s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9d9246b47df54908813a9378be09d3e4 : SUCCESS in 1h 13m 07s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3a61633208c6469188fb8acc22740d1c : SUCCESS in 1h 56m 56s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/25bc2ad7559549a7870179cf5aa2ea09 : SUCCESS in 1h 14m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c2b6b63198fd86baf14009f2f4c3a84ab7fd51be","author":{"_account_id":30449,"name":"jinyuanliu","email":"liujinyuan@inspur.com","username":"jinyuanliu"},"date":"2021-06-20 13:32:09.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0d82be1c4e7c82e2b62214dc608ef4dcca451a00","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-21 09:20:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"71f3ca2540aa711becc660e1250b92990fa25ba9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 09:20:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"57fbade920b8a8620fbc6383ba892148b0556525","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 12:07:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/bd4f75e9f1794476b198443d3f920653 : SUCCESS in 33m 47s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/9601a819564e4dda9ec47eea5b9d2826 : SUCCESS in 1h 02m 07s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/8d5459027ebb4bdb8af53488370cbe39 : SUCCESS in 1h 05m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9613c4aa3234f51a25af402b204fa31 : SUCCESS in 6m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ea633979fc68467ba147ed317931eb21 : SUCCESS in 5m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/476fb39377f9463f83ba495664d00fa2 : SUCCESS in 7m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b58e385cfe784da5a0481f86f1701b62 : SUCCESS in 7m 57s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/84e97b09e86440f4bdaf72bc486d5f06 : SUCCESS in 30m 47s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1aa99b6568b3457bb7c95e8d31a4d6f1 : SUCCESS in 56m 27s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/49bce20301d54f7fa2c49378645de6b1 : SUCCESS in 1h 14m 53s","accounts_in_message":[],"_revision_number":1},{"id":"0d3300fb97c02195a51e20bf13181cc6d721ea64","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 12:07:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7d95fa13b89290c89146a2b1bc8dcf3650a40fcc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 12:09:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0e7f494a2d34e228db93d18b53388c4 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"253c215f1ea1ce9547fff7442c570248494c7115","revisions":{"253c215f1ea1ce9547fff7442c570248494c7115":{"kind":"REWORK","_number":1,"created":"2021-06-20 07:52:43.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/09/797109/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/09/797109/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/09/797109/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/09/797109/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/09/797109/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/09/797109/1"}}},"commit":{"parents":[{"commit":"a51deb4f02574189524357c93a7cf22bf9366c71","subject":"Merge \"[CI] Support building source images with in-review changes\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a51deb4f02574189524357c93a7cf22bf9366c71"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-06-15 07:42:32.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-06-20 07:52:43.000000000","tz":0},"subject":"Fix build of masakari-monitors image","message":"Fix build of masakari-monitors image\n\nThe CentOS 8 Advanced Virtualization repository recently added libvirt\n7.0.0. The current version of libvirt-python fails to build against it:\n\n    running build\n    /usr/bin/pkg-config --print-errors --atleast-version\u003d0.9.11 libvirt\n    /var/lib/kolla/venv/bin/python3.6 generator.py libvirt /usr/share/libvirt/api/libvirt-api.xml\n    Found 480 functions in /usr/share/libvirt/api/libvirt-api.xml\n    Found 0 functions in libvirt-override-api.xml\n    Generated 389 wrapper functions\n    Missing type converters:\n    char ** *:1\n    char **:1\n    ERROR: failed virDomainAuthorizedSSHKeysGet\n    ERROR: failed virDomainAuthorizedSSHKeysSet\n    error: command \u0027/var/lib/kolla/venv/bin/python3.6\u0027 failed with exit status 1\n\nWe need at least libvirt-python v6.10.0.\n\nCloses-Bug: #1931817\nChange-Id: If69536387513ca221e03a80878016532a41aef87\n(cherry picked from commit 715d799ab997cbfe9e444b3f2a72c5dbbbf6728e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/253c215f1ea1ce9547fff7442c570248494c7115"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/253c215f1ea1ce9547fff7442c570248494c7115"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
