)]}'
{"id":"openstack%2Fkolla~797300","triplet_id":"openstack%2Fkolla~stable%2Fvictoria~I4faabdfa9c273fc61b536e6ce88b8d71ab2fc581","project":"openstack/kolla","branch":"stable/victoria","topic":"bug/1928408-stable/victoria","hashtags":[],"change_id":"I4faabdfa9c273fc61b536e6ce88b8d71ab2fc581","subject":"horizon: Correct location of monitoring_policy file","status":"MERGED","created":"2021-06-21 16:27:04.000000000","updated":"2021-06-22 09:49:23.000000000","submitted":"2021-06-22 09:47:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"797300-bug/1928408-stable/victoria","cherry_pick_of_change":792838,"cherry_pick_of_patch_set":4,"meta_rev_id":"6805d72bace1e0f648d7129570906d47921d2b73","_number":797300,"virtual_id_number":797300,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-22 09:47:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-06-22 08:31:51.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2021-06-21 16:37:41.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-06-21 17:46:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2021-06-22 08:31:51.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-21 16:27:04.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"},"state":"REVIEWER"},{"updated":"2021-06-21 16:37:41.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2021-06-21 17:43:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ab0d8a7025803e6b37f8bbc2fb0881beb5aec2e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-21 16:27:04.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"9f94a54b074121adca9489bcc80ace92e9e48c93","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-06-21 16:37:41.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a71e5ad4095a845a7269bb8a80618267e75a6c0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 17:43:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/0d2df61408ad4835948b8087d073b45f : SUCCESS in 30m 52s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/86d8f4231d18402db6f840fa68aa355b : SUCCESS in 34m 52s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/584434013c364f06a76ec49b6d5cf4c7 : SUCCESS in 1h 09m 01s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/17b10a84708841d09540fcbf21a12fa6 : SUCCESS in 1h 05m 56s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/6c377c528d5a4dd4a0da154a7b0fc9cc : SUCCESS in 1h 15m 21s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/840c70a37ab14c91a14e67864e86ab61 : SUCCESS in 25m 32s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/026c0061634b4e2784142c3bfba06fa9 : SUCCESS in 28m 12s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d6e4cdab9dea4d81a5a8e4266dee910f : SUCCESS in 1h 04m 06s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b76bf8a461e41e5ad3215f2d253d6cb : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/58dac98aa4c34cf29cc98b15cc8179ad : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/38a41c9aece84dce980567d80732294f : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/100b85ada6634933a96d1206e653f8ea : SUCCESS in 7m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3bee95a331e741a98b14e823dc9021f7 : SUCCESS in 5m 06s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5de5c05dd9e84ba7a9a624cdf04c0ff6 : SUCCESS in 39m 25s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7c2dc55f98e84c3a90ada1294b2b8d72 : SUCCESS in 41m 10s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/749f749988224cc2b56f7885b294b4df : SUCCESS in 46m 26s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6c26293ca34545a8ab421d85955d194e : SUCCESS in 44m 46s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b46bb3d5531b4f9898ad02e394d45e36 : SUCCESS in 1h 07m 29s","accounts_in_message":[],"_revision_number":1},{"id":"559945aa435cdc572a6d07348c9319060fab0b4f","author":{"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"},"date":"2021-06-21 17:46:35.000000000","message":"Patch Set 1: Code-Review+1\n\nLooks good","accounts_in_message":[],"_revision_number":1},{"id":"ea1659f0300d5648f4fc6b21c58d307a47a496f8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 18:25:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/253265894c3744db8e2f867e3efd292a : SUCCESS in 1h 51m 17s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/cfb9c602b165442c86a590597e47b1d1 : SUCCESS in 44m 43s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/97316408280940b992c2595246bb7446 : SUCCESS in 1h 56m 31s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/5a73aad363254b338dab5764128882a8 : SUCCESS in 42m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cea5db7ce1a5eae7da6a5dcac5de394668ed5629","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-22 08:31:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0ca6f438e75e292d77ab6a7094f7a6d3b1d44452","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-22 08:32:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"821bde4177b8ec4513b2f93f190a2c3c8782b7c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-22 09:47:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/996dfbb78db440969ef9992f7c314896 : SUCCESS in 25m 19s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/26670e436ede477ead9ea2327f917c13 : SUCCESS in 1h 04m 19s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/fc90ee572c3045a4b514df1e0dbe46f3 : SUCCESS in 1h 09m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5642221b33848969ef630de434ccac3 : SUCCESS in 6m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/74c18eb84ab74f73b1c3f7e3d634c457 : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9293322b42054c7ca8cd5d5fad844fba : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84f8a4a4318a4fe4a07b81a16c03a47f : SUCCESS in 7m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d6388b311fe84858842c2bedcb32dafb : SUCCESS in 4m 30s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6f8c49c18e2b41a4afd24a67c985fa46 : SUCCESS in 21m 39s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ccc0afd32cbe4494b42dc6f3339be5b0 : SUCCESS in 49m 17s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/82a15c5bc2154405a2f0905a920c980a : SUCCESS in 59m 00s","accounts_in_message":[],"_revision_number":1},{"id":"1de01fc8923a28c367c46a2c5a56dc76898ca74f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-22 09:47:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6805d72bace1e0f648d7129570906d47921d2b73","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-22 09:49:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd7ab0f82e7749aa94a79a699e908d7c : SUCCESS in 1m 12s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/533c99e767154007a418c85e8dc7d2d3 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"32eb32f1b66416a2ccd385bd4afee765bf1eb712","revisions":{"32eb32f1b66416a2ccd385bd4afee765bf1eb712":{"kind":"REWORK","_number":1,"created":"2021-06-21 16:27:04.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/00/797300/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/797300/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/797300/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/797300/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/797300/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/797300/1"}}},"commit":{"parents":[{"commit":"c87f82fb42fd11f8131c31dbc8bb77ce1c05b137","subject":"Merge \"Fix _get_images_dir when pip-installed with --user\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c87f82fb42fd11f8131c31dbc8bb77ce1c05b137"}]}],"author":{"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","date":"2021-05-24 17:51:23.000000000","tz":-420},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-06-21 16:27:04.000000000","tz":0},"subject":"horizon: Correct location of monitoring_policy file","message":"horizon: Correct location of monitoring_policy file\n\nPatch to correctly copy monitoring_policy.json into\n/etc/openstack-dashboard.  Policy was misplaced, and not being enforced.\n\nNote that by current default policy, admin doesn\u0027t not have Monitoring\naccess.\n\nCloses-Bug: #1928408\nChange-Id: I4faabdfa9c273fc61b536e6ce88b8d71ab2fc581\n(cherry picked from commit f68dfb88fdbad51322727575457b3d78a70466b1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/32eb32f1b66416a2ccd385bd4afee765bf1eb712"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/32eb32f1b66416a2ccd385bd4afee765bf1eb712"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
