)]}'
{"id":"openstack%2Fkolla~798033","triplet_id":"openstack%2Fkolla~master~Ie89f3628c9dfab683973d845a3812bccac9f5bb8","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"Ie89f3628c9dfab683973d845a3812bccac9f5bb8","subject":"Clean up extra spaces","status":"MERGED","created":"2021-06-25 03:21:22.000000000","updated":"2021-07-08 16:36:42.000000000","submitted":"2021-06-28 14:08:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"798033","meta_rev_id":"974e70c761d3029c3946bb0397754b57b14393bc","_number":798033,"virtual_id_number":798033,"owner":{"_account_id":30449,"name":"jinyuanliu","email":"liujinyuan@inspur.com","username":"jinyuanliu"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-28 14:08:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2021-06-28 11:13:51.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2021-06-28 11:13:51.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-25 03:31:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-28 10:49:44.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-06-28 11:13:51.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"81ea71de1336079803b2bbb83bcea5c71cd1216c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30449,"name":"jinyuanliu","email":"liujinyuan@inspur.com","username":"jinyuanliu"},"date":"2021-06-25 03:21:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5a6cd7348dec3d1d4b6814cd55a3efd88d8d1f59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-25 03:31:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4edc0b4ec15c416a8c99e8df96f126e2 : SUCCESS in 6m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e4dc7fec2b854c90a4e8ed331cfbf5d6 : SUCCESS in 6m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0fc46cda30874a21b50736babe4d0d39 : SUCCESS in 6m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1ff89272453948ecaa959349d5a6d281 : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9069c1193554343b0dcab6dd367ad91 : SUCCESS in 7m 48s","accounts_in_message":[],"_revision_number":1},{"id":"acba049493426a9f3c12ca2cb85273f31e23abef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-25 03:32:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cf1fbdd887ef459a99c0cdc747901690 : SUCCESS in 5m 41s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8f24fb05d1db48a3a14fad9da7c170ff : SUCCESS in 6m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"be2a4c83764d8c9aa09b3e5845e2db39614cb285","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30449,"name":"jinyuanliu","email":"liujinyuan@inspur.com","username":"jinyuanliu"},"date":"2021-06-25 04:09:51.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b345637d8d8334c32f0f4c1a86e70970c5a1d89a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-25 04:21:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0fc52837a5924e5c8363afb1bc43b3b0 : SUCCESS in 7m 24s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2afb8859a48b484b91703c0b8465f2a9 : SUCCESS in 6m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b298a5377d4f11f926be0728fd81829e76cef13e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-25 04:21:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8cf31ed42134cfd8b1386c841e01d95 : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aa853e2cb98d41548281a41f6ea6a584 : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ed8fb93e3f5e4177a628cfdfd9d132a6 : SUCCESS in 6m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e8ed650a4d3f44188e5947bca3d677df : SUCCESS in 6m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dace086b88ea4061aa5d4fc996701da5 : SUCCESS in 7m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7bfc34ab7be6414aa35fa9d1d23b8643 : SUCCESS in 5m 34s","accounts_in_message":[],"_revision_number":2},{"id":"aaa06ebde3fefe1aa0720923bf5bf67d97e0094c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-28 10:49:44.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c17b2112425dedd1db95080e8031fb85582efdf3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30449,"name":"jinyuanliu","email":"liujinyuan@inspur.com","username":"jinyuanliu"},"date":"2021-06-28 10:53:42.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"ef56486a578ac6faa5410f8eb645eb79ec206fb8","author":{"_account_id":30449,"name":"jinyuanliu","email":"liujinyuan@inspur.com","username":"jinyuanliu"},"date":"2021-06-28 10:54:26.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d2a117fe26b7dc0f5ed491c124e4a797061bdca7","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-06-28 11:13:51.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\ntrivialfix","accounts_in_message":[],"_revision_number":3},{"id":"ba4f210c647cbe818e55711daec321aa667764a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 12:12:59.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/2570a6ba2a704bfabfa3cbc877061021 : SUCCESS in 1h 11m 42s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/397ab2e314604df68d9dadff8977fd06 : SUCCESS in 1h 04m 52s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/4408ee5058d54492bf27f3aa51904567 : SUCCESS in 1h 17m 42s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c7b4ad992518442b9eb2a4969c8c9cc4 : FAILURE in 21m 54s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e95032b84654410789b535837e6feebd : SUCCESS in 6m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fe91f9e1bd7146ada3570eee5ed6c7b4 : SUCCESS in 6m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0cf68b5243534386a5d5e90898fdcb11 : SUCCESS in 7m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/54990b43e42a4326a6392455cbba6539 : SUCCESS in 6m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf601b3b9e64466ab07c31248438ecc8 : SUCCESS in 7m 29s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2d8cce01a4e6455d9ee9015966c65ba9 : SUCCESS in 55m 47s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c0e12332f8c140e29d4ae081e139032e : SUCCESS in 53m 29s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/79ddcbc927084c4eb6c4d8557cdfb4c1 : SUCCESS in 59m 44s","accounts_in_message":[],"_revision_number":3},{"id":"d5323413648f09eb982426317cc768e0e016399e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 12:13:16.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"63c828d3003b1e1ea73fbdc7f3c1b6691702e096","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 14:08:29.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/fb83d7153c5b409ea8cde43f65da6fb4 : SUCCESS in 1h 05m 02s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/7bfeeb70ce8547ddbde764cd2a2d0689 : SUCCESS in 1h 31m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8abfaf60ea1e40a69407caf9810945c6 : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dd2892d945f9469a855481f0a43c3bd0 : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c7a8fb95dbb5497bbfc296593e5e2775 : SUCCESS in 6m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/089903c753644d5eb84b58088355a4db : SUCCESS in 8m 18s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/70b64e29286647438104958c579ffc47 : SUCCESS in 49m 55s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a67d4040f6bc46208ca84574fe04f14f : SUCCESS in 1h 02m 24s","accounts_in_message":[],"_revision_number":3},{"id":"a9dad4318335f3d2c30e5082144b1f03e8791e5f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 14:08:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"7286d8b71a772915886f59561a77c2cac7efc187","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 14:10:14.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bd40c54f922430a941a22506f24b64f : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":3},{"id":"974e70c761d3029c3946bb0397754b57b14393bc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-08 16:36:42.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/86f796780cd643ab8c53a4c241743b41 : SUCCESS in 3h 07m 55s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2949bd4e86344866b2d7d541408db19a : SUCCESS in 8m 14s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4507e7a555bf47ff857f6c64df83a06e : SUCCESS in 8m 23s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/e8b04549c7004099bcb0f3559eb332f4 : SUCCESS in 2h 28m 22s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"3b5d8ccd658a5d1793b4bf18717c466ffca572c6","revisions":{"7e2100d1c1e8aa4dc067b600014b04f56a4065b0":{"kind":"REWORK","_number":1,"created":"2021-06-25 03:21:22.000000000","uploader":{"_account_id":30449,"name":"jinyuanliu","email":"liujinyuan@inspur.com","username":"jinyuanliu"},"ref":"refs/changes/33/798033/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/33/798033/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/33/798033/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/33/798033/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/33/798033/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/33/798033/1"}}},"commit":{"parents":[{"commit":"8e020c5f86e033d9baba2c80c39f800bbd3cf2de","subject":"Merge \"tox: Use allowlist_externals instead of whitelist_externals\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8e020c5f86e033d9baba2c80c39f800bbd3cf2de"}]}],"author":{"name":"jinyuanliu","email":"liujinyuan@inspur.com","date":"2021-06-25 03:20:21.000000000","tz":480},"committer":{"name":"jinyuanliu","email":"liujinyuan@inspur.com","date":"2021-06-25 03:21:16.000000000","tz":480},"subject":"Clean up extra spaces","message":"Clean up extra spaces\n\nI recently reviewed the code, although these errors are not important, they affect the code specification.\n\nChange-Id: Ie89f3628c9dfab683973d845a3812bccac9f5bb8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7e2100d1c1e8aa4dc067b600014b04f56a4065b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7e2100d1c1e8aa4dc067b600014b04f56a4065b0"}]},"branch":"refs/heads/master"},"6737636cb06b8a63b29407ca4674b2c5c9b35387":{"kind":"REWORK","_number":2,"created":"2021-06-25 04:09:51.000000000","uploader":{"_account_id":30449,"name":"jinyuanliu","email":"liujinyuan@inspur.com","username":"jinyuanliu"},"ref":"refs/changes/33/798033/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/33/798033/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/33/798033/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/33/798033/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/33/798033/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/33/798033/2"}}},"commit":{"parents":[{"commit":"8e020c5f86e033d9baba2c80c39f800bbd3cf2de","subject":"Merge \"tox: Use allowlist_externals instead of whitelist_externals\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8e020c5f86e033d9baba2c80c39f800bbd3cf2de"}]}],"author":{"name":"jinyuanliu","email":"liujinyuan@inspur.com","date":"2021-06-25 03:20:21.000000000","tz":480},"committer":{"name":"jinyuanliu","email":"liujinyuan@inspur.com","date":"2021-06-25 04:09:24.000000000","tz":480},"subject":"Clean up extra spaces","message":"Clean up extra spaces\n\nI recently reviewed the code, although these errors are not important, they affect the code specification.\n\nChange-Id: Ie89f3628c9dfab683973d845a3812bccac9f5bb8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6737636cb06b8a63b29407ca4674b2c5c9b35387"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6737636cb06b8a63b29407ca4674b2c5c9b35387"}]},"branch":"refs/heads/master"},"3b5d8ccd658a5d1793b4bf18717c466ffca572c6":{"kind":"REWORK","_number":3,"created":"2021-06-28 10:53:42.000000000","uploader":{"_account_id":30449,"name":"jinyuanliu","email":"liujinyuan@inspur.com","username":"jinyuanliu"},"ref":"refs/changes/33/798033/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/33/798033/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/33/798033/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/33/798033/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/33/798033/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/33/798033/3"}}},"commit":{"parents":[{"commit":"8e020c5f86e033d9baba2c80c39f800bbd3cf2de","subject":"Merge \"tox: Use allowlist_externals instead of whitelist_externals\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8e020c5f86e033d9baba2c80c39f800bbd3cf2de"}]}],"author":{"name":"jinyuanliu","email":"liujinyuan@inspur.com","date":"2021-06-25 03:20:21.000000000","tz":480},"committer":{"name":"jinyuanliu","email":"liujinyuan@inspur.com","date":"2021-06-28 10:53:34.000000000","tz":0},"subject":"Clean up extra spaces","message":"Clean up extra spaces\n\nI recently reviewed the code, although these errors are not important, they affect the code specification.\n\nChange-Id: Ie89f3628c9dfab683973d845a3812bccac9f5bb8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3b5d8ccd658a5d1793b4bf18717c466ffca572c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3b5d8ccd658a5d1793b4bf18717c466ffca572c6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
