)]}'
{"id":"openstack%2Fkolla~798035","triplet_id":"openstack%2Fkolla~stable%2Fussuri~Ib855c9ff1f0d7b9d0ba559359f6e9952571d2504","project":"openstack/kolla","branch":"stable/ussuri","hashtags":[],"change_id":"Ib855c9ff1f0d7b9d0ba559359f6e9952571d2504","subject":"Clean up extra spaces","status":"ABANDONED","created":"2021-06-25 04:27:54.000000000","updated":"2021-06-28 10:54:42.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b3bdff493964ffdfa74f95b992e5e0547020f5e1","_number":798035,"virtual_id_number":798035,"owner":{"_account_id":30449,"name":"jinyuanliu","email":"liujinyuan@inspur.com","username":"jinyuanliu"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-06-25 04:38:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":-2,"date":"2021-06-28 10:51:57.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-25 04:38:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-28 10:51:57.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"2a1c691bf9a51c7f75144bed77067670d9575326","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30449,"name":"jinyuanliu","email":"liujinyuan@inspur.com","username":"jinyuanliu"},"date":"2021-06-25 04:27:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4f8fd7511c64c0e85dd09cb411c6a7d536991646","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-25 04:38:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/472951259287471cbb0e833c41cd57e8 : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7c5d69bd0d254b94af3c24f3e86aa734 : SUCCESS in 5m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7eba1eb9f574400bbd7d6f54175f5c06 : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7b56b6598a84a669deef279e88b8065 : SUCCESS in 7m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7e0959ac200d4ecda876cb6697961283 : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":1},{"id":"7521de1f862a6bd638e94dd17f00afd9ab60f9c5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-06-28 10:51:57.000000000","message":"Patch Set 1: Code-Review-2\n\nBackports should not be submitted until change is merged to master. I don\u0027t think this is worth backporting.","accounts_in_message":[],"_revision_number":1},{"id":"b3bdff493964ffdfa74f95b992e5e0547020f5e1","tag":"autogenerated:gerrit:abandon","author":{"_account_id":30449,"name":"jinyuanliu","email":"liujinyuan@inspur.com","username":"jinyuanliu"},"date":"2021-06-28 10:54:42.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2fcff1e70fcb0efe0d0edab0dbcf16410d159823","revisions":{"2fcff1e70fcb0efe0d0edab0dbcf16410d159823":{"kind":"REWORK","_number":1,"created":"2021-06-25 04:27:54.000000000","uploader":{"_account_id":30449,"name":"jinyuanliu","email":"liujinyuan@inspur.com","username":"jinyuanliu"},"ref":"refs/changes/35/798035/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/35/798035/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/35/798035/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/35/798035/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/35/798035/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/35/798035/1"}}},"commit":{"parents":[{"commit":"f7556324fd3850d3e63320fdd2d4086055fc9c87","subject":"horizon: Correct location of monitoring_policy file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f7556324fd3850d3e63320fdd2d4086055fc9c87"}]}],"author":{"name":"jinyuanliu","email":"liujinyuan@inspur.com","date":"2021-06-25 04:27:16.000000000","tz":480},"committer":{"name":"jinyuanliu","email":"liujinyuan@inspur.com","date":"2021-06-25 04:27:16.000000000","tz":480},"subject":"Clean up extra spaces","message":"Clean up extra spaces\n\nI recently reviewed the code, although these errors are not important, they affect the code specification.\n\nChange-Id: Ib855c9ff1f0d7b9d0ba559359f6e9952571d2504\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2fcff1e70fcb0efe0d0edab0dbcf16410d159823"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2fcff1e70fcb0efe0d0edab0dbcf16410d159823"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
