)]}'
{"id":"openstack%2Fkolla~798212","triplet_id":"openstack%2Fkolla~master~I822d640a251e6ed9f71c76a922513e23e4218418","project":"openstack/kolla","branch":"master","topic":"change-798212-3","hashtags":[],"change_id":"I822d640a251e6ed9f71c76a922513e23e4218418","subject":"Add missing default policy files for debian-binary-horizon","status":"MERGED","created":"2021-06-27 13:03:38.000000000","updated":"2021-07-29 10:49:27.000000000","submitted":"2021-07-29 10:47:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"798212-change-798212-3","meta_rev_id":"c47e3ed4b59e6459a29f2d61d58139b1e8e34ac8","_number":798212,"virtual_id_number":798212,"owner":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-29 10:47:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-07-01 10:27:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-07-29 09:15:20.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2021-07-29 09:15:20.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2021-07-29 10:49:27.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}],"CC":[{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-27 13:13:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-27 13:19:13.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2021-06-27 13:19:13.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-06-30 16:17:26.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-07-22 11:20:23.000000000","updated_by":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"reviewer":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"state":"CC"}],"messages":[{"id":"2f1d9a617953b9b9c61f40f11c10a580db27db21","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-06-27 13:03:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5b233cecfadefe49d15b04e2d4759467d0054ef5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-06-27 13:04:39.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"3ec28084bcaadd37f06ce640f405b7dd941ef66e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-27 13:13:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/869f01378a0c483a8006a2b4bd4049a7 : SUCCESS in 6m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1de0f2da22bb4dfcbd217a2a996b435f : SUCCESS in 6m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/caeac5bcb546400c89820efd77429635 : SUCCESS in 5m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c4a298b24bb24363b37925fcb79f8f50 : SUCCESS in 6m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74d17e18fef24dcaa55b026f192e2d7c : SUCCESS in 8m 14s","accounts_in_message":[],"_revision_number":2},{"id":"c4446b22c02a7a8960e43ba7edd35b3041e020c6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-27 13:14:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/13c4abd1f5cb48bb827a7e2f72b9c9d6 : SUCCESS in 5m 48s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/88be7163fbc740868f7ea3564adaa5a9 : SUCCESS in 5m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a8779fadbc032051347768d99ce83620e50dcd1c","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-06-27 13:19:13.000000000","message":"Patch Set 2:\n\nTrivial fix.","accounts_in_message":[],"_revision_number":2},{"id":"04a6886a0b97af481766a72471d7ac4a08aecdf4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-06-28 05:36:14.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"77e3f21be540dea3f31fb7b11fc99cb7e9bdc7c6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 05:45:41.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bc098b2bd10f43cd88bd0c736638db75 : SUCCESS in 7m 23s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f53be5bdc9df46c8af0b6b4cc0c306b7 : SUCCESS in 7m 38s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f38b8bdfb6f51a638acd1bad68f938b968ea0eab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 05:48:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a71e4b152fa4171997148fc8a78b637 : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/828f55b9124647bd9224fd38ebe63455 : SUCCESS in 6m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d700a3d4174c4a19891f6a2de1dc9574 : SUCCESS in 6m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2960085e95f44532907da938548523d6 : SUCCESS in 6m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/634fad01d90d4c18a763fed30a7ec7bf : SUCCESS in 8m 29s","accounts_in_message":[],"_revision_number":3},{"id":"71975fe2d5e60451076718dc2ebbdbf0e37329e2","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-06-28 06:32:25.000000000","message":"Patch Set 3: Backport-Candidate+1","accounts_in_message":[],"_revision_number":3},{"id":"e086ef9e5b33039526315fd310af40c727c68a78","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-06-28 06:34:43.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"68e241958f10db65ee4237cbc9480e9d363ff2a0","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-06-30 16:17:26.000000000","message":"Patch Set 3: Code-Review-1\n\nreno missing :(","accounts_in_message":[],"_revision_number":3},{"id":"30725d7a2bb35b3bcdf2a8d1f5293e9b163af75b","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-07-01 10:15:45.000000000","message":"Patch Set 3:\n\n\u003e Patch Set 3: Code-Review-1\n\u003e \n\u003e reno missing :(\n\nOk, I will provide, but I must mention that the same fix for ubuntu referenced in commit message didn\u0027t need to have release note, and have +1 also from you :(","accounts_in_message":[],"_revision_number":3},{"id":"d4892d98678286ca9748e2bc3867fef42bd43532","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-07-01 10:18:30.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"287c5144601531ed2383349f86b42732ed6ff662","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-07-01 10:19:05.000000000","message":"Patch Set 4:\n\n\u003e Patch Set 3:\n\u003e \n\u003e \u003e Patch Set 3: Code-Review-1\n\u003e \u003e \n\u003e \u003e reno missing :(\n\u003e \n\u003e Ok, I will provide, but I must mention that the same fix for ubuntu referenced in commit message didn\u0027t need to have release note, and have +1 also from you :(\n\nDone, reno included.","accounts_in_message":[],"_revision_number":4},{"id":"2ee0a226b0292a78cf18e1112e5a3e0e9b770014","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-07-01 10:27:32.000000000","message":"Patch Set 4: Code-Review+2\n\n\u003e Patch Set 3:\n\u003e \n\u003e \u003e Patch Set 3: Code-Review-1\n\u003e \u003e \n\u003e \u003e reno missing :(\n\u003e \n\u003e Ok, I will provide, but I must mention that the same fix for ubuntu referenced in commit message didn\u0027t need to have release note, and have +1 also from you :(\n\nNobody is a machine ;-)","accounts_in_message":[],"_revision_number":4},{"id":"f68be2cc1c0f505cfddbbe49a08cd2a7367129de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 11:21:26.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/b7bc4b464cbc4a958361ce9ec5f95bc3 : FAILURE in 27m 33s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/2dad240083ec477db81e0c1ff8567c6d : FAILURE in 42m 20s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/51f4fd7affa8408bb7642610ed8c72a5 : FAILURE in 24m 33s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/67bd1c245abf489993aa4a6467efea7f : FAILURE in 28m 15s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/5b03a70539a74e9d97758d49560fb1f6 : FAILURE in 23m 59s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/5ccf5891eda4452d9343ae8d18336a15 : SUCCESS in 45m 10s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/d2c81a9f6f5f4340b9c53018d0a78e3d : SUCCESS in 31m 33s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/29423c7cf85e4924ae8734a443133210 : SUCCESS in 57m 50s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8de81d8816744fcf8750d44de7a48ad4 : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4852ab113909425e88bc79585a2a25d8 : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8a9bc8fb8b8f4993b4a81cd3795e0267 : SUCCESS in 4m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4faa843b06004435bb6cc90c3da0da32 : SUCCESS in 4m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b902b71c04fe44ee9fe0a5507d3a78fd : SUCCESS in 6m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e99af32f643440b794e8776ef2865cae : SUCCESS in 4m 42s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/04a8fe16b46e42a99592311ed164de36 : SUCCESS in 30m 44s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a0c861d4118c4e4db3b91c1f77600669 : SUCCESS in 39m 26s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c6ee633d881a41cda0a12f9bb2e33055 : SUCCESS in 57m 48s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d68b1045d08b4d5e97c75bd73dd67b85 : SUCCESS in 55m 21s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ac4e99f90dfc498597cd4c11381457e2 : SUCCESS in 1h 01m 39s","accounts_in_message":[],"_revision_number":4},{"id":"a36e1440e021cd48458b6bbdb3402862c906bc5b","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-07-01 11:24:16.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"f9251f922eeacc88085217a602fcdf1d4fe0de3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 12:35:58.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/45dc369d50324e448dd0b4a5cc04706a : FAILURE in 23m 45s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/a661d8b61b17462eb49976e041b31dba : FAILURE in 41m 42s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/e71e08be695246bfa7b969d1da7104a5 : FAILURE in 24m 24s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/a2ac2fcb490c4fdc94d3d88e39a07efc : FAILURE in 33m 06s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/f29d4fb84c914d8e88469a00db2cfff1 : FAILURE in 28m 24s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9cd4bb361a0e42989c13aa0a7e215640 : SUCCESS in 41m 00s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/95eb394c3e9b4eb08c73a06b3ca0946f : SUCCESS in 22m 12s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ae9f34273871467d8807ee18bc890a42 : SUCCESS in 1h 05m 39s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a2333c5e2f84319b69f866f8ceec684 : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f05eefc3de944a6b93b8b46b1a91f1db : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a32cb633b23e46f0a239b3a5b8ca6a3f : SUCCESS in 6m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aaa37e65065f4bdbb647761c9d792bb9 : SUCCESS in 5m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33e781aed7c646bb8fe677e858f4e6e8 : SUCCESS in 6m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9596b7b2f64d464ba93c01bd88c2a7fe : SUCCESS in 4m 36s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/300c69f89d9c499e95b436adec0abcee : SUCCESS in 23m 13s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/10a8907248bd42bf8250739d03a4b788 : SUCCESS in 31m 45s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7a6c20ca18964179b34b3acd4a3be6c0 : SUCCESS in 58m 53s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/54774cd1d98044439be199b099013907 : SUCCESS in 50m 18s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/54b1fabc10e24bfda3426901c6b17cb0 : SUCCESS in 59m 25s","accounts_in_message":[],"_revision_number":4},{"id":"42d84cdc418f9a8245f55baa9b5da013624d8605","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-07-01 13:22:02.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"32cc356e9ce5d96baa43c2d7a76eac8b204d6689","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 14:44:23.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/c686e76ee358448c94eecfe502963718 : FAILURE in 33m 30s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/fe5fc0563de44907b872087bf3b4f063 : SUCCESS in 40m 26s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/4049cc9d5d59442ab1be9b438603f223 : SUCCESS in 1h 16m 29s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/30e8329f845341acbc5e4e16d93c3e0d : SUCCESS in 1h 01m 02s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/f63fcbf6465b4d4b9b3910561d5d616a : SUCCESS in 1h 17m 18s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/7868a130230c4f599a97d00d2916932c : FAILURE in 35m 16s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/9c65fa741dc24630bef266486c13f581 : SUCCESS in 22m 07s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/67b4eb5fff084a1d8f0433157eac99d5 : SUCCESS in 1h 00m 12s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa9961a2ec2d4902a1cb4f4a9da63ac7 : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5e06f4fb85954e41a4a96efab1928ccf : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/db3a82b6d99f4625ad47a4241aaf1594 : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5b87d20f11134f76a0f5341fea0e6fca : SUCCESS in 4m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d5ab2d3a83e4b6fba552e6b7db2af61 : SUCCESS in 5m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d9b77bb5f3764e72a5529e83958a00e2 : SUCCESS in 4m 28s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/49bc6aef357340fe9cb15c0652372907 : SUCCESS in 23m 14s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/5e0c64a78b1b49248be33538420b6478 : FAILURE in 38m 27s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/220a49525b4f4cd98cf58d07e2ff309a : SUCCESS in 57m 46s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f69821837f354c818edf556d5fb72170 : SUCCESS in 52m 25s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/38e89fa1165a457cbe7a190dd8dcd7dd : SUCCESS in 1h 03m 21s","accounts_in_message":[],"_revision_number":4},{"id":"db033e30295aa0347043dd066030cf6f52a6e6a4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-08 16:42:04.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/18cfbedac3614ea19547ca0e00a671b1 : SUCCESS in 2h 50m 26s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b3edcc351af44dec91da2f28cae98304 : SUCCESS in 6m 08s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3937a8b6470d4d85805740f8c13d1c10 : SUCCESS in 7m 08s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/c83752fb27d842d7bb73ca75585b5082 : SUCCESS in 2h 15m 12s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"51cca43e5673a0c6cf334499faf04214d4d02d92","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-07-09 08:44:09.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"bdfdf74c5eda046992294d5b31212fc8c4cac4f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-09 09:55:29.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/efb7ad1b7cda47cda3347b1047a9d2e2 : SUCCESS in 39m 19s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/ffaaf8d93a444b559a8ef6fee60cfc50 : SUCCESS in 40m 10s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/726e14cd8338490e9ec30510d6925dc6 : SUCCESS in 56m 40s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/c26c744b001d4d4ea01fc5c396359d34 : SUCCESS in 1h 04m 40s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/51c943cdbfcb4056a3b9c2cb64df9ea6 : SUCCESS in 1h 08m 50s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/139a4e037f1941a19c5151dd16dc2d2a : SUCCESS in 31m 45s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/204f7ccf05744c1095fca9e854b7ca36 : SUCCESS in 17m 22s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/9f4c2e24343b474aae86dcde04510ad2 : SUCCESS in 49m 44s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecfde47fd06b4bcf9dd6d92b0ba9bde4 : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e92be6b83a714e539bda587699de6749 : SUCCESS in 6m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fb1ae31fe90144d8b46b19693337d649 : SUCCESS in 4m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e825ca7744e94fc1ba128b153383156f : SUCCESS in 7m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2874d0c6d378440e91839809ac96921a : SUCCESS in 6m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0da860e9c79a4792a4e6ed4c6c298b15 : SUCCESS in 6m 06s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9387273dfcde4b8d98b7ec2268413f21 : SUCCESS in 21m 03s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/94eb1e0352f94433a765efb5dea028bf : SUCCESS in 34m 57s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b676a00e4fb746b78c1bf96d18145e53 : SUCCESS in 53m 23s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/75d97c0f523a41e9aa14eb7d9cb77cac : SUCCESS in 50m 01s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2aace1f8427e4c75bff0d8c40418abb5 : SUCCESS in 1h 01m 43s","accounts_in_message":[],"_revision_number":4},{"id":"3f17f250203a0c1889cf9e837ab4cd6e4e70a395","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-09 10:09:13.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/aa492d3816024c37b9f5320e2be9b7ef : SUCCESS in 51m 38s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0575a0bb9ab740e2abceef65d3df9d86 : SUCCESS in 40m 32s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0f2f20f46c464f709950c19d9129a653 : SUCCESS in 1h 17m 19s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ec3907ff5641477b806a398f56f61048 : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4390b279c1f44c5b8a546e5282bbb990 : SUCCESS in 5m 33s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/48a175f9b669449db2e2e0f36246047a : SUCCESS in 40m 43s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"bde38556325e49e49a128ee5958d3f96f792da4e","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-07-29 09:15:20.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"b3ae9dda92978fc4e2301625ff7fecd39c150067","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 09:15:46.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"283cca5f95ddd89cbc0f191209bb1470091e4499","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 10:47:42.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/46fd9390afa049eaa8d14372e18c0a3e : SUCCESS in 35m 36s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/5ccfb53bdd854f8990aa3902cfed5f32 : SUCCESS in 1h 09m 11s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/eafd7a06b75440379239f0a6d0e2c849 : SUCCESS in 1h 25m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/557f7d943ad54b1b8dcd878275ac23a5 : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/40cecaec7fea4c0a9f8833f874b997e0 : SUCCESS in 5m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8f4fe2e64db34d66a954915101e39e5e : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbe522d927784e64919fed0b7270979e : SUCCESS in 7m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6b8642278afb4f54b8305fd0584400f3 : SUCCESS in 4m 58s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d348332184a44ec1a216c0d9c4271b91 : SUCCESS in 37m 05s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/56265b40f3154b109552f47986030c2b : SUCCESS in 54m 02s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c15f6ebc5a6c49fdb1686797ec1ce937 : SUCCESS in 1h 09m 52s","accounts_in_message":[],"_revision_number":4},{"id":"7e6b7a188777210d653c6eb222df43edaddab1f3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 10:47:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"4cdf2d834b9d82a51395727d99e619e5c47b4feb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 10:49:27.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57e3e12b07254fc4b1edefa97882d778 : SUCCESS in 1m 11s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/48a01bf1963c4d68964c3b8a5c12cf4b : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"01d02be25ed043b4541db3ebfc102d013cdc578a","revisions":{"24f8432acd39562fa0d4e4f521e1922b8e122009":{"kind":"REWORK","_number":1,"created":"2021-06-27 13:03:38.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/12/798212/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/12/798212/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/798212/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/798212/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/798212/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/12/798212/1"}}},"commit":{"parents":[{"commit":"8e020c5f86e033d9baba2c80c39f800bbd3cf2de","subject":"Merge \"tox: Use allowlist_externals instead of whitelist_externals\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8e020c5f86e033d9baba2c80c39f800bbd3cf2de"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-06-27 12:56:46.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-06-27 13:03:36.000000000","tz":120},"subject":"Add missing default policy files for debian-binary-horizon","message":"Add missing default policy files for debian-binary-horizon\n\nThe debianu-binary-horizon image is missing the default policies files\nwhich should be in /etc/openstack-dashboard/default_policies. By copying\neverything from /etc/openstack-dashboard/policy\ninto /etc/openstack-dashboard, we get the default policy files, as well\nas commented out policy files (e.g. cinder_policy.yaml) and the\nnova_policy.d directory containing api-extensions.yaml.\n\nChange was merged for ubuntu in [1] but unfortunatelly\nnot for debian, this trivial patch fixes it also\nfor debian.\n\n[1] https://review.opendev.org/c/openstack/kolla/+/794589\n\nCloses-Bug: #1933759\nChange-Id: I822d640a251e6ed9f71c76a922513e23e4218418\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/24f8432acd39562fa0d4e4f521e1922b8e122009"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/24f8432acd39562fa0d4e4f521e1922b8e122009"}]},"branch":"refs/heads/master"},"0b0ec2c962f7bf129cdc1b74a0a89ff2059471d0":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-06-27 13:04:39.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/12/798212/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/12/798212/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/798212/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/798212/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/798212/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/12/798212/2"}}},"commit":{"parents":[{"commit":"8e020c5f86e033d9baba2c80c39f800bbd3cf2de","subject":"Merge \"tox: Use allowlist_externals instead of whitelist_externals\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8e020c5f86e033d9baba2c80c39f800bbd3cf2de"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-06-27 12:56:46.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-06-27 13:04:18.000000000","tz":120},"subject":"Add missing default policy files for debian-binary-horizon","message":"Add missing default policy files for debian-binary-horizon\n\nThe debian-binary-horizon image is missing the default policies files\nwhich should be in /etc/openstack-dashboard/default_policies. By copying\neverything from /etc/openstack-dashboard/policy\ninto /etc/openstack-dashboard, we get the default policy files, as well\nas commented out policy files (e.g. cinder_policy.yaml) and the\nnova_policy.d directory containing api-extensions.yaml.\n\nChange was merged for ubuntu in [1] but unfortunatelly\nnot for debian, this trivial patch fixes it also\nfor debian.\n\n[1] https://review.opendev.org/c/openstack/kolla/+/794589\n\nCloses-Bug: #1933759\nChange-Id: I822d640a251e6ed9f71c76a922513e23e4218418\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0b0ec2c962f7bf129cdc1b74a0a89ff2059471d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0b0ec2c962f7bf129cdc1b74a0a89ff2059471d0"}]},"branch":"refs/heads/master"},"b2570baf28b6bddc342c7aa6af37db91428c9f89":{"kind":"REWORK","_number":3,"created":"2021-06-28 05:36:14.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/12/798212/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/12/798212/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/798212/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/798212/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/798212/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/12/798212/3"}}},"commit":{"parents":[{"commit":"8e020c5f86e033d9baba2c80c39f800bbd3cf2de","subject":"Merge \"tox: Use allowlist_externals instead of whitelist_externals\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8e020c5f86e033d9baba2c80c39f800bbd3cf2de"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-06-27 12:56:46.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-06-28 05:36:04.000000000","tz":120},"subject":"Add missing default policy files for debian-binary-horizon","message":"Add missing default policy files for debian-binary-horizon\n\nThe debian-binary-horizon image is missing the default policies files\nwhich should be in /etc/openstack-dashboard/default_policies. By copying\neverything from /etc/openstack-dashboard/policy\ninto /etc/openstack-dashboard, we get the default policy files, as well\nas commented out policy files (e.g. cinder_policy.yaml) and the\nnova_policy.d directory containing api-extensions.yaml.\n\nChange was merged for ubuntu in [1] but unfortunatelly\nnot for debian, this trivial patch fixes it also\nfor debian.\n\n[1] https://review.opendev.org/c/openstack/kolla/+/794589\n\nCloses-Bug: #1933759\nChange-Id: I822d640a251e6ed9f71c76a922513e23e4218418\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b2570baf28b6bddc342c7aa6af37db91428c9f89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b2570baf28b6bddc342c7aa6af37db91428c9f89"}]},"branch":"refs/heads/master"},"01d02be25ed043b4541db3ebfc102d013cdc578a":{"kind":"REWORK","_number":4,"created":"2021-07-01 10:18:30.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/12/798212/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/12/798212/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/798212/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/798212/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/798212/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/12/798212/4"}}},"commit":{"parents":[{"commit":"8e020c5f86e033d9baba2c80c39f800bbd3cf2de","subject":"Merge \"tox: Use allowlist_externals instead of whitelist_externals\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8e020c5f86e033d9baba2c80c39f800bbd3cf2de"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-06-27 12:56:46.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-07-01 10:18:27.000000000","tz":120},"subject":"Add missing default policy files for debian-binary-horizon","message":"Add missing default policy files for debian-binary-horizon\n\nThe debian-binary-horizon image is missing the default policies files\nwhich should be in /etc/openstack-dashboard/default_policies. By copying\neverything from /etc/openstack-dashboard/policy\ninto /etc/openstack-dashboard, we get the default policy files, as well\nas commented out policy files (e.g. cinder_policy.yaml) and the\nnova_policy.d directory containing api-extensions.yaml.\n\nChange was merged for ubuntu in [1] but unfortunatelly\nnot for debian, this trivial patch fixes it also\nfor debian.\n\n[1] https://review.opendev.org/c/openstack/kolla/+/794589\n\nCloses-Bug: #1933759\nChange-Id: I822d640a251e6ed9f71c76a922513e23e4218418\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/01d02be25ed043b4541db3ebfc102d013cdc578a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/01d02be25ed043b4541db3ebfc102d013cdc578a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
