)]}'
{"id":"openstack%2Fkolla~799065","triplet_id":"openstack%2Fkolla~stable%2Fvictoria~I19f57d258021ff660df6b6b95e8205b69f3bea85","project":"openstack/kolla","branch":"stable/victoria","hashtags":[],"change_id":"I19f57d258021ff660df6b6b95e8205b69f3bea85","subject":"CI: Fix epel url override","status":"MERGED","created":"2021-07-01 14:51:14.000000000","updated":"2021-07-02 10:17:13.000000000","submitted":"2021-07-02 10:15:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"799065","cherry_pick_of_change":798995,"cherry_pick_of_patch_set":3,"meta_rev_id":"dbfc2b46305017e7644802f7ccfee3a932d3b6fb","_number":799065,"virtual_id_number":799065,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-02 10:15:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-07-02 09:07:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-07-01 14:59:15.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-07-02 09:07:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-01 14:51:14.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-07-01 14:59:15.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2021-07-01 16:12:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0640dbf71fef0271e93c2618bfef47d571d8b316","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-07-01 14:51:14.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"efdf9b7f0f02d25d655b59903c45c34eceffcb60","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-07-01 14:59:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ece66d1a6ee6aa418cb77032a7df67b85e4f1a3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 16:12:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/c0cfdfe71910424686116c942f9e44f8 : SUCCESS in 1h 08m 38s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/36b0908e68ab49e1be42ed29dd1a6e86 : SUCCESS in 1h 11m 02s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/fa6c0503f3914173bd1f9b7943d49f74 : SUCCESS in 1h 12m 24s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/91b65d85a11346589d3304d9ac912348 : FAILURE in 16m 42s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c1092c9c85c4839b9b7f52962508f31 : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6529359324eb466c80daae4500f26931 : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/37f5119dab3b49fca0826fea5e5f0b20 : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34446bee15c74a3dad4e130096196f99 : SUCCESS in 5m 47s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/80cad2bd9318412d9b3e6e29cdfcdc2d : FAILURE in 24m 30s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2ad105c860724b4d8df5c87c04dd0445 : SUCCESS in 50m 37s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f909dd695bb8482fb526ed65cb630e76 : SUCCESS in 1h 09m 14s","accounts_in_message":[],"_revision_number":1},{"id":"8a2af0be807c6168500ae706bc2a7eee1d28ea7e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 17:12:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4ad7cd339d3f4a87a77377c64c478532 : SUCCESS in 2h 19m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6daf6a30f927d4ef81ba27acaf04cfeda92ae27e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-07-02 07:16:38.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ae7cf4071a11444a1c159fb9c4f8fbd2d6116373","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 08:42:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/0c209c454b49408e81f46b3ba082302f : SUCCESS in 1h 01m 02s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/c9ecd2d21d6142418e8457fcbb8b55cc : SUCCESS in 1h 16m 45s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/5a71302417994238b755b64b17bb0752 : SUCCESS in 1h 12m 48s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/1bb8c68098da48cead3a2e07f2bd97a2 : SUCCESS in 1h 17m 48s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce2ca64b154b42c8b6e69e0229b05884 : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0b423f5cb14a450c92da712a6cef994a : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a4fed7b637fd4f9f85cc132517e25177 : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c09957fd5730445bb76648e16ac0fcfe : SUCCESS in 5m 47s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/07a3cb38c3574abd837d74bc88cb3a80 : SUCCESS in 1h 05m 19s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b90b690839434bf489369dfd628d27ae : SUCCESS in 1h 06m 10s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d1f0ca0a697440fe9ef84a0d837a9dcc : SUCCESS in 1h 05m 57s","accounts_in_message":[],"_revision_number":1},{"id":"c5b80252c17a4fc14bce62a6512346a1b74aecef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 09:03:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3bbd336972824fe493f94a1c319e80e9 : SUCCESS in 1h 45m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c3a951be0befec1011bd4dc39d72e6c0102eb551","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-07-02 09:07:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"098147d27600cc13aa208e9f9af44d929a78da12","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 09:07:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dd7554fa5b088c319d9de384175965656d078124","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 10:15:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/ac7449cb58ee4f7fbc332729d8fec977 : SUCCESS in 1h 01m 55s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/cc3c76d925d24a71b99bd2c53bdbea0d : SUCCESS in 1h 03m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2817d922eafc4dcb87fc9cfea993be20 : SUCCESS in 6m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0b740b4ad6564d01a203cb284299f761 : SUCCESS in 6m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e1acc87a9d4e46a6b5d4cc2d7ade949f : SUCCESS in 7m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3023346189404e7c9365e9711d3dba80 : SUCCESS in 10m 22s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a1d0884cff174acca014263cfe107069 : SUCCESS in 1h 06m 01s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a7269a98b61846c5a5f0113ca355e7fa : SUCCESS in 1h 04m 23s","accounts_in_message":[],"_revision_number":1},{"id":"9f34c085ef054181fac4005fbf0facdc6a2c693f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 10:15:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"dbfc2b46305017e7644802f7ccfee3a932d3b6fb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 10:17:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/101a53fe347a41f4b74edf02fb02aadb : SUCCESS in 1m 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9fde574c6ae3f137172def7fd94d8baa57371b7c","revisions":{"9fde574c6ae3f137172def7fd94d8baa57371b7c":{"kind":"REWORK","_number":1,"created":"2021-07-01 14:51:14.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/65/799065/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/65/799065/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/65/799065/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/65/799065/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/65/799065/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/65/799065/1"}}},"commit":{"parents":[{"commit":"32eb32f1b66416a2ccd385bd4afee765bf1eb712","subject":"horizon: Correct location of monitoring_policy file","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/32eb32f1b66416a2ccd385bd4afee765bf1eb712"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-07-01 07:19:01.000000000","tz":120},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-07-01 14:51:14.000000000","tz":0},"subject":"CI: Fix epel url override","message":"CI: Fix epel url override\n\nepel-release now installs epel*.repos with following baseurl:\nbaseurl\u003dhttps://download.example/pub/epel/$releasever/Everything/$basearch\n\nChange-Id: I19f57d258021ff660df6b6b95e8205b69f3bea85\n(cherry picked from commit acc8cb9903aa2a97342d1ca76bc9e274361194c1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9fde574c6ae3f137172def7fd94d8baa57371b7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9fde574c6ae3f137172def7fd94d8baa57371b7c"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
