)]}'
{"id":"openstack%2Fkolla~799782","triplet_id":"openstack%2Fkolla~master~I055086ab4af37878e81844f1055bc48d84ae9945","project":"openstack/kolla","branch":"master","topic":"bug/1934788","hashtags":[],"change_id":"I055086ab4af37878e81844f1055bc48d84ae9945","subject":"Fix missing pacemaker-cli-utils in Debian hacluster images","status":"MERGED","created":"2021-07-07 10:08:48.000000000","updated":"2021-08-02 12:10:16.000000000","submitted":"2021-08-02 12:08:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":38,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"799782-bug/1934788","meta_rev_id":"5ef10dc83a83288b0f7342a3abd8ed1ce89fb37a","_number":799782,"virtual_id_number":799782,"owner":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-02 12:08:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2021-08-02 10:39:16.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2021-07-30 11:36:29.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2021-08-02 10:39:16.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-07 11:48:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-08 09:01:37.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-07-22 12:52:09.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2021-07-29 09:12:08.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"CC"},{"updated":"2021-07-30 11:01:32.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"}],"messages":[{"id":"61bad1bbf0656c3c356fd1112145b9679ed76ed4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-07-07 10:08:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dc5b78638fad68c03a36af3cc755658b69839cb2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-07 11:48:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/45d6fe75114c4e958e5e10dd1df73ae1 : SUCCESS in 35m 42s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/998d37b6b754499eaaf8bd6a1cd4402b : SUCCESS in 58m 41s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/54dae46ff4c64b62b90ac5e10c8b1068 : SUCCESS in 1h 02m 17s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/97c81400110245c1a72104d331cb51c4 : SUCCESS in 1h 05m 08s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/86e2d862ba1b4d1ba97ef5c824fc7ae8 : SUCCESS in 1h 22m 50s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/3101f8c7aa7a458c87707461b5606397 : SUCCESS in 40m 52s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/228f5cd20d214d5b937a8214f409cf64 : SUCCESS in 42m 01s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/190733b3833b4fbb9138e701dc4f8719 : SUCCESS in 1h 04m 50s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/380903ed3d3649a7abec9c7a2644af61 : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3f0297b2201d48e0b03294168bdc7537 : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/94a122c60eff4dc284ffa241b1c4a113 : SUCCESS in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/853f84b2fd054d9bb8a0a72256fc8124 : SUCCESS in 4m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7f9b6e98ee1450e8930fb86f6e4ac33 : SUCCESS in 5m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/45740ef6d1514d8984f1ff28a1b14ff5 : SUCCESS in 4m 36s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f62800fb95084bc0822d3554a1ec97cf : SUCCESS in 28m 59s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/195a021bf0c1460ca2d5703d44c7d233 : SUCCESS in 40m 50s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/76b01dc3d415479bbd0017b3cd9bcf05 : SUCCESS in 1h 06m 23s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a3aeca82ab154e2ba917e3b621e7b1e3 : SUCCESS in 52m 05s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b56ba8100f424ce0af2a6d3b76eca76a : SUCCESS in 58m 12s","accounts_in_message":[],"_revision_number":1},{"id":"6e5825f9345eea9f11426eaed32b6bae1d7c6e2c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-07 11:53:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/ed085fc45ff9437a81874c61a7d07e4b : SUCCESS in 1h 31m 03s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4ca3feec7a9a4c8d8fb384ac56addb31 : SUCCESS in 46m 53s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f3b8df0e32c140b28d2a4e30d7ea7003 : SUCCESS in 1h 36m 16s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fe02512487b94c18ad22daff9c8a7815 : SUCCESS in 6m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/46fc444c6e4c4329963a560f72d71aa0 : SUCCESS in 6m 31s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/28ca610698524a45b198fa3002d9d03f : SUCCESS in 53m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0239f52ac6ecfb687507e5ac2b6e7d5da27cb29d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-07-08 09:01:37.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"46f372f8cd34c78775b4cf059c8f82a00974857d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-07-08 12:26:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"02d7b82add9384bb8687b2eb40b975b2912606a7","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-07-08 12:27:42.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"4c0d4233f863bee2c29b0b925a57fd5d3cb6c761","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-08 13:39:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/fd0a6a628aa746818932dee542fbc863 : SUCCESS in 26m 35s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/8703c87f60af4e1db2a87b708caf3399 : SUCCESS in 31m 49s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/d1f696d427784caab55adc5f8c16c9d9 : SUCCESS in 1h 10m 49s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/4ea3671af0654438b36663092e02e6f4 : SUCCESS in 1h 05m 17s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/9721a40bca0647a5829d2f3944864241 : SUCCESS in 1h 09m 55s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f7d3fa38d2724bd69580f789658e11d5 : SUCCESS in 29m 29s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/d7656fc7c88547ad91985cc2be3313f4 : SUCCESS in 20m 28s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/85bbeb0ab24b4d4aa6b53008581a4f44 : SUCCESS in 52m 18s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0800fd5ab574c11a10a80b21b06dd6e : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/160206827cf14c73b0e68cba3246a237 : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/535a0894c61f4cb58ac0133907eaff08 : SUCCESS in 5m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/918874a797da41c49f97422d06760115 : SUCCESS in 4m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed136a2a6dd644e9bdf97971b11d1106 : SUCCESS in 7m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cd517c62202b417cb84f4887d1326cf5 : SUCCESS in 4m 27s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b1238f46f63140fd83694eccea21ee16 : SUCCESS in 19m 57s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/fa7c26b6991d4f6697ab46afc39f660c : SUCCESS in 29m 35s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2a3211ffc9e24166970f2670320efdf8 : SUCCESS in 58m 11s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/df967eae130d4614b34bb26a69225703 : SUCCESS in 53m 42s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1d49553e7dfe48479dcc2cb7d4df6970 : SUCCESS in 1h 02m 04s","accounts_in_message":[],"_revision_number":2},{"id":"3488a4d751c9d7ebf69d4b8690614ea8ba788e88","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-08 14:08:07.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/b676b86bb3554089a48dea0a20abf2fc : SUCCESS in 1h 39m 15s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9e0b6529906f45e8b45db2c7d0c54fc9 : SUCCESS in 38m 34s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2f477d0366f449c6b3060906c5250ced : SUCCESS in 1h 19m 07s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/28cffbd7f03e40d0be4229eb375791f0 : SUCCESS in 6m 47s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5b8c6df2e1294691a2a307b11781cc85 : SUCCESS in 6m 30s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/0e881d3a982b4d7ab9e38dee05933b26 : SUCCESS in 1h 19m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"be0bb4e95ffdd2fa76b64bc8a635c6a319e00178","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-07-22 09:08:02.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"970962805a5a2e71a6d77751913348f1e8f987c9","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-07-22 12:52:09.000000000","message":"Patch Set 2: Code-Review-1\n\n(11 comments)","accounts_in_message":[],"_revision_number":2},{"id":"d60ff92c3454749ea7fb7e5bfc307e68f87bdcfd","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-07-29 09:12:08.000000000","message":"Patch Set 2:\n\nFix Radek\u0027s comments and +2 will be given","accounts_in_message":[],"_revision_number":2},{"id":"810e466a08200d7804631a0703b1dab50512d7c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-07-29 09:39:17.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6ce702764ff5f3b4a6d8c922de4ba6652d588ea9","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-07-29 09:39:17.000000000","message":"Patch Set 2:\n\n(11 comments)","accounts_in_message":[],"_revision_number":2},{"id":"11699e7c5398760b85beae300a6c52f4a6ffa08d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 10:58:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/45b1041a16c0485db91ac1e2a38070ef : SUCCESS in 27m 43s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/73ee1a544e5349f99cdcbdf378a0dc57 : SUCCESS in 44m 04s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/6cb8a8646dcb41b080171d705c3abd7f : SUCCESS in 1h 12m 03s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/49f9664246f64eec88a91562671a2afa : SUCCESS in 1h 04m 17s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/7d72228c022549ae90edcaa83ed4fabf : SUCCESS in 1h 08m 00s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9f60f5dc49df44c8bcf344be521c4c6a : SUCCESS in 33m 54s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/42df81bb44cf4ccfb037675e7aaf7ad2 : SUCCESS in 25m 39s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/3714ee6496e44f508c1c40263b0cf68e : SUCCESS in 1h 04m 22s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b58a9d60707440aa1554fa508b33870 : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a23963fdf58c4794b7cd47b26bacac49 : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4846ec68343f4d02b0bc7c62c51fe0e5 : SUCCESS in 5m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a003409bbbab46dfaff974ec8f0df64b : SUCCESS in 5m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47efa30b8e96456b954ed8abb49c6a65 : SUCCESS in 6m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/52f333477a0a46649c60f3e2ccd141b6 : SUCCESS in 4m 51s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0c8c16dfb9354e51a473e90a304a0e72 : SUCCESS in 23m 11s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/938fa27f0fa7441d9d4cebe94977f1de : SUCCESS in 27m 27s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/191b8c0bd575412c8e411cbe74a04e73 : SUCCESS in 49m 15s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f049413fae9d46578f409d6645ab22b9 : SUCCESS in 53m 17s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f6461323033b4d92b49589cf5670b402 : SUCCESS in 1h 04m 26s","accounts_in_message":[],"_revision_number":3},{"id":"498280c816fa719597915ba34275c2c4a9122f42","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 12:51:31.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/68a28781b475456cbfb0604991a19bba : SUCCESS in 2h 40m 58s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1a9f80793f584b2d8a070e1c223de2da : SUCCESS in 2h 33m 23s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/135f13e6126b4b86bde0ebd71347138b : TIMED_OUT in 3h 02m 05s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1c5b70b4e6d143e98aea1f5e74fa8f47 : SUCCESS in 6m 45s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5d0adf25ac5b45a994a267c6dbf4783e : SUCCESS in 10m 19s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/5eef77bbc8fe489ab25e02b17900da50 : SUCCESS in 2h 32m 58s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"58e3d9c6a3375df2432575ade7ebe24232c1a5f6","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-07-30 11:01:32.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"231f0f99ff5797e1a3bec38a48bdf4c143361f67","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-07-30 11:08:41.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b8c1a5ecdb6fa4b364c7e46bef81c1a022854036","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-07-30 11:08:46.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1e4a22e3b722db48f08c25138b1c048aa6c50791","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-07-30 11:10:57.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0e3f1f284a7ce5853d50d4e740aacebb6fa5aa95","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-07-30 11:11:17.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"04efa1dbc03309b1a51eaf394550127473a623c6","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-07-30 11:12:03.000000000","message":"Patch Set 4: Code-Review-1\n\nplease write the reno note and fix other issues as I commented on PS3","accounts_in_message":[],"_revision_number":4},{"id":"ec214783d8f01e40f80e011b9f5e3746c168ef9a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-07-30 11:17:41.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"5e5e0179b0a740512903b0459c330b4bd3d18327","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-07-30 11:17:49.000000000","message":"Patch Set 4:\n\n\u003e Patch Set 4: Code-Review-1\n\u003e \n\u003e please write the reno note and fix other issues as I commented on PS3\n\nDone","accounts_in_message":[],"_revision_number":4},{"id":"dfc63ccf2c3fd50cd12ebbcc7fb10e0bc1a653ad","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-07-30 11:18:11.000000000","message":"Patch Set 5:\n\n(4 comments)","accounts_in_message":[],"_revision_number":5},{"id":"d43b3b315826770a55615378fabd4a630ba02087","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-07-30 11:19:16.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"685fe9f9d841bfb29d9dc4c803e7a8eab9efefdf","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-07-30 11:24:40.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"fe6b0272f8d4e219f29d0c226d3789bfb9e9e367","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-07-30 11:26:46.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"30d4aed0f3302529ed32250f0bb6ef0117d9fcc4","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-07-30 11:26:58.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"5df3881d6bea5b154ec2ba3cdf41f24bc4aee86c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-07-30 11:36:29.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"7eae23cbe8286177a8556c6c21ed2deeef396492","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-30 12:50:27.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/650248c9e82640c8b6650d89c9bfb27c : SUCCESS in 20m 30s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/fa619964281b4cb1862b6856018f0304 : SUCCESS in 52m 37s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/41d979f42711419488414b8efe3950c0 : SUCCESS in 1h 07m 30s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/edc66650a1b2462e9faaedb74dc80dce : SUCCESS in 1h 22m 12s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/769b57a22a1948389a1fda05509b6dc9 : SUCCESS in 1h 16m 27s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/613913467fbd4d5684069e559056d70d : SUCCESS in 25m 17s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/95115e467b214c249584e1d5dba9b16f : SUCCESS in 18m 59s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/42863fdc20414feca23b814a8607409d : SUCCESS in 49m 49s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28f6e45d50d14356b48626e692cac941 : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b8f2df7290764bf3aceaaa845df1a2e6 : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/366b6acdb9a748588e1f1a87c0cca435 : SUCCESS in 5m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e481e07882994ca0bfb1ec4eb709feba : SUCCESS in 4m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a889af98467c4d538b4e319d6c9e9371 : SUCCESS in 6m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e42d5e7f1cc84e8fb304c15d69b50ea6 : SUCCESS in 4m 32s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/85aa1bd471444a6cb526d6815f114730 : SUCCESS in 19m 51s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a37746b482f9454aa020fe470530a4a1 : SUCCESS in 27m 35s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c64caa8f7f5848999bc64b2256e6fd51 : SUCCESS in 53m 23s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/acef2ff43aa14ba4a6aa1990942270cf : SUCCESS in 47m 44s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e2427ac69b194806a2b3277ec3bef02d : SUCCESS in 1h 04m 59s","accounts_in_message":[],"_revision_number":6},{"id":"00d04c0b7c700c77ec579b47a8a76e9e344463e7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-30 12:53:45.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/6cdb7d5ebd6740c5be97f848833b017b : SUCCESS in 44m 20s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/fcfb65d7fa2f4bcf970131ec546395d3 : SUCCESS in 38m 47s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/59cba807fcf545049b61e2b3a9c99b34 : SUCCESS in 1h 19m 43s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ac464e9fb92a4e9980abcb560b653d86 : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bd68c988ef4845d599d4e9da83cde1f2 : SUCCESS in 5m 58s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/7b8ca46a4342408e9363de95d9c66f3a : SUCCESS in 39m 01s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"26b83d6699196e61fe1052447bab4a181e34f19e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-08-02 10:39:16.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"50e3ae951bc2e51a412a6dcc0884ae9709a1c5a2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-02 10:39:34.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"ad943240b739e59e891139d5e33ad2ebeb17ea07","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-02 12:08:55.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/416e2202c598447abe5dc7679b3049fe : SUCCESS in 39m 01s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/c015d12ca0c842508273618ab84ad01b : SUCCESS in 1h 08m 19s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/2022c02b003e4356bc7e05a69f048baf : SUCCESS in 1h 11m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c61bf9ca0dc43308a575d3bd42bcf78 : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3d68fe5130c54c859bf685ec81a695a5 : SUCCESS in 6m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eb7059d362bb4461a16da89e582720ee : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3bd5de4bfc04df6ab60d012d49f44eb : SUCCESS in 7m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/928cd5940c1a47fe9ce7d1d8da703161 : SUCCESS in 4m 18s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/563f45307b03464d916d9b973678ecd0 : SUCCESS in 28m 27s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7a3b3d137f3445b48834b1f696a4338a : SUCCESS in 50m 57s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c2a335c9abcf4829a85dd38fa7f27f68 : SUCCESS in 1h 22m 58s","accounts_in_message":[],"_revision_number":6},{"id":"37b8a9c6aa349d8c8c22c7798ac3f284ed6cc04c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-02 12:08:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"5ef10dc83a83288b0f7342a3abd8ed1ce89fb37a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-02 12:10:16.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fa9d94e6a3049429da04276963dce1a : SUCCESS in 1m 02s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/38994829a81e4231a36360b35dba119d : SUCCESS in 58s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"535c861d99aa0c07ea130d1f30cabb6af3db7ab7","revisions":{"0a13d28c79dcdecb674a46790d643269aa4e89bb":{"kind":"REWORK","_number":1,"created":"2021-07-07 10:08:48.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/82/799782/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/82/799782/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/799782/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/799782/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/799782/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/82/799782/1"}}},"commit":{"parents":[{"commit":"5e0b238b01250392fa9f92c069a7253902d6b325","subject":"Merge \"Bump up cAdvisor\u0027s version\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5e0b238b01250392fa9f92c069a7253902d6b325"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-07-07 09:41:04.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-07-07 10:08:44.000000000","tz":120},"subject":"Fix missing pacemaker-cli-utils in debuntu hacluster images","message":"Fix missing pacemaker-cli-utils in debuntu hacluster images\n\nHacluster images need to have pacemaker cli utils available\nin Debian/Ubuntu and they are provided by pacemaker-cli-utils\npackage. This package is also *recommended* dependency of pacemaker\nand pacemaker-remote in both Debian/Ubuntu systems, but\nunfortunatelly it is not installed via APT by default.\n\nThis patch adding this package to list of package to be installed\non Debian/Ubuntu systems.\n\nCloses-Bug: #1934788\nChange-Id: I055086ab4af37878e81844f1055bc48d84ae9945\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0a13d28c79dcdecb674a46790d643269aa4e89bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0a13d28c79dcdecb674a46790d643269aa4e89bb"}]},"branch":"refs/heads/master"},"cd461f273c55365190bbece7bcef6aebe1c1dc30":{"kind":"REWORK","_number":2,"created":"2021-07-08 12:26:00.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/82/799782/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/82/799782/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/799782/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/799782/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/799782/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/82/799782/2"}}},"commit":{"parents":[{"commit":"5e0b238b01250392fa9f92c069a7253902d6b325","subject":"Merge \"Bump up cAdvisor\u0027s version\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5e0b238b01250392fa9f92c069a7253902d6b325"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-07-07 09:41:04.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-07-08 12:25:25.000000000","tz":120},"subject":"Fix missing pacemaker-cli-utils in debuntu hacluster images","message":"Fix missing pacemaker-cli-utils in debuntu hacluster images\n\nHacluster images need to have pacemaker cli utils available\nin Debian/Ubuntu and they are provided by pacemaker-cli-utils\npackage. This package is also *recommended* dependency of pacemaker\nand pacemaker-remote in both Debian/Ubuntu systems, but\nunfortunatelly it is not installed via APT by default.\n\nThis patch adding this package to list of package to be installed\non Debian/Ubuntu systems.\n\nCloses-Bug: #1934788\nChange-Id: I055086ab4af37878e81844f1055bc48d84ae9945\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cd461f273c55365190bbece7bcef6aebe1c1dc30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cd461f273c55365190bbece7bcef6aebe1c1dc30"}]},"branch":"refs/heads/master"},"f309c94ca40f1ee4ebcf4eac6534ee05793d55f7":{"kind":"REWORK","_number":3,"created":"2021-07-29 09:39:17.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/82/799782/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/82/799782/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/799782/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/799782/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/799782/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/82/799782/3"}}},"commit":{"parents":[{"commit":"5e0b238b01250392fa9f92c069a7253902d6b325","subject":"Merge \"Bump up cAdvisor\u0027s version\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5e0b238b01250392fa9f92c069a7253902d6b325"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-07-07 09:41:04.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-07-29 09:39:14.000000000","tz":120},"subject":"Fix missing pacemaker-cli-utils in Debian hacluster images","message":"Fix missing pacemaker-cli-utils in Debian hacluster images\n\nHacluster-pacemaker images need to have pacemaker cli utils\navailable in all distros and they are provided by pacemaker-cli-utils\npackage in Debuntu systems. Due to a different dependency chain\nthis package is not installed in Debian. This package is also\n*recommended* dependency of pacemaker and pacemaker-remote in both\nDebian/Ubuntu systems, but unfortunatelly it is not installed\non Debian via APT by default.\n\nThis patch adding this package to list of package to be installed\non Debian/Ubuntu systems for both hacluster-pacemaker (required)\nand hacluster-pacemaker-remote (to ease with debugging).\n\nCloses-Bug: #1934788\nChange-Id: I055086ab4af37878e81844f1055bc48d84ae9945\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f309c94ca40f1ee4ebcf4eac6534ee05793d55f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f309c94ca40f1ee4ebcf4eac6534ee05793d55f7"}]},"branch":"refs/heads/master"},"8ddc9cc25c0e181d77b9cfd7551602718244c057":{"kind":"REWORK","_number":4,"created":"2021-07-30 11:08:41.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/82/799782/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/82/799782/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/799782/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/799782/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/799782/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/82/799782/4"}}},"commit":{"parents":[{"commit":"5e0b238b01250392fa9f92c069a7253902d6b325","subject":"Merge \"Bump up cAdvisor\u0027s version\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5e0b238b01250392fa9f92c069a7253902d6b325"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-07-07 09:41:04.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-07-30 11:08:39.000000000","tz":120},"subject":"Fix missing pacemaker-cli-utils in Debian hacluster images","message":"Fix missing pacemaker-cli-utils in Debian hacluster images\n\nHacluster-pacemaker images need to have pacemaker cli utils\navailable in all distros and they are provided by pacemaker-cli-utils\npackage in Debuntu systems. Due to a different dependency chain\nthis package is not installed in Debian. This package is also\n*recommended* dependency of pacemaker and pacemaker-remote in both\nDebian/Ubuntu systems, but unfortunatelly it is not installed\non Debian via APT by default.\n\nThis patch adding this package to list of package to be installed\non Debian/Ubuntu systems for both hacluster-pacemaker (required)\nand hacluster-pacemaker-remote (to ease with debugging).\n\nCloses-Bug: #1934788\nChange-Id: I055086ab4af37878e81844f1055bc48d84ae9945\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8ddc9cc25c0e181d77b9cfd7551602718244c057"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8ddc9cc25c0e181d77b9cfd7551602718244c057"}]},"branch":"refs/heads/master"},"7555c96c61b018b0015fee9f11cd2b8421b22884":{"kind":"REWORK","_number":5,"created":"2021-07-30 11:17:41.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/82/799782/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/82/799782/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/799782/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/799782/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/799782/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/82/799782/5"}}},"commit":{"parents":[{"commit":"5e0b238b01250392fa9f92c069a7253902d6b325","subject":"Merge \"Bump up cAdvisor\u0027s version\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5e0b238b01250392fa9f92c069a7253902d6b325"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-07-07 09:41:04.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-07-30 11:17:39.000000000","tz":120},"subject":"Fix missing pacemaker-cli-utils in Debian hacluster images","message":"Fix missing pacemaker-cli-utils in Debian hacluster images\n\nHacluster-pacemaker images need to have pacemaker cli utils\navailable in all distros and they are provided by pacemaker-cli-utils\npackage in Debuntu systems. Due to a different dependency chain\nthis package is not installed in Debian. This package is also\n*recommended* dependency of pacemaker and pacemaker-remote in both\nDebian/Ubuntu systems, but unfortunatelly it is not installed\non Debian via APT by default.\n\nThis patch adding this package to list of package to be installed\non Debian/Ubuntu systems for both hacluster-pacemaker (required)\nand hacluster-pacemaker-remote (to ease with debugging).\n\nCloses-Bug: #1934788\nChange-Id: I055086ab4af37878e81844f1055bc48d84ae9945\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7555c96c61b018b0015fee9f11cd2b8421b22884"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7555c96c61b018b0015fee9f11cd2b8421b22884"}]},"branch":"refs/heads/master"},"535c861d99aa0c07ea130d1f30cabb6af3db7ab7":{"kind":"REWORK","_number":6,"created":"2021-07-30 11:26:46.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/82/799782/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/82/799782/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/799782/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/799782/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/799782/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/82/799782/6"}}},"commit":{"parents":[{"commit":"5e0b238b01250392fa9f92c069a7253902d6b325","subject":"Merge \"Bump up cAdvisor\u0027s version\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5e0b238b01250392fa9f92c069a7253902d6b325"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-07-07 09:41:04.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-07-30 11:26:39.000000000","tz":120},"subject":"Fix missing pacemaker-cli-utils in Debian hacluster images","message":"Fix missing pacemaker-cli-utils in Debian hacluster images\n\nHacluster-pacemaker images need to have pacemaker cli utils\navailable in all distros and they are provided by pacemaker-cli-utils\npackage in Debuntu systems. Due to a different dependency chain\nthis package is not installed in Debian. This package is also\n*recommended* dependency of pacemaker and pacemaker-remote in both\nDebian/Ubuntu systems, but unfortunatelly it is not installed\non Debian via APT by default.\n\nThis patch adding this package to list of package to be installed\non Debian/Ubuntu systems for both hacluster-pacemaker (required)\nand hacluster-pacemaker-remote (to ease with debugging).\n\nCloses-Bug: #1934788\nChange-Id: I055086ab4af37878e81844f1055bc48d84ae9945\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/535c861d99aa0c07ea130d1f30cabb6af3db7ab7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/535c861d99aa0c07ea130d1f30cabb6af3db7ab7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
