)]}'
{"id":"openstack%2Fkolla~802206","triplet_id":"openstack%2Fkolla~stable%2Fwallaby~I0bb8b2429906e698858f6b699e4c4d8bacc2dd74","project":"openstack/kolla","branch":"stable/wallaby","topic":"bug/1934763-stable/wallaby","hashtags":[],"change_id":"I0bb8b2429906e698858f6b699e4c4d8bacc2dd74","subject":"Fix prometheus-elasticsearch-exporter container.","status":"MERGED","created":"2021-07-26 08:39:48.000000000","updated":"2021-07-27 08:04:03.000000000","submitted":"2021-07-27 08:02:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"802206-bug/1934763-stable/wallaby","cherry_pick_of_change":799620,"cherry_pick_of_patch_set":1,"meta_rev_id":"a1efe8e6f17f26344743e8335da82b10750aa9f5","_number":802206,"virtual_id_number":802206,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-27 08:02:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"recommended":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"all":[{"value":1,"date":"2021-07-26 09:17:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":2,"date":"2021-07-27 06:38:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-07-26 10:23:58.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":1,"date":"2021-07-27 06:38:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-26 09:17:56.000000000","updated_by":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"reviewer":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"state":"REVIEWER"},{"updated":"2021-07-26 10:03:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-27 06:38:06.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"9f47fb15e1b8721533e783d4e671e52547c057d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-07-26 08:39:48.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"036564bfbbc663586f013a895d62a86680acc7a8","author":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"date":"2021-07-26 09:17:56.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"cb7c5c5949c56899d9dd3cfdf526dc3c50ee39fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-26 10:03:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/e64e9eaa18d64b4cb7b9e47be11792a9 : SUCCESS in 26m 19s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/f0ddbd8efc89473b963b2f7dda34e8ec : SUCCESS in 33m 54s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/bd421ffc5ba3450a83c10ec2f336eafa : SUCCESS in 1h 01m 10s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/7b32bb335c934d5cbd100e9fa3335e71 : SUCCESS in 1h 09m 28s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/4a224432b58842caaa72d123c3c78aa4 : SUCCESS in 1h 18m 41s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/a37cc9e00bda49b4be51103022043e1f : SUCCESS in 40m 17s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/fd53cf3931014a9e825183b5b0d11f49 : SUCCESS in 26m 09s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/df07ccfe20244576aebec264bae8c751 : SUCCESS in 56m 53s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcf7d55f0d0e4e2c9baf61197ca590d6 : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d660c73fa6a943838a929356dc0ae681 : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/79c83a7b509448b4a5c226943f41e13a : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1eba5febe61c48fd9c0b862a8e77eec7 : SUCCESS in 5m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fcc80b8db95841ea942397b4fdba22e4 : SUCCESS in 6m 18s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f7486c8a132f4d3289881c669cb608b6 : SUCCESS in 27m 26s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/5ee9e94342304611909e23f49670a8b0 : SUCCESS in 41m 16s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e9149732ca004a9d983c3659c22983ad : SUCCESS in 52m 50s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2f59c544a4c94f7cbadfaad56a77fd16 : SUCCESS in 53m 26s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/fa4790a814e045aaa660436e759193a0 : SUCCESS in 1h 01m 39s","accounts_in_message":[],"_revision_number":1},{"id":"d3416ae0a93d3baa9c41a2b48ecf536b80c8bdf5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-07-26 10:23:58.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"50612c53972392400e81b334c4720b08ef47e849","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-26 11:34:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/d50eacc87c3e4a61b259a4f2ef03129e : SUCCESS in 1h 39m 45s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2d8f483fd9a542bc8d7bcd3befcbc00f : SUCCESS in 1h 08m 15s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5223cc663dfd4d89b7c97d342e17037f : SUCCESS in 2h 27m 34s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/4c466e239ad4405cb9a96d59d9369207 : SUCCESS in 1h 23m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c95dbba76ca3156cd7d6d4a3fc3b65d99aff9317","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-07-27 06:38:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ae532082026f0f7ba41aa70e78cfa64420e8bae7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 06:38:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ec1197cf36ee7b2505fec811ec4a9afe1e00fa05","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 08:02:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/15a97993b4a7469cb97052e81a314473 : SUCCESS in 50m 23s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/ab17c807793f472bb1ea9c34a981bafc : SUCCESS in 1h 12m 39s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/c75f5ace68a54a6f83d2a12ce52edb26 : SUCCESS in 1h 23m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52fada4fe94d4fbbaaa61b201cbb864b : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/721f897728a0441f9dadea7f46b8da45 : SUCCESS in 5m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/55c071b0247a44cb9925ca5a445c8c4c : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d568065b1126478fb5200ea89e1cbe32 : SUCCESS in 6m 46s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/48a47c10d829473a84bc82a42f158619 : SUCCESS in 38m 48s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/df59ef5b36724d14985891923ce26730 : SUCCESS in 1h 02m 37s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e01783721abe4bce944d4b799252e14f : SUCCESS in 1h 12m 59s","accounts_in_message":[],"_revision_number":1},{"id":"f1c9a93c4b1d39f12afb0cc019d79ee9665c6579","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 08:02:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a1efe8e6f17f26344743e8335da82b10750aa9f5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 08:04:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f219cb832f3478b9924a29b6cdf0f17 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"601872b9255e2abc97201e3d29e45fa0c0eee684","revisions":{"601872b9255e2abc97201e3d29e45fa0c0eee684":{"kind":"REWORK","_number":1,"created":"2021-07-26 08:39:48.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/06/802206/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/06/802206/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/802206/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/802206/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/06/802206/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/06/802206/1"}}},"commit":{"parents":[{"commit":"f0eb095e72b0be9066d584be6968e7ee716e45c4","subject":"gnocchi: add python3-rados to gnocchi-base source image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f0eb095e72b0be9066d584be6968e7ee716e45c4"}]}],"author":{"name":"Gaël THEROND (Fl1nt)","email":"gael.therond@bitswalk.com","date":"2021-07-06 10:04:17.000000000","tz":120},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-07-26 08:39:48.000000000","tz":0},"subject":"Fix prometheus-elasticsearch-exporter container.","message":"Fix prometheus-elasticsearch-exporter container.\n\n* Replace current URL with the proper target.\n\nChange-Id: I0bb8b2429906e698858f6b699e4c4d8bacc2dd74\nCloses-Bug: #1934763\n(cherry picked from commit 1f8ac21622295fa9f858aaf7308160a92fe69240)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/601872b9255e2abc97201e3d29e45fa0c0eee684"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/601872b9255e2abc97201e3d29e45fa0c0eee684"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
