)]}'
{"id":"openstack%2Fkolla~802207","triplet_id":"openstack%2Fkolla~stable%2Fvictoria~I0bb8b2429906e698858f6b699e4c4d8bacc2dd74","project":"openstack/kolla","branch":"stable/victoria","topic":"bug/1934763-stable/victoria","hashtags":[],"change_id":"I0bb8b2429906e698858f6b699e4c4d8bacc2dd74","subject":"Fix prometheus-elasticsearch-exporter container.","status":"MERGED","created":"2021-07-26 08:40:01.000000000","updated":"2021-07-28 10:59:23.000000000","submitted":"2021-07-28 10:58:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"802207-bug/1934763-stable/victoria","cherry_pick_of_change":799620,"cherry_pick_of_patch_set":1,"meta_rev_id":"542b3689fe9372dbdf04eabfe7a5ab49b80c3348","_number":802207,"virtual_id_number":802207,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-28 10:58:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"recommended":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"all":[{"value":1,"date":"2021-07-26 09:18:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":2,"date":"2021-07-27 06:38:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-07-26 10:24:05.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":1,"date":"2021-07-27 06:38:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-26 09:18:08.000000000","updated_by":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"reviewer":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"state":"REVIEWER"},{"updated":"2021-07-26 09:56:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-27 06:38:12.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"18840554fcdf1d5d35e771007d81cb5bb572d31a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-07-26 08:40:01.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* docker/prometheus/prometheus-elasticsearch-exporter/Dockerfile.j2","accounts_in_message":[],"_revision_number":1},{"id":"d337e7850abac003993684bdd5f887ba1e0a0e61","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-07-26 08:40:49.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"b78ebe507391de7c77e30f1950f572b4340dec08","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-07-26 08:40:53.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"0a3ccf758a86cdddb316ae8364e09360139f676d","author":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"date":"2021-07-26 09:18:08.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"a5edb97375fb18197a2985dbd260953e3f26dab9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-26 09:56:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/3b29f30bf58a4c65bdae8682a2014d34 : SUCCESS in 28m 19s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/5a83e9e9c6484486ac8dd8e930b17cc2 : SUCCESS in 28m 59s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/fe00b6099a5a449e95a9f97c474df784 : SUCCESS in 1h 12m 29s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/37a8358c5e4349acadba6b02ba1cd634 : SUCCESS in 1h 02m 16s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/6b98bdadb63d4033ae273eea61e86c01 : SUCCESS in 1h 10m 18s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b07b621ed47f46e5933041a69fa2c580 : SUCCESS in 26m 58s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/e574b98ab6bf49cd848d3a3844232dea : SUCCESS in 25m 39s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/5830cff0ee2d47e78a52c959b8410227 : SUCCESS in 59m 18s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fff203a1c404092afbe9f578845e5f3 : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fc39bbcf15024d7c9a8b3fad997df13e : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f173d8e2be74477f80cc83312df218da : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce62e5a9385e4e5a81c86a7bfdde72cb : SUCCESS in 6m 20s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b32e2aea305449d69d6d8cb79e31a8e6 : SUCCESS in 25m 13s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0f60439499784bf7b1d7afe5f332f581 : SUCCESS in 30m 22s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7d8e5770570243bd8158b475d09b3cd6 : SUCCESS in 52m 30s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0789fdacc0a24f3db8f73a81ea4df151 : SUCCESS in 48m 20s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d5252df486d84f19b1094f55685539b9 : SUCCESS in 1h 01m 30s","accounts_in_message":[],"_revision_number":2},{"id":"ec59138abeb50f49ebea7b10d2c4a7610bcb94a9","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-07-26 10:24:05.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"094f2406f44438fb364f4681eb4db95c9dc96f80","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-26 10:56:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/7c7b2526be03462897647d4bfe9a5970 : SUCCESS in 1h 30m 49s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a1734e34faff46459e58d9d8862cbb0b : SUCCESS in 1h 08m 19s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8e4b7c448c254f8d8336a2f6e28090fd : SUCCESS in 1h 52m 50s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/b3aa39e331694289a450654691056f2f : SUCCESS in 1h 12m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d675e67ce8795bffdeed2527d60b4c6621435d24","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-07-27 06:38:12.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7ff33a95283ad00928a78fe69707e41295253e25","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 06:38:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d54acd7cdbb776fed1a0efb9fabdaef704f6bdbb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 08:02:45.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/e76a4e89c85a491b970f6ea8f296d332 : SUCCESS in 31m 17s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/8c1fb0bce8fa4a8a92e9df329395f8c2 : SUCCESS in 1h 04m 48s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/cc2a8f4b9dff4551807272fc8e9c57a6 : FAILURE in 27m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6afc8b4f698d4b08b8b440219f7da2de : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c9b7480a59664f9e92e1a51d84bb5e6c : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/323fee6d55bb43fbbe634a24440a4135 : SUCCESS in 6m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b3e55e80d4c4974bf6b21e4f403da84 : SUCCESS in 7m 19s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ae21ce0245f040a3935ae5df113db8f7 : SUCCESS in 29m 24s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/15fd749295124756b9ff0acba536a6b4 : SUCCESS in 57m 59s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2e7aab52030b4c4f8024e11f851de38a : SUCCESS in 1h 04m 52s","accounts_in_message":[],"_revision_number":2},{"id":"e609efe62ba5ea0728743292c24114c46d9edda3","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-07-27 08:03:16.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"288cd4557697fa74f67dc3f550bd2a1b49455a2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 09:22:12.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/00b46afbaca140cab9ca60c4fec6e1db : SUCCESS in 25m 01s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/d8b24653e7ef4410bd436c78898addf9 : FAILURE in 27m 37s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/f858a045b0ad4321be9c28122f78b5da : SUCCESS in 1h 00m 33s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/22a375c4ac4440a6a7756ee51c02048e : SUCCESS in 1h 03m 02s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/32622d39c1aa465aa261ebc15d5cfcbd : SUCCESS in 1h 14m 41s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/6b0572f848e64fe8adfb2a90392a9956 : SUCCESS in 22m 59s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/7f0a16714673452ab7acff5a04d79580 : FAILURE in 19m 57s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c8910ac761a549a7b6d88f3f64e097d8 : SUCCESS in 52m 37s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d6a1f5272784c549865d7893c67131d : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8ba7cb69b02e40f49d3c592768dd8629 : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4229e247b64f4a808df91fd1698730eb : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c0b6c61264745e8bce9ea0a6dff416d : SUCCESS in 6m 06s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/78d3b20836bf4c9093e5b80ce4431232 : SUCCESS in 21m 36s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/36f0d910fdb54d5fa4879da10c9b6ccc : SUCCESS in 21m 45s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f0604ce796a84fb9a1b6f000ae691f47 : SUCCESS in 1h 00m 25s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/702a995311ed4f289d8c2288afe5ffb9 : SUCCESS in 58m 07s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f8f6f0d7f84643589b38711b237f3f0f : SUCCESS in 1h 06m 36s","accounts_in_message":[],"_revision_number":2},{"id":"f5a925f84d9b684c0209bdfe1b0ab2793876cdbf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 13:16:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/c5cdc81d692744c295b37b31e1b1598f : SUCCESS in 1h 32m 18s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/039e32f814144c31b15c19217f1ad0f2 : SUCCESS in 2h 45m 23s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e50523a800534b5cbe828c7b087ad0cf : SUCCESS in 2h 06m 05s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/9693644fe08c423d892fd8c6d2a8ac6b : SUCCESS in 1h 12m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"227fe9d64411ca2fd8a566c6e912b3cea3095f44","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-07-28 08:32:42.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"98894eef7606cf2e0895d431771e93226ab5bd32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 09:45:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/b2deba1a75dd41b7863d19fffa546ccc : FAILURE in 21m 47s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/51636fe74ff04509a93f436644d47c92 : SUCCESS in 27m 50s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/a8092c4b95fc4e249a66a507c51287bb : SUCCESS in 1h 01m 21s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/1699254bec7f45d4804f64e42a3cf4cd : SUCCESS in 1h 03m 11s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/c1a7ee886e974e4da0a591f3260f3270 : SUCCESS in 1h 10m 37s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/bac7c236f32d4e608c69d58bf3fd13d8 : SUCCESS in 21m 37s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/8376ea71cfda4269ba615d3b9c0d5b8a : FAILURE in 18m 52s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/90f8262735ef44618e168e01e9a18c6d : SUCCESS in 55m 19s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8abdbe3b5fde4a3d8e2c8ac1c56b60a5 : SUCCESS in 5m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f55c9d2274b8416792e894f737b97f55 : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/540d433d1fa740b68b65cded8c95e447 : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff60b2048e564f84b6ce879fa14a2282 : SUCCESS in 7m 06s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4d1543520bf34186a1dde852d5526494 : SUCCESS in 24m 15s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f2c30bfa5bc2416995cce6ee21d50d32 : SUCCESS in 26m 44s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d08bfd738f494091a7aa4621d9ac98a4 : SUCCESS in 52m 01s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5282f5cbcd3e418ebd0a91f2e05bdf65 : SUCCESS in 52m 45s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9a0d87a753cc47049a3ba2db9c95bbb9 : SUCCESS in 1h 03m 09s","accounts_in_message":[],"_revision_number":2},{"id":"efe56d1ec4754b2f5c021c46f70f044e6ae0b9a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 09:45:41.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1aac212824ab49bdfccf44397e795fbd20a32003","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 10:40:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/37ebd01c6a364cefac4f1c12390c704b : SUCCESS in 2h 01m 43s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8b5b3deb8492414487ecad07b6c10667 : SUCCESS in 33m 25s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e016e572fbe1412ba7527191543eadcb : SUCCESS in 2h 07m 07s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/f842b9e5062844aca90c2f614e86e68f : SUCCESS in 34m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4ab032986eb36925da7d1a0c161a2e9036b60b7f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 10:58:08.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/1ecd986094394d9dacb1abd654e4d21e : SUCCESS in 54m 35s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/5ed7212707af4f35917c59864e2a2c00 : SUCCESS in 1h 04m 02s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/e4ae1b7e2c0d4cd8a8b679acf364e39e : SUCCESS in 1h 09m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ea1f8c49b2a46d5a30f98fe3e3958eb : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a7ad8f170d394668ae02df232459f80f : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b3e0f7dbf12545929ccc63aaacae61d5 : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0a948b1ac3d421a95ee0a121191ebe6 : SUCCESS in 6m 59s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d2638bb43141408cbfb1153a70157c06 : SUCCESS in 20m 09s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8488f29d740f47b5bbd4dc2378dc8e7a : SUCCESS in 50m 52s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0022c88b821a4163882d409f36740b35 : SUCCESS in 1h 02m 15s","accounts_in_message":[],"_revision_number":2},{"id":"a6a9d0f7ebea1e624fcb53a8e92d266bdd27786b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 10:58:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"542b3689fe9372dbdf04eabfe7a5ab49b80c3348","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 10:59:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57f890b80ed24115a85a8a26f3e712d4 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"cbc404a7265402355d1ae754aecac6fae0201013","revisions":{"60c3baae269b3cd3c6fb890c123c53a6768b5ef1":{"kind":"REWORK","_number":1,"created":"2021-07-26 08:40:01.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/07/802207/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/07/802207/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/802207/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/802207/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/802207/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/07/802207/1"}}},"commit":{"parents":[{"commit":"b9eac37a6fe71307b78727a8d0bf810698457f26","subject":"gnocchi: add python3-rados to gnocchi-base source image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b9eac37a6fe71307b78727a8d0bf810698457f26"}]}],"author":{"name":"Gaël THEROND (Fl1nt)","email":"gael.therond@bitswalk.com","date":"2021-07-06 10:04:17.000000000","tz":120},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-07-26 08:40:01.000000000","tz":0},"subject":"Fix prometheus-elasticsearch-exporter container.","message":"Fix prometheus-elasticsearch-exporter container.\n\n* Replace current URL with the proper target.\n\nChange-Id: I0bb8b2429906e698858f6b699e4c4d8bacc2dd74\nCloses-Bug: #1934763\n(cherry picked from commit 1f8ac21622295fa9f858aaf7308160a92fe69240)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/60c3baae269b3cd3c6fb890c123c53a6768b5ef1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/60c3baae269b3cd3c6fb890c123c53a6768b5ef1"}]},"branch":"refs/heads/stable/victoria"},"cbc404a7265402355d1ae754aecac6fae0201013":{"kind":"REWORK","_number":2,"created":"2021-07-26 08:40:49.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/07/802207/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/07/802207/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/802207/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/802207/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/07/802207/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/07/802207/2"}}},"commit":{"parents":[{"commit":"b9eac37a6fe71307b78727a8d0bf810698457f26","subject":"gnocchi: add python3-rados to gnocchi-base source image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b9eac37a6fe71307b78727a8d0bf810698457f26"}]}],"author":{"name":"Gaël THEROND (Fl1nt)","email":"gael.therond@bitswalk.com","date":"2021-07-06 10:04:17.000000000","tz":120},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-07-26 08:40:47.000000000","tz":0},"subject":"Fix prometheus-elasticsearch-exporter container.","message":"Fix prometheus-elasticsearch-exporter container.\n\n* Replace current URL with the proper target.\n\nChange-Id: I0bb8b2429906e698858f6b699e4c4d8bacc2dd74\nCloses-Bug: #1934763\n(cherry picked from commit 1f8ac21622295fa9f858aaf7308160a92fe69240)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cbc404a7265402355d1ae754aecac6fae0201013"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cbc404a7265402355d1ae754aecac6fae0201013"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
