)]}'
{"id":"openstack%2Fkolla~802208","triplet_id":"openstack%2Fkolla~stable%2Fussuri~I0bb8b2429906e698858f6b699e4c4d8bacc2dd74","project":"openstack/kolla","branch":"stable/ussuri","topic":"bug/1934763-stable/victoria-stable/ussuri","hashtags":[],"change_id":"I0bb8b2429906e698858f6b699e4c4d8bacc2dd74","subject":"Fix prometheus-elasticsearch-exporter container.","status":"MERGED","created":"2021-07-26 08:41:02.000000000","updated":"2021-07-27 07:53:25.000000000","submitted":"2021-07-27 07:52:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"802208-bug/1934763-stable/victoria-stable/ussuri","cherry_pick_of_change":802207,"cherry_pick_of_patch_set":2,"meta_rev_id":"a7883aeeeab023b80cb31833da6afe65383cb6f5","_number":802208,"virtual_id_number":802208,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-27 07:52:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"recommended":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"all":[{"value":1,"date":"2021-07-26 09:18:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":2,"date":"2021-07-27 06:37:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-07-26 10:23:31.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":1,"date":"2021-07-27 06:37:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-26 09:18:22.000000000","updated_by":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"reviewer":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"state":"REVIEWER"},{"updated":"2021-07-26 10:01:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-27 06:37:58.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"517513420ede631c79c3c3a8136d1847b94aec55","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-07-26 08:41:02.000000000","message":"Patch Set 1: Cherry Picked from branch stable/victoria.","accounts_in_message":[],"_revision_number":1},{"id":"cf8d86bd69ac6e584afff86196a4f88529cecac4","author":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"date":"2021-07-26 09:18:22.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6b3f610a9c06dc413c5a84633dddbc3a1237641f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-26 10:01:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/b70f221544754091bdf74eb4b00590bd : SUCCESS in 30m 54s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/9878db93edd241b298d815944cb70645 : SUCCESS in 28m 31s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/1d90914655e24db6928f55788416465f : SUCCESS in 1h 08m 41s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/9f82ff2b162a414db8ad6c31a8ceaa3a : SUCCESS in 55m 30s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/89a30d17c48f42eaa2bbce318d68bae6 : SUCCESS in 1h 14m 51s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/bfbc1d15d52a43bda201a3f726a3bced : SUCCESS in 32m 28s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/b10b21368f194153895fe1fe97cedd96 : SUCCESS in 24m 39s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/adebb65cd6144c00ad8cee836ff16c01 : SUCCESS in 1h 01m 50s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9911c9e7878b433bb7d2169e3d4d1d00 : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0e8b15b0e4214027b6a1ce276707bb3e : SUCCESS in 4m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f7a96027d3e34ff9b358000967387e4c : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84d5f57d515b487f82cd4f2c4bbcf0fa : SUCCESS in 6m 04s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f3d44718503f406faa70994acd37c3be : SUCCESS in 24m 53s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b59b79195d734e1987620b9e9f9ca87e : SUCCESS in 31m 30s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d2a528f947c7467497c67c363fb8c82d : SUCCESS in 1h 02m 25s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f7850f9f540e401ea4c7b9a136e0ab60 : SUCCESS in 56m 05s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5f4594b93a184f4d928b39081e001dc4 : SUCCESS in 1h 08m 24s","accounts_in_message":[],"_revision_number":1},{"id":"c775fcda91efc9f054c02b95f9645d69df281b57","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-07-26 10:23:31.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"45adc68defdc7148d92745dc986758aafa5b6c73","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-26 13:18:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/d3b9cdc2dffc4d0e998e4f10e92b631c : SUCCESS in 2h 00m 33s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3afb91ad328a4aac86004b09e9da81aa : SUCCESS in 1h 02m 58s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8b52a8ae0e8a4990bb0827a20e0a81f7 : SUCCESS in 1h 56m 22s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/d0be25c94a7448d690e23915af4a5fda : SUCCESS in 1h 25m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3840099067739c4b0460555aaa3fb8dd722a2d07","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-07-27 06:37:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6a0f734ba14ff8cdee52ee93c489672fc8bb80fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 06:38:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"523ff0dad168d73c6ca63812e1c8f2e9b349c634","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 07:52:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/83d85481d8ce4985bfd8b32e5ff2806e : SUCCESS in 30m 53s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/392b060e2dc7428b8f0c694b6581337b : SUCCESS in 56m 44s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/51a213b98dec4babbda48e5896a8d436 : SUCCESS in 1h 03m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89e608e7608d46589f38daacf6b90590 : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c916896816bf4cd58191742537d7bd2c : SUCCESS in 5m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/abe83ba483dc4b2297282886e66b0495 : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bd9b7fb654347cea96e92052fcfef82 : SUCCESS in 6m 56s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/94c65b54db4540449c90a9702a9c1181 : SUCCESS in 31m 03s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/02985cb70e7140269931abafc7d0b6f4 : SUCCESS in 1h 00m 49s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8020bbf46e264d2fb32e16c28f305879 : SUCCESS in 1h 12m 50s","accounts_in_message":[],"_revision_number":1},{"id":"858f2e5cf5e6a3cae3c43a1f30f5833c5dc7bca4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 07:52:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a7883aeeeab023b80cb31833da6afe65383cb6f5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 07:53:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdb7414a99de4f5e8728708ecf6587f8 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f28f4a9425b3ac13902bbc3a71e07339874acabd","revisions":{"f28f4a9425b3ac13902bbc3a71e07339874acabd":{"kind":"REWORK","_number":1,"created":"2021-07-26 08:41:02.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/08/802208/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/08/802208/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/802208/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/802208/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/08/802208/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/08/802208/1"}}},"commit":{"parents":[{"commit":"4d122108323482511ffe86ea2a580c26367ebe2c","subject":"gnocchi: add python3-rados to gnocchi-base source image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4d122108323482511ffe86ea2a580c26367ebe2c"}]}],"author":{"name":"Gaël THEROND (Fl1nt)","email":"gael.therond@bitswalk.com","date":"2021-07-06 10:04:17.000000000","tz":120},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-07-26 08:41:02.000000000","tz":0},"subject":"Fix prometheus-elasticsearch-exporter container.","message":"Fix prometheus-elasticsearch-exporter container.\n\n* Replace current URL with the proper target.\n\nChange-Id: I0bb8b2429906e698858f6b699e4c4d8bacc2dd74\nCloses-Bug: #1934763\n(cherry picked from commit 1f8ac21622295fa9f858aaf7308160a92fe69240)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f28f4a9425b3ac13902bbc3a71e07339874acabd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f28f4a9425b3ac13902bbc3a71e07339874acabd"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
