)]}'
{"id":"openstack%2Fkolla~802352","triplet_id":"openstack%2Fkolla~stable%2Fwallaby~I4ce8244a3fb034493a75f4096de2640e11b48931","project":"openstack/kolla","branch":"stable/wallaby","topic":"bug/1931867-stable/wallaby","hashtags":[],"change_id":"I4ce8244a3fb034493a75f4096de2640e11b48931","subject":"Fix missing templating block for kolla-toolbox.","status":"MERGED","created":"2021-07-27 07:59:50.000000000","updated":"2021-07-28 12:45:07.000000000","submitted":"2021-07-28 12:43:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"802352-bug/1931867-stable/wallaby","cherry_pick_of_change":796279,"cherry_pick_of_patch_set":6,"meta_rev_id":"88582761d7acb0ef3abfef0ee42a40cc6d8d4428","_number":802352,"virtual_id_number":802352,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-28 12:43:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-07-28 08:40:41.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-07-28 11:13:42.000000000","_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-07-28 11:13:42.000000000","_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-27 09:24:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-28 11:13:42.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"}],"messages":[{"id":"5a665b91a9c71794a8023923d7605a78be1b8398","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-07-27 07:59:50.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"ee3cde042b2e5f82b7d3aced5dbebf9a95b6776d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 09:24:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/6ea0c190f9bc49e0b2c529c0f669cfc8 : FAILURE in 21m 30s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/6b2a45c32fa5416b801e4dcf17622c32 : SUCCESS in 45m 11s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/a91ff4e6ffc3477abd1af8b3c032221a : SUCCESS in 59m 50s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/db2e7eb08bc54462987c9c8fa674e911 : SUCCESS in 58m 53s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/1f7e9469374e4d96ae34836a28f5e371 : SUCCESS in 1h 21m 44s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/89208d2ec48e490b8b5e65aa3d6e7b00 : SUCCESS in 32m 48s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/ba63e5f3a81d453d81f721034ff22754 : SUCCESS in 15m 58s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/514969f88ed7479c8ff773fc919990ff : SUCCESS in 54m 05s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5f9df8a1c75445fb5685efe103b6690 : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/670374c35ee249f9be9d9fa6e9965984 : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c3d9b4f508874f7182173ec10800a96a : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/599fcb8efab942748797d4d5fe89b7cb : SUCCESS in 4m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b603fd8377ff48d487521b8be6488b5c : SUCCESS in 6m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2cd2ac4027c24e76bfd4513692ff1e3d : SUCCESS in 4m 35s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cb48abe4acd7473f959acd28305ea8ec : SUCCESS in 21m 07s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/73046460640844a7a9b4526013aacb30 : SUCCESS in 27m 06s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/80419e86e32848c79fe5659571bf9ef7 : SUCCESS in 48m 54s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e36c0cd8fd7e4354b2a53271c8784f55 : SUCCESS in 56m 01s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c4cb67fe22544a368fb42ea7a6170bd5 : SUCCESS in 55m 58s","accounts_in_message":[],"_revision_number":1},{"id":"f09de8ffcfa4f7302d17827b913bae4c26ead3c8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 11:15:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/75fe29709e844722865789b16e89cfdc : SUCCESS in 2h 45m 37s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/07b23ff8f75c42a994374a073d0f1860 : SUCCESS in 2h 47m 02s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3de3b9b4becb4c01b207c7a138ee2fbc : TIMED_OUT in 3h 02m 02s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/18abdef4034648beb600c7cd5a938aa9 : SUCCESS in 2h 43m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"166b317429f460df6d5ad3feca1e3dc9e5bbb57e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-07-28 08:40:41.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"75f5f8bcab0609e907ffc9f0846bdae9e80effb4","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2021-07-28 11:13:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3416be32f240e87a203fa1aa78277eda4ef05c9d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 11:14:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"30d1d4d2c1ba31aaa2619e8f704207943c988955","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 12:43:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/cdf52210b4bc46dcb02901b7bc759f66 : SUCCESS in 39m 08s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/1c15dd8099134e9f92b679266f06c5a2 : SUCCESS in 1h 04m 45s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/aa8c1fb1acf246848607af1286b5cb10 : SUCCESS in 1h 25m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d53bfb5c0f4b4d2588f500f0a24c0996 : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/60533d5da01f453191491b48e6228ca5 : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/35e4ba667c27488e96d4c688d98fb186 : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c9d67892d054876bbc2316c33a40d2b : SUCCESS in 6m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/67c756da45984ae3acd5db61cf57134f : SUCCESS in 4m 57s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a154225adc9947c0aa68c7b0dbdde68b : SUCCESS in 35m 23s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/94f702aa9281489c8eaab06e5109d70e : SUCCESS in 57m 10s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/edff0174d9094c32871f61beaf09e519 : SUCCESS in 1h 05m 15s","accounts_in_message":[],"_revision_number":1},{"id":"a78fa1ccc145cd9dbc64ec0441e6f9eaca80f28a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 12:43:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"88582761d7acb0ef3abfef0ee42a40cc6d8d4428","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 12:45:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/597c569778ea44e8b9daaf4e8f20a5e2 : SUCCESS in 1m 09s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/508e5925e52346f9b6ee6b3337a2f278 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"52a43b781c38e008d84fbc52198a5d6272e97b4f","revisions":{"52a43b781c38e008d84fbc52198a5d6272e97b4f":{"kind":"REWORK","_number":1,"created":"2021-07-27 07:59:50.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/52/802352/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/52/802352/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/802352/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/802352/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/802352/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/52/802352/1"}}},"commit":{"parents":[{"commit":"f0eb095e72b0be9066d584be6968e7ee716e45c4","subject":"gnocchi: add python3-rados to gnocchi-base source image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f0eb095e72b0be9066d584be6968e7ee716e45c4"}]}],"author":{"name":"Gaël THEROND (Fl1nt)","email":"gael.therond@bitswalk.com","date":"2021-07-23 14:04:52.000000000","tz":120},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-07-27 07:59:50.000000000","tz":0},"subject":"Fix missing templating block for kolla-toolbox.","message":"Fix missing templating block for kolla-toolbox.\n\nChange-Id: I4ce8244a3fb034493a75f4096de2640e11b48931\nCloses-Bug: #1931867\n(cherry picked from commit fb218ff9bf3803188991ae0f1a8f66ffa5176872)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/52a43b781c38e008d84fbc52198a5d6272e97b4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/52a43b781c38e008d84fbc52198a5d6272e97b4f"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
