)]}'
{"id":"openstack%2Fkolla~802353","triplet_id":"openstack%2Fkolla~stable%2Fvictoria~I4ce8244a3fb034493a75f4096de2640e11b48931","project":"openstack/kolla","branch":"stable/victoria","topic":"bug/1931867-stable/victoria","hashtags":[],"change_id":"I4ce8244a3fb034493a75f4096de2640e11b48931","subject":"Fix missing templating block for kolla-toolbox.","status":"MERGED","created":"2021-07-27 07:59:59.000000000","updated":"2021-07-28 12:27:31.000000000","submitted":"2021-07-28 12:26:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"802353-bug/1931867-stable/victoria","cherry_pick_of_change":796279,"cherry_pick_of_patch_set":6,"meta_rev_id":"64c999adea5bc435c1165e86de8cca8c45a90e59","_number":802353,"virtual_id_number":802353,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-28 12:26:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-07-28 08:40:28.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-07-28 11:12:30.000000000","_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-07-28 11:12:30.000000000","_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-27 09:16:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-28 11:12:30.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"}],"messages":[{"id":"e51fa24643f308b0315bceb388e759d29d0df08b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-07-27 07:59:59.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"22499f1f42bb0b4d4b8d46d98c661267e6a1e81e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 09:16:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/7d0406775b614adcb9fbf9030c26ddc5 : SUCCESS in 22m 34s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/a224e709ac7a40a79c4e35bbeccf6e04 : SUCCESS in 25m 31s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/0294fda154094b2296dbc9758df397c1 : SUCCESS in 58m 26s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/354815a1e6954f3da149804f92dbe43a : SUCCESS in 55m 59s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/74ae0815485c4a9186b79f0f74cd0c9d : SUCCESS in 1h 14m 00s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/df9a6df27acf48c2ae5d1054b4e7cb1e : SUCCESS in 19m 45s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/334a9c40e57b49f484320e1aaa3da8fd : FAILURE in 18m 50s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/46538e8623ff4bb899f3be8ead05729b : SUCCESS in 51m 40s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36ef7860f7ee45718525c6abac33bc04 : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ad2bc6dd4e29499b804444046a516c54 : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b42a5df6dd124c6f97b797ea439078ff : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be0f8bd7a07f4e32b8bf47799fa78c30 : SUCCESS in 5m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8e2c73e8c1d746369424910c6e83f57c : SUCCESS in 4m 30s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/28142cb25b504269b147df70f13df8d3 : SUCCESS in 19m 52s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f930439a419e40b7ab09651125410139 : SUCCESS in 24m 03s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e4944b54005149f4a274491cd4c42bdf : SUCCESS in 56m 14s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3d92895df9bf457bbec7c080e19e2a4d : SUCCESS in 54m 20s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b4f9429efcac4401b2581b96138b91ba : SUCCESS in 59m 01s","accounts_in_message":[],"_revision_number":1},{"id":"ea6c9967005b48cafc8df5fda2036c49258491e4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 12:38:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/7ddf434c30424efd8f2b03b2516f57c3 : SUCCESS in 2h 56m 28s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4a2d7c76984747f3ab613669e4cf9ee3 : SUCCESS in 1h 30m 07s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a5f1ae9d1f8447da96bc84a2af414aa7 : SUCCESS in 2h 12m 12s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/bc8a013d81da4a02bc3aa69639f944cb : SUCCESS in 2h 42m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e0124bc780fecb581e16bbfbfa7f4dde673c406b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-07-28 08:40:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"796fec25d0259fe59e16d178acdea7933e27bb97","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2021-07-28 11:12:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7be8825d70c37260f91d8dd101ebd3f589c59dcc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 11:12:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7e063cf29e7e1611ffb9638e4c9c97c34e06cd22","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 12:26:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/e73213b6001a4606a2fcfa988a06028d : SUCCESS in 27m 11s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/ad2bce19a0e94ff5bd4f8a94c1a9e3aa : SUCCESS in 1h 04m 45s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/6e43940acee749cda5f160b33169cfaf : SUCCESS in 1h 09m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2519759620e46058f6ced336e62b7b4 : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cb5e8e17c9bb4c2ea402a975da90521f : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b1b0b2ae8e4545aaafe3a7e4dd82eedc : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/717b3f7022714d7fa7c2eb4c2242f6f8 : SUCCESS in 6m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6d35eedb837f4f538982d8d6a18bd9fb : SUCCESS in 5m 11s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/656abae24d5d4612a61329d40f942943 : SUCCESS in 21m 01s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4271b52656424d88b7ae4a0fffd788c3 : SUCCESS in 49m 15s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/de27d8475d5f4323934fac5d9536ec35 : SUCCESS in 1h 07m 14s","accounts_in_message":[],"_revision_number":1},{"id":"bbf44f65b1b53efd59cf1902ac1bdff67f0655c0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 12:26:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"64c999adea5bc435c1165e86de8cca8c45a90e59","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 12:27:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab26a562ed964094ab9d2eb26fe2f0ba : SUCCESS in 1m 02s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/62a40159403b4605bc8bea02c7f9b8bd : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"be0373c92859070b24665b5d4146c87bec9ccee5","revisions":{"be0373c92859070b24665b5d4146c87bec9ccee5":{"kind":"REWORK","_number":1,"created":"2021-07-27 07:59:59.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/53/802353/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/53/802353/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/802353/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/802353/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/802353/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/53/802353/1"}}},"commit":{"parents":[{"commit":"b9eac37a6fe71307b78727a8d0bf810698457f26","subject":"gnocchi: add python3-rados to gnocchi-base source image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b9eac37a6fe71307b78727a8d0bf810698457f26"}]}],"author":{"name":"Gaël THEROND (Fl1nt)","email":"gael.therond@bitswalk.com","date":"2021-07-23 14:04:52.000000000","tz":120},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-07-27 07:59:59.000000000","tz":0},"subject":"Fix missing templating block for kolla-toolbox.","message":"Fix missing templating block for kolla-toolbox.\n\nChange-Id: I4ce8244a3fb034493a75f4096de2640e11b48931\nCloses-Bug: #1931867\n(cherry picked from commit fb218ff9bf3803188991ae0f1a8f66ffa5176872)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/be0373c92859070b24665b5d4146c87bec9ccee5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/be0373c92859070b24665b5d4146c87bec9ccee5"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
