)]}'
{"id":"openstack%2Fkolla~802355","triplet_id":"openstack%2Fkolla~stable%2Fussuri~I75927b6016bdb7577fbff354632bdab6407f359c","project":"openstack/kolla","branch":"stable/ussuri","topic":"bug/1931980-stable/ussuri","hashtags":[],"change_id":"I75927b6016bdb7577fbff354632bdab6407f359c","subject":"Improve offline build scenario.","status":"MERGED","created":"2021-07-27 09:34:37.000000000","updated":"2021-07-29 11:20:46.000000000","submitted":"2021-07-29 11:19:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"802355-bug/1931980-stable/ussuri","cherry_pick_of_change":796480,"cherry_pick_of_patch_set":3,"meta_rev_id":"3e9b95e6b1ae2a53e3f7d7a6e9a59b7e250d5fd1","_number":802355,"virtual_id_number":802355,"owner":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-29 11:19:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2021-07-29 10:01:32.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2021-07-29 09:12:44.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2021-07-29 10:01:32.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-27 09:48:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-29 09:12:44.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2021-07-29 10:01:32.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"}],"messages":[{"id":"303627dbb365b37bc2a28a3c0568c9c2c57a01be","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"date":"2021-07-27 09:34:37.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* docker/base/Dockerfile.j2","accounts_in_message":[],"_revision_number":1},{"id":"e939083cdee769cb44f53962b56342ce944f39b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 09:48:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/605894e64e2244dab30cf1770f7a277f : FAILURE in 5m 28s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/980ceef29b404a9a988bd832f7a8d485 : FAILURE in 5m 58s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/1972074208c64b44b895cec9f5722391 : FAILURE in 9m 50s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/367f84f7701f48e887ae70aa3ce16064 : FAILURE in 10m 13s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/b1e6bb64cee24d81819271422bcf4535 : FAILURE in 9m 32s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/4616082e5ec34b6dba89494ce4611443 : FAILURE in 4m 16s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/90c0c9e43a6e4ac78c3630916f592a17 : FAILURE in 4m 25s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/93c3663a924b4b7b8b5023853bd29c89 : FAILURE in 8m 46s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48f22f0c42e942268332f384e3203aec : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9588c472600949c9adde11270407834c : SUCCESS in 4m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/acc181d32f2643cebe067536471bbeae : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/840f96bebd144eefad8d59532c227d30 : SUCCESS in 6m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2d5249becd74483d93e7d40a9e5cfd8a : SUCCESS in 4m 58s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4ca01aeecdd341a3a5c654aa04e3b0c0 : FAILURE in 4m 00s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d0335469d3644ed8a7e5a189e2ef827c : FAILURE in 4m 23s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ae5d119a2d02476eb4d0f3b11b668426 : FAILURE in 8m 54s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/83c42a91c3674eec838fdeeff13da760 : FAILURE in 8m 34s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/fab23b5bbbfc4979b7ccf5caebd17270 : FAILURE in 11m 17s","accounts_in_message":[],"_revision_number":1},{"id":"c44cb96ab8ea45e7eeb9703c7ad461933ff12bca","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 10:10:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/03c6d3929606470588c81f46cd0a404d : FAILURE in 12m 52s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0d8429f94ec14bb7a5e9167a18d133a3 : FAILURE in 6m 46s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d55a1368d50b4803a7a26d6a37994bc2 : FAILURE in 19m 33s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/0921beef7d3d4cbdb5b75fa29144028b : FAILURE in 7m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6a3a086869b7cfb01bbc16b244d5db2989f1697f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"date":"2021-07-29 07:18:55.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"b0713d1d45c2b71359f47d7966042ce4cdb2a211","author":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"date":"2021-07-29 07:19:09.000000000","message":"Patch Set 2:\n\nConflicts resolved.","accounts_in_message":[],"_revision_number":2},{"id":"f2313de0335e136490ea0ebcfff8547675711fe5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 08:56:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/b5cb4be1200a4f6b816a9ee44db545ae : SUCCESS in 24m 34s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/a8b4acf5dc7542ceb096f8d1ebb6e093 : SUCCESS in 30m 40s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/9a1c1f04c140427a911acf7cc0686d9e : SUCCESS in 1h 02m 55s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/c2e4b5afc4d64282962bcbb1f92983a5 : SUCCESS in 1h 02m 16s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/a87d3274d04b43038397c115c29dc865 : SUCCESS in 1h 16m 58s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/5f0c8e2245ef497794478f12a1c44542 : SUCCESS in 24m 40s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/ec809a2bf3ba4beb82eb9c16f56d8b96 : SUCCESS in 19m 39s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d73f3905bf4041d7b09483ac9cc39cb9 : SUCCESS in 1h 00m 20s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d9f4796f3b24bbaa07cde7a98013df9 : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c28016fa6ceb45a0b699c35eb54ecd0c : SUCCESS in 4m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1eaa9cad1df9405d905c3e3a0d2cad35 : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f21a7c36b2844370b1fe52f3d4582c8c : SUCCESS in 6m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5266562790c740d08992d67e9e29f87b : SUCCESS in 4m 08s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1916d98135134b9c9002ca12099d23b1 : SUCCESS in 23m 17s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/443060dbf66e44b18fc28820c7161b4d : SUCCESS in 31m 24s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/341270b853e94ad18530f059b86b1751 : SUCCESS in 56m 44s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f24d20522a72468d80a37d64a083cf01 : SUCCESS in 53m 31s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a4373ae48cf844139dfd2d1793cd5598 : SUCCESS in 1h 12m 04s","accounts_in_message":[],"_revision_number":2},{"id":"bd3f4176fa90e3bf50fb4b6b655f7966c48fdbdc","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-07-29 09:12:44.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"9b6b3489258ad4e5bd29b6701067f5569fe2b42c","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-07-29 10:01:32.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"df71bd50e50deaf19119328bcca761ec2536f772","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 10:01:56.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c364fc8a597f73dc48ea982b0d931d3d36666eb3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 10:10:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/0de99ebce1ea453aa2020e55677cf514 : SUCCESS in 2h 13m 47s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b55d320b91a24700854d8e1e1f4669bc : SUCCESS in 1h 55m 35s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/80f6a7c3c9454ff4b6768bfc4a99d16e : SUCCESS in 2h 49m 48s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/81f79f07df9948e7bea37ddcf667c325 : SUCCESS in 1h 46m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3e60bcdf16499d5fd64ee61e56e51ccaed2683f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 11:19:18.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/a5c7b2db08a34352980383a1c01aeb44 : SUCCESS in 29m 59s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/4ba1ba6156dc4b8c938c0b03299f17a6 : SUCCESS in 1h 07m 16s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/d8742539edb94917a6890d00bc4028de : SUCCESS in 1h 12m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efa28616244d4c239a46165fffd3d98d : SUCCESS in 6m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/382ab77fb22146289f32145f918c9f0c : SUCCESS in 5m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/75f2d702555b45b38af12d483562376e : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0883dd518b24340bf15be75c203d397 : SUCCESS in 7m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/37e9e4c5bc494440af42ad85f11f002d : SUCCESS in 5m 08s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/495279ac968046d9ac5728d781e876e0 : SUCCESS in 27m 40s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/831f2ff6d37d44368cab3847db724fc6 : SUCCESS in 57m 14s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/87e255ac969d4dd1a30d6c15d9138776 : SUCCESS in 1h 10m 05s","accounts_in_message":[],"_revision_number":2},{"id":"c74c89990d5fb39f52eba4a0020dd2e03efee43b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 11:19:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"3e9b95e6b1ae2a53e3f7d7a6e9a59b7e250d5fd1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 11:20:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e93c075c4a0f4b3ba95f361483feebbe : SUCCESS in 1m 08s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e6d613f8fb504ecdb406110c50670139 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2df836b0fa625db2b5c3fd92dc424bea482b7686","revisions":{"39c262ea89eb26d8f47afb86b73b6da953cf6ea8":{"kind":"REWORK","_number":1,"created":"2021-07-27 09:34:37.000000000","uploader":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"ref":"refs/changes/55/802355/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/802355/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/802355/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/802355/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/802355/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/802355/1"}}},"commit":{"parents":[{"commit":"f28f4a9425b3ac13902bbc3a71e07339874acabd","subject":"Fix prometheus-elasticsearch-exporter container.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f28f4a9425b3ac13902bbc3a71e07339874acabd"}]}],"author":{"name":"Gaël THEROND (Fl1nt)","email":"gael.therond@bitswalk.com","date":"2021-06-17 14:42:42.000000000","tz":120},"committer":{"name":"Gaël THEROND","email":"gael.therond@bitswalk.com","date":"2021-07-27 09:34:37.000000000","tz":0},"subject":"Improve offline build scenario.","message":"Improve offline build scenario.\n\n  * Use distribution based dumb-init package.\n\nChange-Id: I75927b6016bdb7577fbff354632bdab6407f359c\nRelated-Bug: #19311980\n(cherry picked from commit 5940d1758fabf47d15bac5c08d0afa57fbb35964)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/39c262ea89eb26d8f47afb86b73b6da953cf6ea8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/39c262ea89eb26d8f47afb86b73b6da953cf6ea8"}]},"branch":"refs/heads/stable/ussuri"},"2df836b0fa625db2b5c3fd92dc424bea482b7686":{"kind":"REWORK","_number":2,"created":"2021-07-29 07:18:55.000000000","uploader":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"ref":"refs/changes/55/802355/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/802355/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/802355/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/802355/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/802355/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/802355/2"}}},"commit":{"parents":[{"commit":"f28f4a9425b3ac13902bbc3a71e07339874acabd","subject":"Fix prometheus-elasticsearch-exporter container.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f28f4a9425b3ac13902bbc3a71e07339874acabd"}]}],"author":{"name":"Gaël THEROND (Fl1nt)","email":"gael.therond@bitswalk.com","date":"2021-06-17 14:42:42.000000000","tz":120},"committer":{"name":"Gaël THEROND","email":"gael.therond@bitswalk.com","date":"2021-07-29 07:18:38.000000000","tz":0},"subject":"Improve offline build scenario.","message":"Improve offline build scenario.\n\n  * Use distribution based dumb-init package.\n\nChange-Id: I75927b6016bdb7577fbff354632bdab6407f359c\nRelated-Bug: #19311980\n(cherry picked from commit 5940d1758fabf47d15bac5c08d0afa57fbb35964)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2df836b0fa625db2b5c3fd92dc424bea482b7686"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2df836b0fa625db2b5c3fd92dc424bea482b7686"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
