)]}'
{"id":"openstack%2Fkolla~802357","triplet_id":"openstack%2Fkolla~stable%2Fvictoria~I75927b6016bdb7577fbff354632bdab6407f359c","project":"openstack/kolla","branch":"stable/victoria","topic":"bug/1931980-stable/victoria","hashtags":[],"change_id":"I75927b6016bdb7577fbff354632bdab6407f359c","subject":"Improve offline build scenario.","status":"MERGED","created":"2021-07-27 09:37:22.000000000","updated":"2021-07-29 10:45:35.000000000","submitted":"2021-07-29 10:43:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"802357-bug/1931980-stable/victoria","cherry_pick_of_change":796480,"cherry_pick_of_patch_set":3,"meta_rev_id":"c3527d40e4fea5d6a6a5257cdf9b545c7cb945f6","_number":802357,"virtual_id_number":802357,"owner":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-29 10:43:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-07-29 08:40:36.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2021-07-29 09:12:29.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2021-07-29 09:12:29.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-27 09:54:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-28 08:42:16.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-07-28 16:41:51.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"CC"},{"updated":"2021-07-29 09:12:29.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"ae0cb2b2139f7ae96376993fa7b5b11e2bd5d4f5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"date":"2021-07-27 09:37:22.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* docker/base/Dockerfile.j2","accounts_in_message":[],"_revision_number":1},{"id":"b71fcd832c44bf18bbdea49f36bab399254b73c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 09:54:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/4314f6ebea714d7da8968904df6a351e : FAILURE in 5m 50s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/921620020f104faab777c6406edc7d07 : FAILURE in 6m 00s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/83ac6c70e8c2454a8c806f8c4af57196 : FAILURE in 11m 15s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/8d362187c60542fd8a7a5dd476dce17e : FAILURE in 10m 48s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/c384c81a29064e3885e84785e4f62127 : FAILURE in 10m 24s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/57b213ae55aa4e52a9042a535fb975c6 : FAILURE in 4m 39s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/7bd900783c964031bcb3af47ac911d03 : FAILURE in 4m 40s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c5c222eb94a542878de335b3a99e08c8 : FAILURE in 8m 24s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78f275832f23473fba5bf15b0a7a2e72 : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be547203eb0c42d989536ec7eb56d11d : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8156fa67651c4236b5893a2f008201fa : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff4a6e3959ef4a9f9f6bbfd7fb9871a8 : SUCCESS in 7m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c8c8cf8a9e2b4c72ac864d6e06b9d9a2 : SUCCESS in 4m 42s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e48c5228d393454fb697c0d8142b7e06 : FAILURE in 4m 18s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a143ea9463f84343a401662cb3988aac : FAILURE in 4m 10s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/44b04b7557cc4a60989ddef9b51382ea : FAILURE in 9m 15s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/69c04a55e3ed4b8a99c08f5843083af2 : FAILURE in 10m 19s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0576687dc0cf42fc9767c6f0187dd874 : FAILURE in 9m 07s","accounts_in_message":[],"_revision_number":1},{"id":"c3f1dd578b501e8d8cf1326e31fe3bc1e6d648d5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 14:12:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/31456a81096c4d5899f03e546b86ece8 : FAILURE in 9m 18s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f27e33d94d4a4236a82e169cf218f97e : FAILURE in 9m 39s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8ab3c88c744b48dbbb4135f759e08249 : FAILURE in 15m 20s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/b4c814f29f824e0fb4c4415ea53faedc : FAILURE in 5m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7ff412d6fc5152eaaa43fbd08711fb6c95fdc13e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-07-28 08:42:16.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d571c79581c500091f68a3ee5f39f73d74a23bee","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"date":"2021-07-29 07:20:28.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"f55a010ddc2256761369bb53b8d1df6a55173a32","author":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"date":"2021-07-29 07:20:42.000000000","message":"Patch Set 2:\n\nConflicts resolved.","accounts_in_message":[],"_revision_number":2},{"id":"141c45cefc9eee093152b08d531b820eba6711c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 08:30:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/6bb4b2af8ebd47f3a88dbe4b3851b13c : SUCCESS in 49m 35s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/7dce8205802f4f82af99bfd742d4ae10 : SUCCESS in 26m 01s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/8d39b38968a24d8ab80b8c40168e6d2e : SUCCESS in 1h 00m 09s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/76ccc672dff44793a56b2283bede8467 : SUCCESS in 1h 00m 09s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/8d1ee44e5ff44e6cbc90bd0657d1ec06 : SUCCESS in 1h 06m 44s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/6d607ca3f1644ff6943879d779b52b4a : SUCCESS in 22m 59s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/d1dd872e5b52473a90832e97adfbdd8b : FAILURE in 18m 26s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/cde0eddb280b44d3bdb1bbc65fc137d5 : SUCCESS in 54m 05s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35b36933a24c442895da48a161e66cd2 : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b79fdd30b2f447c2b1dafc2d27f43d95 : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d661385e02074f779a86872bf3dabd1c : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c0ff060c85c414c84c69c8f7d0fd68f : SUCCESS in 5m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/76e0ad74e3d04294b4db5ca491b71890 : SUCCESS in 4m 33s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ea0c5ace5abc4a6296c96409b661f398 : SUCCESS in 20m 09s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4eaa9642ab594df59154e5d7b525ffdb : SUCCESS in 20m 29s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7b0ff16e6165451897699b56d86bcf60 : SUCCESS in 48m 14s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f03238589b764c78bd2c401fa7ba0791 : SUCCESS in 48m 54s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8e3a7fa7ccc64e44bf60a0388d9c61a6 : SUCCESS in 1h 00m 27s","accounts_in_message":[],"_revision_number":2},{"id":"72d00395a619661b7359ea07d8d075aa460a4399","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-07-29 08:40:36.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"0e837e99ed4ba6165c1cf8dccbd6b7daf0d79d16","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-07-29 09:12:29.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"4f3a732c65a2f42773b47889d5ce26fe5774590d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 09:12:53.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"45a8b5ecd8125f7303792dc7a68054cea251b874","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 10:00:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/19f3b2f645864465a70f217faf5c79f0 : SUCCESS in 2h 06m 40s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/71708202582f4f1e8f8d4fd34c809de2 : SUCCESS in 1h 47m 11s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/dadec6bf8f204a8abeeb8b6725adb1af : SUCCESS in 2h 38m 02s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/29887474c3cd408098b190aaebc7beac : SUCCESS in 1h 46m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b976bdc1b85ed3402080f959b0ae15c5021a7517","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 10:43:44.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/d0dbbadbbb9f4f1a852324c355d6c213 : SUCCESS in 28m 46s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/72abb8cb60ca4c7cbf0bdcd9c8024ad8 : SUCCESS in 1h 04m 21s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/b8a702756d934f9fa0d6e27a67d4010d : SUCCESS in 1h 16m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3eabcd1f409a4a0d91b66b2b5832921d : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/83eefda9682a42efa486a12fd80dbd4d : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/894fdd8d17654b84897a132e88e7cd8b : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/081a62c8d83f46ba948fca039d3a41fe : SUCCESS in 7m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/133a5416f70f4daab82dec17c17da688 : SUCCESS in 6m 15s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4c722dd5b8b546608a5b0378d2cb11b8 : SUCCESS in 26m 57s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3e7c3e41019146e1bff56c43d1668656 : SUCCESS in 54m 40s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9318cc8f6e944cc08221f73af1cf28b4 : SUCCESS in 1h 02m 58s","accounts_in_message":[],"_revision_number":2},{"id":"11aeb66e609576c9c8bd3f8af95538441ec03102","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 10:43:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"c3527d40e4fea5d6a6a5257cdf9b545c7cb945f6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 10:45:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b4a0721d1cb4710865e756ff47c05cb : SUCCESS in 1m 28s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/462d13fef764410da3ca4e5be4abbddd : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"865e3f1bb848f3dbf64bb2e52dd51bc3209e9719","revisions":{"4c459f7c4a94dcecdbc8700bdd7b0b8c81768ccf":{"kind":"REWORK","_number":1,"created":"2021-07-27 09:37:22.000000000","uploader":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"ref":"refs/changes/57/802357/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/802357/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/802357/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/802357/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/802357/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/802357/1"}}},"commit":{"parents":[{"commit":"b9eac37a6fe71307b78727a8d0bf810698457f26","subject":"gnocchi: add python3-rados to gnocchi-base source image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b9eac37a6fe71307b78727a8d0bf810698457f26"}]}],"author":{"name":"Gaël THEROND (Fl1nt)","email":"gael.therond@bitswalk.com","date":"2021-06-17 14:42:42.000000000","tz":120},"committer":{"name":"Gaël THEROND","email":"gael.therond@bitswalk.com","date":"2021-07-27 09:37:22.000000000","tz":0},"subject":"Improve offline build scenario.","message":"Improve offline build scenario.\n\n  * Use distribution based dumb-init package.\n\nChange-Id: I75927b6016bdb7577fbff354632bdab6407f359c\nRelated-Bug: #19311980\n(cherry picked from commit 5940d1758fabf47d15bac5c08d0afa57fbb35964)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4c459f7c4a94dcecdbc8700bdd7b0b8c81768ccf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4c459f7c4a94dcecdbc8700bdd7b0b8c81768ccf"}]},"branch":"refs/heads/stable/victoria"},"865e3f1bb848f3dbf64bb2e52dd51bc3209e9719":{"kind":"REWORK","_number":2,"created":"2021-07-29 07:20:28.000000000","uploader":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"ref":"refs/changes/57/802357/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/802357/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/802357/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/802357/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/802357/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/802357/2"}}},"commit":{"parents":[{"commit":"b9eac37a6fe71307b78727a8d0bf810698457f26","subject":"gnocchi: add python3-rados to gnocchi-base source image","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b9eac37a6fe71307b78727a8d0bf810698457f26"}]}],"author":{"name":"Gaël THEROND (Fl1nt)","email":"gael.therond@bitswalk.com","date":"2021-06-17 14:42:42.000000000","tz":120},"committer":{"name":"Gaël THEROND","email":"gael.therond@bitswalk.com","date":"2021-07-29 07:20:19.000000000","tz":0},"subject":"Improve offline build scenario.","message":"Improve offline build scenario.\n\n  * Use distribution based dumb-init package.\n\nChange-Id: I75927b6016bdb7577fbff354632bdab6407f359c\nRelated-Bug: #19311980\n(cherry picked from commit 5940d1758fabf47d15bac5c08d0afa57fbb35964)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/865e3f1bb848f3dbf64bb2e52dd51bc3209e9719"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/865e3f1bb848f3dbf64bb2e52dd51bc3209e9719"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
