)]}'
{"id":"openstack%2Fkolla~802663","triplet_id":"openstack%2Fkolla~master~I670fa84476e35e55cbed2d22e48836ed5e28495f","project":"openstack/kolla","branch":"master","topic":"bug/1934753","hashtags":["feature","needs-reno","partof-bugfix"],"change_id":"I670fa84476e35e55cbed2d22e48836ed5e28495f","subject":"Config: allow to set group for user","status":"MERGED","created":"2021-07-28 07:27:59.000000000","updated":"2021-08-17 12:01:59.000000000","submitted":"2021-08-17 10:28:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"802663-bug/1934753","meta_rev_id":"bbf2043ee1b3b74b5f08ba830e58dbc34f62e824","_number":802663,"virtual_id_number":802663,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-17 10:28:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"all":[{"value":1,"date":"2021-08-16 02:51:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2021-08-17 09:08:44.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2021-08-11 09:21:33.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":2,"date":"2021-08-05 14:54:08.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-08-17 09:11:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2021-08-17 09:08:44.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-28 07:28:24.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2021-07-28 08:52:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-30 11:43:36.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2021-08-05 14:54:08.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-08-16 02:51:35.000000000","updated_by":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"reviewer":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"state":"REVIEWER"},{"updated":"2021-08-17 09:11:42.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"}],"messages":[{"id":"ab9655f986d7c2d21540fa1466511d8faaf7aeaa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-07-28 07:27:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b86d700e8751aa43e8717f654dfa5e2f36c61395","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-07-28 07:28:24.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"b96e602cbcfbb0b4660d41bf43a1790a96a76b8c","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-07-28 07:28:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cda675e07b6e747a4ae2a694c66f863cdd11498f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 08:52:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/70370615c12c4eaf969513861178fca8 : SUCCESS in 23m 36s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/7fd70df41d894f4c9d2848df0941ff15 : SUCCESS in 44m 19s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/699450f37676469da533f26ac2abb715 : SUCCESS in 1h 11m 59s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/25537e87469a42089c1bcf6ebe38c691 : SUCCESS in 1h 03m 17s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/a1d3ae1627ec437e9d0a21548c98bebe : SUCCESS in 1h 22m 48s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/bd9b061da10a48c08fc17098e091a01b : SUCCESS in 33m 16s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/462933ca2825483a8979175bda6ff827 : SUCCESS in 20m 35s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/cdfa2bbed2c64431a59cd71cb05ddc90 : SUCCESS in 1h 00m 14s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd0157d9b70e487cbc4aa21b69cab162 : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a5b495a2903940d896525df5c8ebea4e : SUCCESS in 5m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d3fa065b50c446fba46a8abb9ff7971e : SUCCESS in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2cc0715861c3423e97644dd4f179f102 : SUCCESS in 5m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acd9e698bdd24956a09b88263151a714 : SUCCESS in 8m 36s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/fdb6619eae134dbda49f4a8c477967fd : SUCCESS in 20m 55s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/406d5821c4e1480eb905d2ab4a314efe : SUCCESS in 33m 45s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/89a16c466ffd4c21b5146f43af6d5fdc : SUCCESS in 58m 23s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a0c2490f713b44ac9e5674931c5e9761 : SUCCESS in 51m 56s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/27e4d76174cd47f3ba3e18e11663e207 : SUCCESS in 1h 02m 36s","accounts_in_message":[],"_revision_number":1},{"id":"b7680e0a87d9b5e0d4537082550fb288307bd0bd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 09:32:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/00a6c7936ea3402a94a66487580ebd33 : SUCCESS in 2h 02m 13s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c415586dcbfa4a2c93095b94dbec262d : SUCCESS in 1h 21m 30s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/93a061f3118a4de8aec4827579a9091f : SUCCESS in 1h 16m 53s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/978e55d73b314a2098e2ac88e0eb59f8 : SUCCESS in 4m 19s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/659ccba9fd9e456e85b5f106a1f1a8c0 : SUCCESS in 6m 44s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/0767797e6ab44be789882f8afab11b94 : SUCCESS in 1h 52m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"44cc8db662235b759c02eb97a840feb4817a9ca4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-07-28 11:16:44.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2350733961815f57701c09fcb1ad17791ca2c408","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 12:39:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/dd8dae13fcdd445b9e066db8945fa786 : SUCCESS in 36m 15s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/3ca28b8edf094b598273692e85151691 : SUCCESS in 44m 23s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/473e3edf401249609662c2187efc758c : SUCCESS in 1h 02m 26s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/dcbce5c1ba25442d947ba5dba8eb81e8 : SUCCESS in 1h 07m 59s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/299c18f2954e4b0e9985416bbb4a2c4d : SUCCESS in 1h 17m 08s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f8ad7acd828043639f7b17d400b1b0b4 : SUCCESS in 28m 40s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/cc7c85a4d9ae4714afed70fab6163929 : SUCCESS in 29m 25s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/8db81986d8414cbfb01d3dc461a52b67 : SUCCESS in 1h 03m 11s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f345b8478a946c69e0dc0f3c97bab52 : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3c1679ab1b444b4a8ed2a5ac19681508 : SUCCESS in 5m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/42693beef25a4041bbb9d50d2bc17307 : SUCCESS in 4m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/083df385a5ac4bec9a880253c9ccbc82 : SUCCESS in 4m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77b6c1d7a5744822a9919668c3ccfd42 : SUCCESS in 6m 46s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6499e5b695764ac89dea64bd824cbc36 : SUCCESS in 26m 47s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/604c8066a03f401b8716e5ef32208741 : SUCCESS in 41m 02s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3c962085f8224b728ce2ce9185ba85a6 : SUCCESS in 49m 14s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5318ee90ffee4b52a8c7c48d3ea48f38 : SUCCESS in 50m 56s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5825f387bd9e45419679f4d2ae337e3b : SUCCESS in 1h 02m 29s","accounts_in_message":[],"_revision_number":2},{"id":"6297f0b120759b055c6954c8e2e0afd453217e50","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 13:41:37.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/1dd1a3d84f8547b7980125cb767db484 : SUCCESS in 2h 14m 50s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/cf9cac98fbc44ff98a96c9fd6042f31d : SUCCESS in 1h 34m 12s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/38e79568cc6a4d098a1521929f7a3817 : SUCCESS in 2h 22m 46s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/46f7bcc7b6eb42d4b79b6b08cf4b8fb0 : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c31d327745a844539ed29321c536d81b : SUCCESS in 6m 48s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/88974633846b47c694604d5098912c38 : SUCCESS in 1h 34m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8100c9811fc8dc4b8b2af689d7919e7664e9d0b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-07-29 10:22:24.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"c23d4ede910f5d8207abd3b7027ae5fe1062329b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 11:50:06.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/485c018037bc4a26903f93b95d3c422a : SUCCESS in 28m 53s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/de4e645908f74e93b1ab4d6f7aad1ee2 : FAILURE in 39m 15s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/a79546c9e00b45348c5e45cb97f93896 : SUCCESS in 1h 12m 21s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/918e2c4c736a4b978e143e9a50adc47b : SUCCESS in 1h 03m 44s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/32c26032ac084641a5ddc90876fb3ff8 : SUCCESS in 1h 20m 39s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/0fc213e1367442a2a2c8d8f4ee3aa31f : SUCCESS in 41m 17s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/7e73fc0b159a45ea919cb87020be7449 : SUCCESS in 21m 43s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/efea2d0bfdb94dba9002e40377241f06 : SUCCESS in 1h 09m 03s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0108e43f352646079dbb6a9ba4f55a9d : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fc9d29d2a192469eace5e3bd6fff384e : SUCCESS in 6m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a678ce0629904c0590dc3b098421ee0e : SUCCESS in 5m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0ebd1d4b990a4a63a2539cda09bdb728 : SUCCESS in 4m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2da1843707f4476689a8b380cb626497 : SUCCESS in 8m 06s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4e1d086a134b413f9296cb9f0cab69e0 : SUCCESS in 21m 19s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/157ac0e6f60b4d9e99bfbdc20e1f5174 : SUCCESS in 37m 20s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8efe1c367bac4b0eb4ab1c4f5a59ea77 : SUCCESS in 49m 31s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9d78eb8cbdff49cb99807a28a33779ae : SUCCESS in 53m 13s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ba28950d52dd41feb4fa2c38f0d464ab : SUCCESS in 1h 02m 44s","accounts_in_message":[],"_revision_number":3},{"id":"84dd5127a841fcb1291613cc8c746e9ce3ee53c7","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-07-29 11:50:37.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"c80f8249a8d8f9c89ae4a08237e9f0384078fbf2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 13:11:02.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/4767ee483d634c69924e21b4733331fe : SUCCESS in 2h 19m 32s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b302133970bc49fcb607b14d285ab69a : SUCCESS in 1h 27m 05s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d7cf33689784421084d2dbc61b195b49 : SUCCESS in 2h 26m 11s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8237a4a2b1824c0688be0763746b3206 : SUCCESS in 8m 32s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fd02e352272d494da445a66485c87311 : SUCCESS in 7m 13s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/c3b7ef7ec9f246f5b8553bd8add89812 : FAILURE in 2h 16m 28s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e60849b9207c48dce1edb89ab16ff26d4c0c98b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 13:26:58.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/86d2e432d92e4560ba88fbef6a114143 : SUCCESS in 34m 00s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/71d6219da00a456b85eddfe8c6d54dcc : SUCCESS in 44m 14s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/80d31aff8f794e0b8d36a324c8aa196a : SUCCESS in 1h 00m 22s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/b486a9a88b684b3b8bf46f4199953f8a : SUCCESS in 1h 06m 47s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/d5ea31360fbe4cbb9b55abed63785126 : SUCCESS in 1h 34m 27s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/45496d94118642269caa791d97723bea : SUCCESS in 26m 45s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/0f3ee87bfa6a4ba39294097392ebba86 : SUCCESS in 17m 25s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/2a2e312cf44d42cda52a2503251db793 : SUCCESS in 1h 04m 47s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d999698ac6c489b8d15039c94c55c84 : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6ff9d66ca78649d09fb1de164c6768ee : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a7541f37bc648889e00da33a3269e36 : SUCCESS in 4m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3b74ee9bcc4147d78194cfb9abe32750 : SUCCESS in 4m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f402f9fa95794bfab52890016a164942 : SUCCESS in 5m 45s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0b31485f842244728326bb1f4e85e848 : SUCCESS in 20m 09s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4df591e01bc649f48896c86ce2748f31 : SUCCESS in 29m 05s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8cefcd00e7b14e1382dc91f0362babfc : SUCCESS in 51m 53s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ad95d0e86ec2490fb73bbadfaa34801b : SUCCESS in 52m 18s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/288c62e2d8424612bd8cfe4ec5554899 : SUCCESS in 59m 49s","accounts_in_message":[],"_revision_number":3},{"id":"812308891d385e3ea7eea4daa6920c34462d5aba","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-07-30 10:56:20.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"f71cbedd7280516e489ed70d7f13711cc8335cd6","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-07-30 11:43:36.000000000","message":"Patch Set 3: Code-Review-1\n\nneeds a reno (new feature) and a unit test","accounts_in_message":[],"_revision_number":3},{"id":"f8e8a62fd95af89d2f289821366ca2523d0a5683","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-08-02 12:18:29.000000000","message":"Hashtag added: feature","accounts_in_message":[],"_revision_number":3},{"id":"eaed221f85b983d5fbf2660fbe26982f33665d3b","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-08-02 12:18:34.000000000","message":"Hashtag added: needs-reno","accounts_in_message":[],"_revision_number":3},{"id":"b0b689454c6ba67ac49ee620edf1c8d5fc2ba9d3","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-08-02 12:18:47.000000000","message":"Hashtag added: needs-testsuite","accounts_in_message":[],"_revision_number":3},{"id":"c0f8a4b6854eefbeec846605b28c3fb82325af9d","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-08-05 13:49:30.000000000","message":"Hashtag added: partof-bugfix","accounts_in_message":[],"_revision_number":3},{"id":"5b2b1bbccaae4b3d99abf1fe4b0be2773e650889","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-08-05 14:23:21.000000000","message":"Hashtag removed: needs-testsuite","accounts_in_message":[],"_revision_number":3},{"id":"0cc5f50f338f06a5670fc297d999f384e6700d79","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-08-05 14:23:52.000000000","message":"Patch Set 3:\n\nunittest can be skipped as there are no relevant unit tests and I will not be forcing writing them on you; the change is simple enough to understand it\u0027s not breaking","accounts_in_message":[],"_revision_number":3},{"id":"6a7f705a868c99dc17e1998328d7689b0d0f100f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-08-05 14:50:55.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"7eb4f753eff8a254d42b00809e8ac03efbb8b958","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-08-05 14:54:08.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"8d87d747e7f0c0f9d8a10b60cc815f01009c36b2","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-08-05 14:54:39.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0d9c50db7ff4118aa184d670aaea1f1cbe7629e7","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-08-05 14:56:32.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"a66604cb24b2650bec14664d54806688f7aefb01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-05 16:06:05.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/99062148829d4518bf7e113cfb666efc : SUCCESS in 32m 35s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/d76a69abcd4142aa8849d4fe83d11d04 : SUCCESS in 35m 27s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/51ca5ed9759a4921b196afde5f7d800c : SUCCESS in 1h 08m 36s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/58dd2f78526f474095cf69e6a7cfe248 : SUCCESS in 1h 07m 08s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/21d88d04aa8b4ecc82a5f31943067ddd : SUCCESS in 1h 11m 24s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/77757171c08a40b5b1f115a740544b53 : SUCCESS in 30m 15s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/18977036f013499e9f7f4d31735e85a9 : SUCCESS in 18m 08s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/6bac1fc1fc384cd0b380b2ad84992473 : SUCCESS in 1h 01m 02s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/457b45d147344af5963f89027865c0cc : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e4f19723bca4f36a630fce4da3d41b2 : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3ce5677bc8064a149380a2787f25516f : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e9940b20e24d46938f6f6e53ad5fa643 : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab9ef3f59af843de90c0e1c0a824743a : SUCCESS in 7m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f3c4439b9b8648bdbecb97d2c0a17eb6 : SUCCESS in 5m 20s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/aa6935112dec403a9f88fb3aa2061531 : FAILURE in 26m 09s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7ac4a1837ef1420097e1efb41e9f0541 : FAILURE in 31m 35s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9331ced75e5a4a96ab4ae43396d5175c : SUCCESS in 56m 56s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9ee41d7e5b714435aa63be7b46503ee0 : SUCCESS in 54m 19s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d023aaef0d134712a183242c8805faf5 : SUCCESS in 1h 05m 13s","accounts_in_message":[],"_revision_number":4},{"id":"618c676639a2bddaa1b90356df2fe22319882723","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-05 16:59:42.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/4cf0fcabf01a4c318de6ee885e24c610 : SUCCESS in 1h 31m 34s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3e64a3d4273342d58a82b1bb59bcd92d : SUCCESS in 1h 18m 55s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d3c4871867ee4b67a20c796fb6308f1c : SUCCESS in 2h 06m 52s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/23c7c0e902ef450e829b9c52e2826866 : SUCCESS in 10m 44s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/60967d24badd4a63ae61f297c6ad3ad1 : SUCCESS in 12m 00s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/b8859b4e93644e2d9431ba7b33a8bd75 : SUCCESS in 1h 23m 47s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"caa246275009a178577526393586fd00a97e4e19","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-08-06 10:07:21.000000000","message":"Patch Set 4:\n\nintermittent CI failure - grafana apt download, going to recheck","accounts_in_message":[],"_revision_number":4},{"id":"e3c622c47c468e4467689d0484d18e414fd3d6ea","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-08-06 10:07:27.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"f3a0b6599907aadf9bac479aaed356b084ac8ad6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-06 11:28:11.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/604df83c127645268cca0fd1d3f609fa : SUCCESS in 27m 07s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/ffbf3be4ee374e059439f4ad264ddc1c : SUCCESS in 39m 15s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/44804a4f92d64d4382a581f7836b56ad : SUCCESS in 1h 08m 03s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/93c6baa63a2149faa75bf4a6623528d5 : SUCCESS in 59m 31s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/17692de51ebc48a780dcccf45ed6be52 : SUCCESS in 1h 15m 09s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/89cc091c75be4d0d8f4513ea35f111b1 : SUCCESS in 23m 48s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/b1b73d4f01654df894a0048a1c2d35b2 : SUCCESS in 17m 27s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/0e8cb4664aa441a6844a5cc77f297381 : SUCCESS in 1h 00m 14s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d350705b8a143efb15271150a6ae2ea : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/86806dceee634a6e9d46362d8b85bb2b : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0989cba5fb0f4713af938a31409ee077 : SUCCESS in 5m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3246fb33cbbc47de836defb08d8c8ea0 : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ee25014bd7c4ea1a9ed5ee83a3d70d8 : SUCCESS in 5m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/357f3664495e487ba822161021c6b658 : SUCCESS in 4m 30s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8568b4e8572647378b2254764d1b0082 : SUCCESS in 20m 42s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/08b06f82b9d74f0da7df89537c64ada6 : SUCCESS in 30m 21s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/767c2a048dfb4f2e900f7d9e20044a51 : SUCCESS in 49m 13s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6983087a45754a7292551b15946ab249 : SUCCESS in 44m 33s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/608f7d2bbb3842f4ab14c17bbce9d1e6 : SUCCESS in 59m 29s","accounts_in_message":[],"_revision_number":4},{"id":"d884423769e6c497728034fbdf7d5df821fc29be","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-06 12:00:12.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/792fae733cce4fe8851fe670af0e148c : SUCCESS in 1h 50m 33s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/012ce4f0437b4ab4b1f81118f1362c7d : FAILURE in 1h 18m 59s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b33a8754ddca44b6ac4e10ddbc5580ba : SUCCESS in 1h 47m 05s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/537ecc4527da49b68a54e6114d437015 : SUCCESS in 7m 00s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/998291d5ddee4aefb572b5566134c53c : SUCCESS in 5m 09s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/22f2d39c8f494199b99e415c270e1f64 : SUCCESS in 37m 36s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"7e033fdef003afebe8ecb92138b57e32ba92ceba","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-08-11 09:21:33.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"1f1e96d5b6d38476fc769c6916cd7c49a98c03c4","author":{"_account_id":32291,"name":"wushiming","email":"wushiming@yovole.com","username":"wushiming"},"date":"2021-08-16 02:51:35.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"3f3e96cbc0e95d3a4236999b7a71c37d523a1621","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-08-17 09:08:44.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"9c59f7f04d3ef397d92eabf22108c064208540eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 09:09:12.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"4d09307e1c34535a9dc7326222e9520da6e0f182","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-08-17 09:11:42.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"29ccc291d8550ceb816723d59305fb01e89a6f2a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 10:28:05.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/78bea7d1f9db478f8e7df84c291f8a1f : SUCCESS in 39m 42s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/f325084f4bb24d9e8e5d5a96cdebff03 : SUCCESS in 1h 13m 53s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/de12f516a2ee4b888dc9c3f95e101584 : SUCCESS in 1h 11m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8ed5f9ab7a5487a88607bc2373541d8 : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e1e07892b5824e27b673fb13fd51208a : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/279c8cddc6234358a7ad2ae147422bbc : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d27508c738754fe28d9bd1bfcda54a41 : SUCCESS in 7m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c1b6358ca16c4054a7ff5bfb4c4ca6a1 : SUCCESS in 5m 07s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/38a097674017404e80bf6ee817c01596 : SUCCESS in 31m 41s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/fd4f911bc8bf46ae9a6f63a050e962ba : SUCCESS in 52m 53s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8371f667ed064ef893ebf5a77f32c1ff : SUCCESS in 1h 03m 59s","accounts_in_message":[],"_revision_number":4},{"id":"b1491259e242e2472be889ec520f09723ccb662c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 10:28:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"790a4c9eb85826955e92071d0d3c4dfad4ce345f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 10:29:31.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/726ed6b1d26448f9bab018a1512c42ad : SUCCESS in 57s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/92849cb88add4e3e81ba026f57f12a61 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":4},{"id":"c0d964e5f581b86050e1a6f99354c808a3eb65ba","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-08-17 12:00:17.000000000","message":"Patch Set 4:\n\nyou may start backporting to wallaby now","accounts_in_message":[],"_revision_number":4},{"id":"bbf2043ee1b3b74b5f08ba830e58dbc34f62e824","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-08-17 12:01:59.000000000","message":"Patch Set 4:\n\n\u003e Patch Set 4:\n\u003e \n\u003e you may start backporting to wallaby now\n\nDone, thanks for reminder.","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"65684250175117b47d41e29d1bc3f9116c0084b5","revisions":{"9c23d295c36d99f058a425be9591ae8c97fcf2ad":{"kind":"REWORK","_number":1,"created":"2021-07-28 07:27:59.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/63/802663/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/63/802663/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/802663/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/802663/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/802663/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/63/802663/1"}}},"commit":{"parents":[{"commit":"a4a0d40f4b419bc3dbb2e9d420276770d7e0c482","subject":"Merge \"Add masakari-dashboard to debian binary horizon image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a4a0d40f4b419bc3dbb2e9d420276770d7e0c482"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2021-07-28 07:25:45.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2021-07-28 07:26:59.000000000","tz":120},"subject":"[WIP] config: allow to set group for user","message":"[WIP] config: allow to set group for user\n\nAccording to recent IRC discussions is could be useful for haclient.\n\nChange-Id: I670fa84476e35e55cbed2d22e48836ed5e28495f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9c23d295c36d99f058a425be9591ae8c97fcf2ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9c23d295c36d99f058a425be9591ae8c97fcf2ad"}]},"branch":"refs/heads/master"},"498347dc9608f3ac8dc77452e27303b098cb7454":{"kind":"REWORK","_number":2,"created":"2021-07-28 11:16:44.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/63/802663/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/63/802663/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/802663/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/802663/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/802663/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/63/802663/2"}}},"commit":{"parents":[{"commit":"a4a0d40f4b419bc3dbb2e9d420276770d7e0c482","subject":"Merge \"Add masakari-dashboard to debian binary horizon image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a4a0d40f4b419bc3dbb2e9d420276770d7e0c482"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2021-07-28 07:25:45.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-07-28 10:15:41.000000000","tz":120},"subject":"[WIP] config: allow to set group for user","message":"[WIP] config: allow to set group for user\n\nAccording to recent IRC discussions is could be useful for haclient.\n\nChange-Id: I670fa84476e35e55cbed2d22e48836ed5e28495f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/498347dc9608f3ac8dc77452e27303b098cb7454"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/498347dc9608f3ac8dc77452e27303b098cb7454"}]},"branch":"refs/heads/master"},"55ea6d2d7e22404879796681e32415bf102d8e9d":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-07-29 10:22:24.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/63/802663/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/63/802663/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/802663/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/802663/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/802663/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/63/802663/3"}}},"commit":{"parents":[{"commit":"a4a0d40f4b419bc3dbb2e9d420276770d7e0c482","subject":"Merge \"Add masakari-dashboard to debian binary horizon image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a4a0d40f4b419bc3dbb2e9d420276770d7e0c482"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2021-07-28 07:25:45.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-07-29 10:14:57.000000000","tz":120},"subject":"Config: allow to set group for user","message":"Config: allow to set group for user\n\nAccording to recent IRC discussions is could be useful for haclient.\n\nChange-Id: I670fa84476e35e55cbed2d22e48836ed5e28495f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/55ea6d2d7e22404879796681e32415bf102d8e9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/55ea6d2d7e22404879796681e32415bf102d8e9d"}]},"branch":"refs/heads/master"},"65684250175117b47d41e29d1bc3f9116c0084b5":{"kind":"REWORK","_number":4,"created":"2021-08-05 14:50:55.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/63/802663/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/63/802663/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/802663/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/802663/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/63/802663/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/63/802663/4"}}},"commit":{"parents":[{"commit":"ea6ceef686c018d810a2ff20ca4ba81aa77164e5","subject":"Merge \"drop leftovers of RHEL support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ea6ceef686c018d810a2ff20ca4ba81aa77164e5"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2021-07-28 07:25:45.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-08-05 14:46:50.000000000","tz":120},"subject":"Config: allow to set group for user","message":"Config: allow to set group for user\n\nAccording to recent IRC discussions is could be useful for haclient.\nPartial-Bug: #1934753\n\nChange-Id: I670fa84476e35e55cbed2d22e48836ed5e28495f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/65684250175117b47d41e29d1bc3f9116c0084b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/65684250175117b47d41e29d1bc3f9116c0084b5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
