)]}'
{"id":"openstack%2Fkolla~802753","triplet_id":"openstack%2Fkolla~master~Ifff0aa1ec29e94b8067c7e1564b5dbf04fd5c2af","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"Ifff0aa1ec29e94b8067c7e1564b5dbf04fd5c2af","subject":"test-requirements: remove cap on PrettyTable","status":"ABANDONED","created":"2021-07-28 16:24:56.000000000","updated":"2021-07-29 08:16:48.000000000","total_comment_count":2,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"931b86e0b89a0fff39412c285022d9937dc3d3c8","_number":802753,"virtual_id_number":802753,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-07-28 18:12:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":-1,"date":"2021-07-28 18:50:21.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-28 16:33:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-28 17:53:54.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"8bbb32a183074b3b11eec6516b331256466e09de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-07-28 16:24:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3edb98a376c3ec94ae10937bd9e256b7e2f154d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 16:33:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32bb2b2407924804868742e713220a57 : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0a3b490c7f1244718f23a01f07bdb37b : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/72db7bbed38b4a3f886eb60d996556f2 : SUCCESS in 5m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/509a4c48efad4798bc29f1d40162e23c : SUCCESS in 5m 35s (non-voting)\n- requirements-check https://zuul.opendev.org/t/openstack/build/8fc897c372d34077a60ee0c1e50883a3 : SUCCESS in 3m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fe5ff5a5a844ea8a98dbd46d6550c12 : SUCCESS in 7m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/68a551d88c7947f1a36640942df5141e : SUCCESS in 4m 53s","accounts_in_message":[],"_revision_number":1},{"id":"bd9a0231c1c5ad244d8c5b29cf8b193e116fb228","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 16:37:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/64244a5975e94855bc15c72581c89920 : SUCCESS in 5m 52s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/48a66428212545318e4e7ee8082f486b : SUCCESS in 7m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"12bbeb8eda35bdb9a6edee7e0f52ddda84669e1b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-07-28 17:53:54.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1e6fa6dd27308e42c6745513c8b1d00c31493e10","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-07-28 17:54:19.000000000","message":"Patch Set 1: -Code-Review\n\neh, the commit message is missing numbered refs","accounts_in_message":[],"_revision_number":1},{"id":"d37bf1bc9277b2a6c69430924b86e5b53df5b269","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-07-28 17:56:28.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1: -Code-Review\n\u003e \n\u003e eh, the commit message is missing numbered refs\n\ncould you expand?","accounts_in_message":[],"_revision_number":1},{"id":"f43d5b6de2a01ad4193ff46750bc558dd2f91a88","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-07-28 17:58:28.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e \u003e Patch Set 1: -Code-Review\n\u003e \u003e \n\u003e \u003e eh, the commit message is missing numbered refs\n\u003e \n\u003e could you expand?\n\nyou wrote [1] and [2] but they are missing","accounts_in_message":[],"_revision_number":1},{"id":"309059838baf027e6467edeee2c08d2eca8a040f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-07-28 18:00:06.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"6f65ce31c23a08a30cddbbc7a6a06e8f0cf8dc0a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-07-28 18:01:10.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"df739c57001f317f9a759d1a3a0c3fad5229e21a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-07-28 18:01:50.000000000","message":"Patch Set 2: Code-Review+2\n\nI wonder if we use PrettyTable anywhere at all","accounts_in_message":[],"_revision_number":2},{"id":"d78d1b97df100f8742726b90100b893b169e1901","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 18:12:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5454e47e0bed4331a92639e1e3f742b0 : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/459f3b1ea5284f0ba97c60eebc4bf6f6 : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/22e2673805af489a9d2eb7c697d1cf31 : SUCCESS in 4m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2114ac8a190d49f5b47ee6c0e2551c28 : SUCCESS in 4m 41s (non-voting)\n- requirements-check https://zuul.opendev.org/t/openstack/build/c18dd98155fd4a3f9c5cba9f8c2dbd0e : SUCCESS in 3m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecab69bcc1854d1db6051dbd2a2c9460 : SUCCESS in 6m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dabafc107f424ec49bf1a1ef4cdc152d : SUCCESS in 5m 32s","accounts_in_message":[],"_revision_number":2},{"id":"5505a4cbd869c177df446943fdb75a800f101e22","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 18:15:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/673bd618361d451f9d9de8d4ef6f1eb3 : SUCCESS in 5m 52s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dd2cd76101c54e839c1e8b8efb1d70d3 : SUCCESS in 6m 09s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e38e6ed823bba3a4e96ccce846e90706e89a6f6a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-07-28 18:50:21.000000000","message":"Patch Set 2: Code-Review-1\n\nyeah, it can be dropped: https://review.opendev.org/c/openstack/kolla/+/802769","accounts_in_message":[],"_revision_number":2},{"id":"931b86e0b89a0fff39412c285022d9937dc3d3c8","tag":"autogenerated:gerrit:abandon","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-07-29 08:16:48.000000000","message":"Abandoned\n\nwe drop it","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c072bd664cedf30e7afa16f5748e962851c38257","revisions":{"7ca16764dc389f9dd11869f8fea3c4a6261a8fa8":{"kind":"REWORK","_number":1,"created":"2021-07-28 16:24:56.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/53/802753/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/53/802753/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/802753/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/802753/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/802753/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/53/802753/1"}}},"commit":{"parents":[{"commit":"53b391d3612a322ad13e1ec65e2ed9ba7e94a3c1","subject":"Add boto3 as s3 dependency for Glance container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/53b391d3612a322ad13e1ec65e2ed9ba7e94a3c1"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2021-07-28 16:24:01.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2021-07-28 16:24:01.000000000","tz":120},"subject":"test-requirements: remove cap on PrettyTable","message":"test-requirements: remove cap on PrettyTable\n\nopenstack/requirements got rid of it in\n09b8c864b4656b7a58210d75e54457452aaaecc5:\n\nRemove cap on PrettyTable\n\nPrettyTable was capped at a \u003c 0.8, which meant we were getting the\nveritably ancient 0.7.2 release first release in April 2013 (!) [1].\nThe project is now being maintained as a Jazzband project [2], meaning\nwe can remove the cap. The API is pretty much unchanged so this should\nhave little impact for projects that depend on this library.\n\nChange-Id: Ifff0aa1ec29e94b8067c7e1564b5dbf04fd5c2af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7ca16764dc389f9dd11869f8fea3c4a6261a8fa8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7ca16764dc389f9dd11869f8fea3c4a6261a8fa8"}]},"branch":"refs/heads/master"},"c072bd664cedf30e7afa16f5748e962851c38257":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-07-28 18:01:10.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/53/802753/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/53/802753/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/802753/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/802753/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/802753/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/53/802753/2"}}},"commit":{"parents":[{"commit":"53b391d3612a322ad13e1ec65e2ed9ba7e94a3c1","subject":"Add boto3 as s3 dependency for Glance container","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/53b391d3612a322ad13e1ec65e2ed9ba7e94a3c1"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2021-07-28 16:24:01.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-07-28 18:01:06.000000000","tz":0},"subject":"test-requirements: remove cap on PrettyTable","message":"test-requirements: remove cap on PrettyTable\n\nopenstack/requirements got rid of it in\n09b8c864b4656b7a58210d75e54457452aaaecc5:\n\nRemove cap on PrettyTable\n\nPrettyTable was capped at a \u003c 0.8, which meant we were getting the\nveritably ancient 0.7.2 release first release in April 2013 (!) [1].\nThe project is now being maintained as a Jazzband project [2], meaning\nwe can remove the cap. The API is pretty much unchanged so this should\nhave little impact for projects that depend on this library.\n\n[1] https://pypi.org/project/prettytable/#history\n[2] https://github.com/jazzband/prettytable\n\nChange-Id: Ifff0aa1ec29e94b8067c7e1564b5dbf04fd5c2af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c072bd664cedf30e7afa16f5748e962851c38257"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c072bd664cedf30e7afa16f5748e962851c38257"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
