)]}'
{"id":"openstack%2Fkolla~802755","triplet_id":"openstack%2Fkolla~stable%2Fwallaby~I199185e24cedd2e282c53460a24aeffc83478a12","project":"openstack/kolla","branch":"stable/wallaby","topic":"boto3-glance","hashtags":[],"change_id":"I199185e24cedd2e282c53460a24aeffc83478a12","subject":"Add boto3 as s3 dependency for Glance container","status":"MERGED","created":"2021-07-28 16:46:36.000000000","updated":"2021-07-29 11:20:50.000000000","submitted":"2021-07-29 11:19:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"802755-boto3-glance","meta_rev_id":"1e017abec2c4c2460d447243a8555a84011b55ce","_number":802755,"virtual_id_number":802755,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":0,"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-29 11:19:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"recommended":{"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},"all":[{"value":1,"date":"2021-07-29 10:22:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":1,"date":"2021-07-29 07:14:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},{"value":2,"date":"2021-07-29 10:02:13.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2021-07-29 08:41:20.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":0,"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},{"value":1,"date":"2021-07-29 10:02:13.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":0,"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":0,"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-28 16:52:12.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-07-28 16:52:12.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-07-28 16:52:12.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},"state":"REVIEWER"},{"updated":"2021-07-28 16:52:12.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"state":"REVIEWER"},{"updated":"2021-07-28 16:52:12.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},"state":"REVIEWER"},{"updated":"2021-07-28 18:10:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-29 10:02:13.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"}],"messages":[{"id":"6d4b3f27f95f9f639d9449850eccd6b210543c26","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-07-28 16:46:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4205ecc97d17d4553f7169b76102f0e8525ac1c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 18:10:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/a41de8be24e849419cf6c8153c997199 : SUCCESS in 25m 35s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/5c22ee0e37e147cdb205ddc48faf72d6 : FAILURE in 39m 14s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/57c7d3c8a5b74ddc887fa777912ff954 : SUCCESS in 1h 08m 17s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/b41859c7abcf4fc8b80e52b6c9fb756e : SUCCESS in 1h 04m 54s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/beee4d7fe7be4b9cb12b65b1b7930081 : SUCCESS in 1h 18m 41s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f6fb7b82a9624be08d2a786e05c4de52 : SUCCESS in 33m 24s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/cc3acdc7e06c46f38b338f3f5558657b : SUCCESS in 19m 04s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/964333cd133445f7982707bd06dfc5e2 : SUCCESS in 1h 04m 48s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26377d2f9b4347e3a3d756396974b44c : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0143b977dc0d4913916e0db8724be187 : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f1d8f52bb7349bf907e0be015dd09ab : SUCCESS in 4m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5ae58f1c29a842c090562b7a3f69f543 : SUCCESS in 4m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb9320c3f35041d49ee93258600c6437 : SUCCESS in 6m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/325d14684e69470dae626dc3764f0618 : SUCCESS in 4m 59s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9b7a09d696564e909aa83feabeda0ad2 : SUCCESS in 19m 54s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/626302543eb64d1e831473ed6cca4a21 : SUCCESS in 29m 53s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/420e70465c234c0a858b8affef96d96c : SUCCESS in 57m 52s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b06a4c8216c14c0bae96ef548cf99c7e : SUCCESS in 53m 12s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b8c1b25a18f649e48f44503e1ed4ec4e : SUCCESS in 1h 02m 14s","accounts_in_message":[],"_revision_number":1},{"id":"1f494b9af517dab072117dd5b27fc44ebe8530bf","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-07-28 19:04:40.000000000","message":"Patch Set 1:\n\nrecheck ERROR:kolla.common.utils:grafana Failed with status: error","accounts_in_message":[],"_revision_number":1},{"id":"e0fad3d34054783a1988ae13d31db4f1dd606fba","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 19:12:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/a84ae4f0a3c44349bcaa0b7e731c73ab : SUCCESS in 2h 10m 38s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4b3820a3fc07465cb2a16a5037fa25b6 : SUCCESS in 2h 12m 44s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8eb31e587e4a4a82b9b3f531d24d70ce : SUCCESS in 1h 51m 34s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/e8cf42e25046451c9808c8a611e94f81 : SUCCESS in 1h 21m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"19a27622549243ce461eac117a109be6429caddb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 21:15:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/ca2ce1f7153240b8914c32cb5239a534 : SUCCESS in 23m 53s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/6def1ea9743d40769ba8cdc557ca9a50 : SUCCESS in 40m 17s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/71c840f2328d4849a84dac5807f02362 : SUCCESS in 1h 08m 45s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/d9911bfdc2cc41859382398ed1922bf4 : SUCCESS in 1h 04m 01s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/e23fcea2681e4a138871a91afd42a746 : SUCCESS in 1h 15m 26s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f60ffef0235142f1a54fd4607eef6b03 : SUCCESS in 32m 46s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/0844a70f4b554d369225edb6f992ae03 : SUCCESS in 18m 53s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/eee2f585c3e5415e9dc9e76c240bcee7 : SUCCESS in 59m 26s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/885093e71ce2469cb75c435552ecdaee : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/04dc4bf7b3d94b8e9cf1b042284008e2 : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a701604123d245bcb4aee218eb8baa77 : SUCCESS in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/87c3f70e311144049f9cbc59dbcf003a : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aceaba9b13614762ad7cd65b836f6a02 : SUCCESS in 6m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8ff98d3d91ed425c897a7cb83c8379cd : SUCCESS in 4m 50s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1cc51dffb99242acaf7750c8bda795d0 : SUCCESS in 22m 13s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3ad2dc40b64d44f2b655f94ecc01da34 : SUCCESS in 37m 03s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/27cc09b774b54a8fbcaa0da81460c807 : SUCCESS in 52m 28s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7bb461fc44474d4bbd9e836efc3fd7f7 : SUCCESS in 51m 09s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1596fd844d3e45fa810f0c8b03b28ac5 : SUCCESS in 1h 08m 24s","accounts_in_message":[],"_revision_number":1},{"id":"211a0135918ab23d1ec76d1a88b206b74391a112","author":{"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},"date":"2021-07-29 07:14:16.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"9f51ec15ed141a90755341c46826973cd86c1758","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-07-29 08:41:20.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"604e4b1da226655ab0fa8dac779b0ddc82bc7cfe","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-07-29 10:02:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5c5c12a6fe7f10602a9eb0652c61a36cf6721909","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 10:02:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bb3e2a0d11ae2d8c5f9ef9c0354645dd115f46d9","author":{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},"date":"2021-07-29 10:22:51.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"43b82f927fb2b38060ff1a722d725ebc4492632e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 11:19:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/4883b72227264b679d92e03b6f23b92a : SUCCESS in 45m 15s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/028bccfcff4948339b92a0e4b4a45a62 : SUCCESS in 1h 07m 46s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/78f7d390514847c192fda7295dc1c6a7 : SUCCESS in 1h 13m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e0098d5763c40e5bd14881f3c0b0071 : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/28a6c52c98a143729949fe859d89790c : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cf9b04a12f1f458c8d8b09f11e9d4962 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95137173f27c4872b2bdb4bc80a7df97 : SUCCESS in 8m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c19b3422fb1c48f49a7b109bd3fe5a3d : SUCCESS in 5m 48s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f77e8999a41f4ae381d450d688889657 : SUCCESS in 38m 21s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c825f9a5498a45e28399bf4492ad77b4 : SUCCESS in 47m 53s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9d890b758dae4d63803d4a5b6ca09694 : SUCCESS in 1h 05m 27s","accounts_in_message":[],"_revision_number":1},{"id":"220be2b96c5e8885238d76cff72cbbe239aee692","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 11:19:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1e017abec2c4c2460d447243a8555a84011b55ce","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 11:20:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecb3b64fcc98469bbefc673310045d96 : SUCCESS in 1m 02s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/69879c7870904cdab9df93d8d90f4864 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a7c5173b366ff8ebcc1efdbe51f594fef6342ba3","revisions":{"a7c5173b366ff8ebcc1efdbe51f594fef6342ba3":{"kind":"REWORK","_number":1,"created":"2021-07-28 16:46:36.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/55/802755/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/802755/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/802755/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/802755/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/802755/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/802755/1"}}},"commit":{"parents":[{"commit":"83f0f86e6668447b6671e19d1ae842ed13b0aeca","subject":"Merge \"Fix missing templating block for kolla-toolbox.\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/83f0f86e6668447b6671e19d1ae842ed13b0aeca"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-06-21 21:57:16.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-07-28 16:45:15.000000000","tz":180},"subject":"Add boto3 as s3 dependency for Glance container","message":"Add boto3 as s3 dependency for Glance container\n\nboto3 library is needed [1] for glance_store and didn\u0027t installed yet\nsince [2] adds s3 support back. this change adds both binary and source\nparts to install boto3 library.\n\n1. https://github.com/openstack/glance_store/blob/04e5ead7c000211a4c10104ed2bb65c9df7681ae/setup.cfg#L76\n2. I203134837319080ead96da69048baf90086d2117\n\nCloses-Bug: #1884259\nChange-Id: I199185e24cedd2e282c53460a24aeffc83478a12\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit 53b391d3612a322ad13e1ec65e2ed9ba7e94a3c1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a7c5173b366ff8ebcc1efdbe51f594fef6342ba3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a7c5173b366ff8ebcc1efdbe51f594fef6342ba3"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
