)]}'
{"id":"openstack%2Fkolla~802757","triplet_id":"openstack%2Fkolla~stable%2Fussuri~I199185e24cedd2e282c53460a24aeffc83478a12","project":"openstack/kolla","branch":"stable/ussuri","topic":"boto3-glance","hashtags":[],"change_id":"I199185e24cedd2e282c53460a24aeffc83478a12","subject":"Add boto3 as s3 dependency for Glance container","status":"MERGED","created":"2021-07-28 16:48:42.000000000","updated":"2021-07-29 11:35:04.000000000","submitted":"2021-07-29 11:33:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"802757-boto3-glance","meta_rev_id":"311ba0c6363c64da800c4bb3ed97d4df4c643be3","_number":802757,"virtual_id_number":802757,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":0,"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-29 11:33:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"recommended":{"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},"all":[{"value":1,"date":"2021-07-29 10:22:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":1,"date":"2021-07-29 07:14:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},{"value":2,"date":"2021-07-29 10:02:25.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2021-07-29 08:40:52.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":0,"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},{"value":1,"date":"2021-07-29 10:02:25.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":0,"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":0,"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-28 16:51:39.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-07-28 16:51:39.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-07-28 16:51:39.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},"state":"REVIEWER"},{"updated":"2021-07-28 16:51:39.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"state":"REVIEWER"},{"updated":"2021-07-28 16:51:39.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},"state":"REVIEWER"},{"updated":"2021-07-28 18:09:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-29 10:02:25.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"}],"messages":[{"id":"0a89e4fbbc732421dd0b51c116493cc3e1745627","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-07-28 16:48:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d9b74f170e7649a571d0d1daf9901c491271eb25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 18:09:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/e4c84145b5e64c729c923cfee597ea01 : SUCCESS in 25m 36s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/7a93c43b67724c2584ff5f0727b8b2b9 : SUCCESS in 30m 46s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/fa1da64e8f2144eda91e8f65b2634d6e : SUCCESS in 1h 15m 31s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/c8c854872e3c45f4b353c784f514bcb5 : SUCCESS in 56m 58s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/d9419997d5ba447b92104dce39333539 : SUCCESS in 1h 15m 11s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/785486a5ed2a4e3ba42be583c879be14 : SUCCESS in 25m 35s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/c9b8c38b7b6d4c6d9da8a3f74a8df82b : SUCCESS in 19m 43s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f297c55aa06942c3809fb4f1bb1f4209 : SUCCESS in 57m 48s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7501c4e75fa43f9a96dbac84ba5f2c0 : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/85b8eeeb6b154509a9c97e39ce2b9d48 : SUCCESS in 4m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fd3033704e8b4aed802c776af289dc90 : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0a9370c2ec54eb1adaab8c8be9d2e2a : SUCCESS in 6m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/00772b392b6249f380b385b18bc8679a : SUCCESS in 4m 40s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a8fd7d4b60d3478a9a8606339e922963 : SUCCESS in 23m 00s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/67c52acb9ad44d6484550371c2a640f2 : SUCCESS in 28m 14s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5a49729275b2471cbe90b0c69063eaf3 : SUCCESS in 58m 04s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/62c8d396f9214bb799e50844f033ace8 : SUCCESS in 52m 55s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ac6a0823837c4b0c8ab8c9ae09e07ca0 : SUCCESS in 57m 19s","accounts_in_message":[],"_revision_number":1},{"id":"4a5eb4b269989362ea6d55c552dbd88779edeb10","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 19:03:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/15b9106b10df4374b5a206297616beb5 : SUCCESS in 1h 24m 23s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4d713004321c4eb587f6117c328733cc : SUCCESS in 1h 10m 21s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3070dc00eaa149d88dbb0cd2a589bfef : SUCCESS in 2h 01m 09s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/9af16754aa0f4e27b9c3b22802bf838f : SUCCESS in 1h 17m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0a41f3c31acfc9e24287788eca42158136248b81","author":{"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},"date":"2021-07-29 07:14:31.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a796ba743e5ba07dde082f8d0c3e71960e84c9ed","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-07-29 08:40:52.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7eac76408ea4a14004ff5b82d67b936480eb1f07","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-07-29 10:02:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"06086cf8f002b95e276ed13319c2b8da62ac0329","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 10:02:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"143aaf4303874f4fd055ce767a7b0bd566cc361d","author":{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},"date":"2021-07-29 10:22:16.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"873f7013da4daf4c8a6a54fa224d2b57740685aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 11:33:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/66677e92f7c64b6a9ab7ca53f634b4e6 : SUCCESS in 34m 11s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/2cf5b6bed00349289b4b03b99293c86b : SUCCESS in 1h 01m 27s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/650f36b62ef34e948855d620203d0b5d : SUCCESS in 1h 04m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62fa60a35a72425db8ac69c462935dfb : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b7bf27bd9462410daaf35caae8e7b925 : SUCCESS in 5m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a7304aa9f6144ca5809df678e6b61c86 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8c0b1650a5b4b9f9a043658b607cbf8 : SUCCESS in 8m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d11bc4812f214dd2b87ba7eb5c913c41 : SUCCESS in 6m 34s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/68a6f6885ea042b6bdd7825493bc0e4d : SUCCESS in 30m 35s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d2c9dc7314a5494c808909b324edae27 : SUCCESS in 1h 07m 40s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/fd4b6b85b7694e1989381d6b813997e9 : SUCCESS in 1h 05m 34s","accounts_in_message":[],"_revision_number":1},{"id":"28d76c35a61ee330a02b16f86a669595b1489ef6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 11:33:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"311ba0c6363c64da800c4bb3ed97d4df4c643be3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 11:35:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bee552308394d219e704f19584daba2 : SUCCESS in 1m 05s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ea76cf6e3abe43b2bea69939b98fd55c : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9f3660a6c2a51fd75c4b04512fe5a45140b3feae","revisions":{"9f3660a6c2a51fd75c4b04512fe5a45140b3feae":{"kind":"REWORK","_number":1,"created":"2021-07-28 16:48:42.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/57/802757/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/802757/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/802757/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/802757/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/802757/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/802757/1"}}},"commit":{"parents":[{"commit":"09cecc92ecf74527b79b5aab660d059711396802","subject":"Fix missing templating block for kolla-toolbox.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/09cecc92ecf74527b79b5aab660d059711396802"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-06-21 21:57:16.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-07-28 16:48:29.000000000","tz":180},"subject":"Add boto3 as s3 dependency for Glance container","message":"Add boto3 as s3 dependency for Glance container\n\nboto3 library is needed [1] for glance_store and didn\u0027t installed yet\nsince [2] adds s3 support back. this change adds both binary and source\nparts to install boto3 library.\n\n1. https://github.com/openstack/glance_store/blob/04e5ead7c000211a4c10104ed2bb65c9df7681ae/setup.cfg#L76\n2. I203134837319080ead96da69048baf90086d2117\n\nCloses-Bug: #1884259\nChange-Id: I199185e24cedd2e282c53460a24aeffc83478a12\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit 53b391d3612a322ad13e1ec65e2ed9ba7e94a3c1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9f3660a6c2a51fd75c4b04512fe5a45140b3feae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9f3660a6c2a51fd75c4b04512fe5a45140b3feae"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
