)]}'
{"id":"openstack%2Fkolla~802845","triplet_id":"openstack%2Fkolla~master~Ia42c8c630c049b3e28c032799528265879b161fa","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"Ia42c8c630c049b3e28c032799528265879b161fa","subject":"sync requirements with openstack/requirements","status":"ABANDONED","created":"2021-07-29 08:56:10.000000000","updated":"2021-08-02 12:18:01.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"667cb49759bfa4a48f513f9141ca7685aa597ca2","_number":802845,"virtual_id_number":802845,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-07-29 10:29:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":-1,"date":"2021-07-29 09:04:57.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-29 09:04:57.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-07-29 09:37:31.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"CC"},{"updated":"2021-07-29 10:29:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5476f46fc334f67a93294490777d1a8b999dbdb5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-07-29 08:56:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f73a1f6a1682124cef537557848c15261e23c418","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-07-29 09:04:57.000000000","message":"Patch Set 1: Code-Review-1\n\nOur requirements are the minimum we need. Upper constraints are community-wide maximums. They don\u0027t need to be the same.","accounts_in_message":[],"_revision_number":1},{"id":"f9930a6fbee707ede6b721c83c18669fc218abad","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-07-29 09:24:51.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1: Code-Review-1\n\u003e \n\u003e Our requirements are the minimum we need. Upper constraints are community-wide maximums. They don\u0027t need to be the same.\n\nAnd one day CI will enforce newer versions that users have and things break as we did not notice that we used deprecated ways.","accounts_in_message":[],"_revision_number":1},{"id":"116fb5550a5edaba65acf39f38782b6c72687646","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-07-29 09:37:31.000000000","message":"Patch Set 1:\n\nWell, we need to bump from time to time. Just perhaps not to the latest ones?","accounts_in_message":[],"_revision_number":1},{"id":"124fd76597c18ef1a8002a40c76818e8e6a3f023","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-07-29 09:40:53.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e Well, we need to bump from time to time. Just perhaps not to the latest ones?\n\nLet us see what Zuul say.","accounts_in_message":[],"_revision_number":1},{"id":"9a4f516c1844e76e47637096a9d57c182b692e60","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-07-29 09:44:19.000000000","message":"Patch Set 1:\n\nIt\u0027s not a great argument for kolla, but in general, this wouldn\u0027t give distros much scope to package kolla, since we specify an exact version.","accounts_in_message":[],"_revision_number":1},{"id":"2374b38918f540d8ae918b12117f5e20721ce4e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 10:29:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/7db362c4b4784fb88684fc18449d128b : SUCCESS in 32m 18s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/d6b94b43cec14f0094e8251096a3a04a : SUCCESS in 51m 44s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/4fadc238da424258bc7b414dcc21936d : SUCCESS in 1h 07m 54s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/34bb9001e6e049bfa42a86627300cd48 : SUCCESS in 1h 02m 21s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/fc0af9dcd37845dbbf95e43946bedd35 : SUCCESS in 1h 25m 25s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9108df578b1749f2abbd450bb0c51208 : SUCCESS in 39m 42s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/6939d04e35484f97be6ae423564cae6f : SUCCESS in 34m 15s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/9748da1907f54808a16f271cbc95cc30 : SUCCESS in 1h 10m 57s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80536ed358924740a2d86490706d3da8 : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/740d112fc08b4e80b3b73de8eee1319e : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d6a6c4ed30db46c380ab92535815a635 : SUCCESS in 4m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/82829d3e052f451c80548f5b2a1f60c8 : SUCCESS in 4m 34s (non-voting)\n- requirements-check https://zuul.opendev.org/t/openstack/build/c51973791af4415889be3fcbce4846e5 : SUCCESS in 3m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b543ff5f833d49d5bf62a58617765e9a : SUCCESS in 7m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1ba41f5368354ac9ba9ad303ec974393 : SUCCESS in 4m 47s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e92f78d110d44710b5ec5f586980655f : SUCCESS in 46m 12s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/705ad398f78447e4ada18ea815335cdd : SUCCESS in 31m 52s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9b96fa8e3b47459ca630718a8aa5d43c : SUCCESS in 51m 31s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d185e25cc8fd4144bc3928a517345ef1 : SUCCESS in 49m 28s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a6d93c173c5d4f85aab655adb52596d9 : SUCCESS in 58m 26s","accounts_in_message":[],"_revision_number":1},{"id":"b05638c39d31cad3175c4e27115f0d09c539d34c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 13:46:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/63c07f6016da4ffe8849c60194f943be : SUCCESS in 2h 06m 48s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a8f67a5af68a4c35a6466f04fedc8fb1 : SUCCESS in 2h 00m 41s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/cf0743fa0a684c6b8b38439aa1b9b2c3 : SUCCESS in 2h 17m 05s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f0021eb9825644eeafe304a596a15f5a : SUCCESS in 7m 00s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/63272dd73e6240f6b458f72f8f12116a : SUCCESS in 7m 29s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/7b70a6932606461f8d6efa35bc0e7b9f : SUCCESS in 2h 02m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2556b08279f92de61051fa5659ccb77bc86d5176","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-07-29 14:28:14.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e It\u0027s not a great argument for kolla, but in general, this wouldn\u0027t give distros much scope to package kolla, since we specify an exact version.\n\nWe already require versions not present in Debian \u0027bullseye\u0027. And this distro was not yet even released.","accounts_in_message":[],"_revision_number":1},{"id":"667cb49759bfa4a48f513f9141ca7685aa597ca2","tag":"autogenerated:gerrit:abandon","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-08-02 12:18:01.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"78d95a91a2574af11fc28ddea16520bdbe56f790","revisions":{"78d95a91a2574af11fc28ddea16520bdbe56f790":{"kind":"REWORK","_number":1,"created":"2021-07-29 08:56:10.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/45/802845/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/45/802845/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/45/802845/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/45/802845/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/45/802845/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/45/802845/1"}}},"commit":{"parents":[{"commit":"5e69c63318aab4e0a70b411860553dc2e7bab743","subject":"test-requirements: remove not needed entries","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5e69c63318aab4e0a70b411860553dc2e7bab743"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2021-07-29 08:55:09.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2021-07-29 08:55:09.000000000","tz":120},"subject":"sync requirements with openstack/requirements","message":"sync requirements with openstack/requirements\n\nBumped all versions to current upper-constraints.txt ones.\n\nChange-Id: Ia42c8c630c049b3e28c032799528265879b161fa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/78d95a91a2574af11fc28ddea16520bdbe56f790"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/78d95a91a2574af11fc28ddea16520bdbe56f790"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
