)]}'
{"id":"openstack%2Fkolla~802926","triplet_id":"openstack%2Fkolla~master~Ie7f26fa016d6fcafc6eee860b827bca868b4c487","project":"openstack/kolla","branch":"master","topic":"remove-patrole","hashtags":[],"change_id":"Ie7f26fa016d6fcafc6eee860b827bca868b4c487","subject":"Remove patrole from version-check","status":"MERGED","created":"2021-07-29 18:43:30.000000000","updated":"2021-08-03 08:08:53.000000000","submitted":"2021-08-03 08:07:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"802926-remove-patrole","meta_rev_id":"09afbb70179615df684278e84b2071abe5c1fd60","_number":802926,"virtual_id_number":802926,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-03 08:07:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-08-03 06:50:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-07-30 10:03:29.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-08-03 06:50:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-29 20:08:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-30 10:03:29.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-07-30 11:13:27.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2021-08-03 06:50:23.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"65232f98ce4c439487c98012306b0ea4366f7123","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-07-29 18:43:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"01acc166e192254e5b0c8aee70a0d68a1ee0c011","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 20:08:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/50d77cbee8a94f158e61549e8c684e85 : SUCCESS in 38m 38s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/0a4eb066f6eb451c9226a7f1ab5733ba : SUCCESS in 42m 37s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/666ecf49ca054ea0923b217ad0b637dc : SUCCESS in 1h 08m 28s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/f0b947aa2ec24218b94af610f9a37975 : SUCCESS in 1h 12m 32s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/51ffe45e4ec445719cf94fa3283a5dfa : SUCCESS in 1h 22m 49s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/2e14636dabcb4b31a33b07281767ac53 : SUCCESS in 25m 38s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/79502a41f989488f8ed6f44ec1f9514f : SUCCESS in 21m 16s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/baf058775acb472e8c289963b9f244ef : SUCCESS in 1h 03m 45s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36c8ee0d45894dd1a7103f2056b27666 : SUCCESS in 7m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9878a874228643a1917e6335dec767f8 : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3331f4bedb594c24bf803166413ff910 : SUCCESS in 6m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/77e00340005142c2876ff65a76cb702c : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b69358cbc2b477b831ecbb1feb81748 : SUCCESS in 6m 42s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/27c44dbe1c9845028def34fa1bae64bb : SUCCESS in 21m 57s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f197c1aad229459d90cd1d01be861e15 : SUCCESS in 27m 35s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7071556e7efe4f4f93b950b694ec48ba : SUCCESS in 53m 18s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2045649a79234aa8bfe084c4a6e5965c : SUCCESS in 56m 40s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e84f5196369c4bada047e07f6b7dc415 : SUCCESS in 58m 06s","accounts_in_message":[],"_revision_number":1},{"id":"895f20b780f8608a4d19da98b1ae09bb3a0cd01b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 20:31:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/976afab33def499d8e690f9fb8625cbd : SUCCESS in 42m 28s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/cc3c3e9a78ca4dc1a0d3331e54f1ab54 : SUCCESS in 43m 03s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2baa5f3c22b44feb8fe73a53a1277ff2 : SUCCESS in 1h 45m 36s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3f14f0bf07074fc78fa87fe9b28a8f4a : SUCCESS in 4m 40s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a8c3fe18c9d5462e8aad062b825cdc41 : SUCCESS in 7m 13s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/baf0b9a887914462857514949ffcddea : SUCCESS in 42m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a63a5c2c3626acb74c2604c697871506f3ce58e1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-07-30 10:03:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b4dd24c8dd7dff3bab38bc00ce6fddceee7dc734","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-08-03 06:50:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"004157e3dbd4f913c57af35a1379ec113b2368b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-03 06:50:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"65e36b66311c247c6056e9dbc0ec8488fc89190c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-03 08:07:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/614965dddf414f8b97f135637ab86faf : SUCCESS in 38m 24s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/d678410af8294d0293551b7d51c3c75a : SUCCESS in 1h 09m 09s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/0ba2c6258346416e92c7f456472ef690 : SUCCESS in 1h 15m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f196f389ea34331935db9c4c2a8cda8 : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/465fe2b9669548caac738eee95899ca8 : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/df3a94fed1414715ad89d987e2b9c867 : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c1a62257a0e4be7a1666da3be0a9c8d : SUCCESS in 6m 21s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/cf4ba9d6127943be8273b75e54a34d73 : SUCCESS in 28m 49s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b7653e4e87984d618439fae0315f6951 : SUCCESS in 53m 42s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2f3f3248904f4cb6be111544189885de : SUCCESS in 1h 03m 27s","accounts_in_message":[],"_revision_number":1},{"id":"078832f4cac10e35eadb22b0815cc38939466cca","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-03 08:07:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"09afbb70179615df684278e84b2071abe5c1fd60","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-03 08:08:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62ebe706a2f24c2b9fb6ff8ea072a121 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f8824b60d5c3f4af4cddabcb868021abe2182645","revisions":{"f8824b60d5c3f4af4cddabcb868021abe2182645":{"kind":"REWORK","_number":1,"created":"2021-07-29 18:43:30.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/26/802926/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/26/802926/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/802926/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/802926/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/26/802926/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/26/802926/1"}}},"commit":{"parents":[{"commit":"b2cfc361b9c5f8dbb36dd41060af763618e1bd18","subject":"Merge \"Remove tempest image\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b2cfc361b9c5f8dbb36dd41060af763618e1bd18"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-07-29 18:42:42.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-07-29 18:42:42.000000000","tz":0},"subject":"Remove patrole from version-check","message":"Remove patrole from version-check\n\nPatrole was used by tempest which is now removed.\n\nChange-Id: Ie7f26fa016d6fcafc6eee860b827bca868b4c487\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f8824b60d5c3f4af4cddabcb868021abe2182645"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f8824b60d5c3f4af4cddabcb868021abe2182645"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
