)]}'
{"id":"openstack%2Fkolla~803617","triplet_id":"openstack%2Fkolla~master~I359982ddd144e07f8296b384638cf0b5331e495c","project":"openstack/kolla","branch":"master","topic":"avoid-force-in-groupadd","hashtags":[],"change_id":"I359982ddd144e07f8296b384638cf0b5331e495c","subject":"Avoid the use of --force in groupadd","status":"MERGED","created":"2021-08-05 15:00:52.000000000","updated":"2021-09-30 20:23:56.000000000","submitted":"2021-09-30 20:22:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"803617-avoid-force-in-groupadd","meta_rev_id":"36480260f1e28639e20df6440cc9001359f402f6","_number":803617,"virtual_id_number":803617,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-30 20:22:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-09-30 18:46:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-09-28 08:23:04.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-09-30 18:46:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-05 15:05:44.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2021-08-05 15:28:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-11 09:51:24.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-08-23 13:06:55.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"5bc3f3d6a2913538b6a152d3c50f06ec7d4c5221","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-08-05 15:00:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9d1a8f116d9b2ed9a91cacfe9b394d1d195c343c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-08-05 15:05:44.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"24278cbb37d27e130a2f438b77018a909f1efe68","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-08-05 15:06:26.000000000","message":"Topic bug/1934753 removed","accounts_in_message":[],"_revision_number":2},{"id":"18c6f525604c2ed4bb97d8d3a4ab94c64d9fe255","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-08-05 15:07:44.000000000","message":"Uploaded patch set 3: New patch set was added with same tree, parent, and commit message as Patch Set 2.","accounts_in_message":[],"_revision_number":3},{"id":"c517bb651a65cf9b44fca7c8a7cb7d2b406e09fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-05 15:28:17.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/ef80cda58b484a4fb2299c062e09965b : FAILURE in 7m 00s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/e182b72ac9fb4f079255b14e85fb4557 : FAILURE in 7m 50s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/d2879d484ebd477583f04730efc9d464 : FAILURE in 12m 14s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/a12a521087644218bbb4f18feb8519d1 : FAILURE in 12m 03s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/4a1cf8193deb4bd89f4f33e1757c6bcd : FAILURE in 13m 52s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/7f8f49a4dde04bf58cc91f66caa2ea78 : FAILURE in 5m 48s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/d6a77a64ed3b46febaa82319b95cce30 : FAILURE in 4m 57s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/556c78a00a7b4a39b78a50eb8844775c : FAILURE in 16m 31s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/930a4220471e4d50b5363866b521b75a : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b54267ef1fe04d56aaf9f3eba06a6392 : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b85d7e03a834ad5954ea3f5cf8e23cc : SUCCESS in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/56dae746c53f48dbaf2d29a41de45d32 : SUCCESS in 4m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37ff2b00ec044ae6a069ae48b241472b : SUCCESS in 7m 27s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/494422e454584a86b333a09fa7a8dacd : FAILURE in 5m 28s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/1b293db410ba47acbf99c2d0abbee3f9 : FAILURE in 6m 19s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7699c05e8e0a40cda10a1eeef9616cae : FAILURE in 10m 14s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/80dad74ac6cf47e780ad367facf25175 : FAILURE in 11m 41s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/96989a6126d34904a6947e3f271b2c0f : FAILURE in 10m 28s","accounts_in_message":[],"_revision_number":3},{"id":"0a6bf98faf609ce3a5ba20255fef4cde421aa3e7","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-08-05 15:36:29.000000000","message":"Patch Set 3:\n\nwow, turns out we had a little bug already","accounts_in_message":[],"_revision_number":3},{"id":"fce33532b66f2f52d70c925df2a8378a79d1252f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-05 15:42:22.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/bcec96b1c85340fa9654236244480afd : FAILURE in 10m 37s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4ffd639799674390a35acee26686002f : FAILURE in 7m 24s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9c195918289848149ee92d5438839781 : FAILURE in 22m 36s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/03e390784ac84425a6bfdb0b6b668629 : SUCCESS in 6m 12s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ecf41786f2b041eab68dc9b265596f5c : SUCCESS in 5m 30s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/bec0fe705e9b470eb8d11d1d9d7f43e1 : FAILURE in 7m 26s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"fa9e7ba41bfea7ffda1b6bead57644d6b6cd7bbf","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-08-06 10:10:03.000000000","message":"Patch Set 3:\n\ngroupadd: GID \u002765534\u0027 already exists in CI","accounts_in_message":[],"_revision_number":3},{"id":"6f1ba7333205233ff132bf9d1ece9bd7ff50467b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-08-06 11:16:24.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b09fcb8b5bab85a55c3c911a89bd7ce1e6ea94e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-08-06 12:06:03.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"ccabad748273b6b7d257e069315329795b49b842","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-06 13:24:25.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/51403fc71e88446fb1e96e5a0858f49d : SUCCESS in 23m 18s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/187f579280a3455896fc800ef933cd32 : SUCCESS in 28m 33s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/d2e3a256a26744f59898c7da2a9d0d6e : SUCCESS in 1h 07m 47s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/91977c2bd4944aa3b14816874e1f46cb : SUCCESS in 59m 27s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/7b30f64de6f74c7faa0070860b495e4f : SUCCESS in 1h 16m 37s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/976c7ab701f846e59b29c73d6aacf0f7 : FAILURE in 25m 00s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/6c48a64564c64300ab287bd966c1df8f : FAILURE in 16m 55s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f9ef381c6b934665a7d3e519f3437310 : SUCCESS in 53m 16s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b2a4f1cd15844278847d3f4bcbed659 : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5f0a870b0d5e49578ea6f1a8c86efe9e : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a6f9422da154494eab6a84ed199028fd : SUCCESS in 4m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/812b040fee4348e6a33004595a68a7d4 : SUCCESS in 5m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8680212f49d348ed978c0d0427f6a84f : SUCCESS in 6m 51s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/934be52148e04adca307cca0faac17b3 : FAILURE in 21m 24s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e571601a52464f58b3f67adf333a018d : FAILURE in 35m 48s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/694192e8e933403baf222cee43d1146d : SUCCESS in 48m 44s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/bdba953a401445de8f7cf453839d7b9a : SUCCESS in 53m 40s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3534b8f374c74e0085e8bdf334c06907 : SUCCESS in 1h 05m 34s","accounts_in_message":[],"_revision_number":5},{"id":"89e70396be3aa42daab2dc35ce35187843cced67","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-06 13:42:54.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/c1e7a949596744afa877e365d97523d2 : SUCCESS in 58m 32s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3693f91c834e45f49ba48e40906bcc8b : FAILURE in 1h 28m 21s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/708dbb8b41f04cedbdc0964c4127e937 : SUCCESS in 1h 31m 28s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8d3243d043e3494fb4270510a84fed28 : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1566001eff9d452f9bdd9cc24d4e9d66 : SUCCESS in 6m 57s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/1ae139d29b9d46b7bb9819c28720fd83 : FAILURE in 47m 30s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"3d70b793a21b713450cc8e0a3ba91acc227136f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-08-09 12:25:31.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"dc06c753fd1c60ad675b28666af1370acd45f4ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-09 12:45:11.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/d76ef6776a8b47f68ddc0ed8e4a25abf : FAILURE in 7m 57s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/51d8d0a7c771425ea67b39f51219979d : FAILURE in 6m 42s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/106d6f70d820414bad6a24538f5b91d0 : FAILURE in 13m 15s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/7db4803c51da44c6908d31c8b3bba980 : FAILURE in 13m 15s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/52391cf0573d47abb23e129181adcd63 : FAILURE in 10m 37s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/65e4c0c1963c497d9c5d4067b6255f6d : FAILURE in 5m 19s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/e19c3ebdc77b49cc954d8f33865a385f : FAILURE in 5m 02s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/38fcd52ee09c42689fa33cbfb397c35d : FAILURE in 14m 29s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14aa8fb504e641e8973e3f99afe959cf : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bf62b288f98542009fad5fd3ec9ad99c : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/936d43426e4f48d78ead2362ab78ec1a : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/19b5ced9c8c6457cb9e8ffdba963b562 : SUCCESS in 4m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a1b1a42c7614e50b906f1f69a3f34fe : SUCCESS in 6m 25s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/22b4677ed28340d3937222fc27a7f1ff : FAILURE in 4m 45s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a322933dbf244330838e5b669e8b83a1 : FAILURE in 5m 03s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/19b4811019d645e092bd3d421e14d583 : FAILURE in 10m 22s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/91f3517ed2b24d4aa0f9f17e47a4bf29 : FAILURE in 10m 16s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/266e5fc5d99442dc97befe305833ace6 : FAILURE in 8m 53s","accounts_in_message":[],"_revision_number":6},{"id":"c6644d4c89f33b84c5a8b44e6971c71e3ab3f0f3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-09 12:50:18.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/fb740837019342c5abc90ad4b1d8cda1 : FAILURE in 11m 40s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b44df66ba37b4a6e85212f3ad586926a : FAILURE in 8m 48s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6b3e5c44120b47a29472575776ca1c1c : FAILURE in 20m 13s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b40e1b78513848deb64a43a5a396d732 : SUCCESS in 5m 35s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7dbd1fd0c36247f39826905521587b42 : SUCCESS in 6m 33s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/741db67189804af48f63ee18d7949370 : FAILURE in 7m 05s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"280f9ccefc088e5f56337b7dbfb9e2c6682c0f99","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-08-09 13:35:53.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"645d39d7563a529543d3626dbb09e2347aa4606b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-09 14:59:50.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/f33df915e4724d53af327f39d49ffeb5 : SUCCESS in 43m 32s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/764c0b5702aa4153b1382c86142e681b : SUCCESS in 55m 03s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/9ae3260a80b04543abf4380b40a2430f : SUCCESS in 1h 03m 56s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/5879a72928cf4647baa4726d85e3622f : SUCCESS in 1h 03m 33s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/0df48f2725c1443581ccba67393df89d : SUCCESS in 1h 13m 51s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f714cfd53de743f5a3fc92af37524513 : SUCCESS in 58m 09s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/7a03e8f4234e408ea8955fbdde29eae9 : SUCCESS in 33m 15s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f200d36eb752414598c2312c2f0417fc : SUCCESS in 1h 03m 51s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0d6d897c9b0492fb1e5c5b1f33c9a53 : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7d17fc1174294f7585be2311a3cdc5eb : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4e027d00156d47819568b8fdd92f3ed0 : SUCCESS in 4m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/61d911fdc8ed4cb5b50f672fb0783d45 : SUCCESS in 3m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d549c41f21847c2b28b9cf2b39fc43c : SUCCESS in 6m 39s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/04df115b67054572b8e5142c928a0cc9 : SUCCESS in 44m 32s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/97c56b105117491ebd47d266aaa8484b : SUCCESS in 52m 16s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9ecb824b133643f69f707f9a76c0d27c : SUCCESS in 51m 06s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f60f10c6a7da4543bc379e27be70a6de : SUCCESS in 55m 39s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d4546cb1d1944396b52add6ead587f4a : SUCCESS in 1h 07m 23s","accounts_in_message":[],"_revision_number":7},{"id":"143f0d7b17f2c64d88c26a92b4cf72d1821e99b3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-09 15:10:49.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/097a064df745437381444e71652ab9da : SUCCESS in 1h 20m 51s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/16692545eb044338a493bab24f20a799 : SUCCESS in 56m 17s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f051a592333e409f9d30a2b6cefa3de9 : SUCCESS in 1h 28m 38s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f2626cea7ff44828adcf203d6fdcf640 : SUCCESS in 4m 55s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0ff739f0074c42b59d85644c7638298d : SUCCESS in 5m 04s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/30e13fd06dba41b897435c4d363d954e : SUCCESS in 1h 03m 58s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"463b6e9b2c8f1d608761b6a74d2431ff21b88b30","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-08-11 08:58:34.000000000","message":"Patch Set 7: Code-Review+1\n\ncan\u0027t really +2 my own change on master","accounts_in_message":[],"_revision_number":7},{"id":"da142742cb4dea41468434de61043bbd8af87a6f","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-08-11 09:21:05.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"d71487a99071c66d8374d4d4a2cffc67222b2826","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-08-11 09:51:24.000000000","message":"Patch Set 7: Code-Review+2\n\nSo, I feel it\u0027s one of those fixes we will need to fix later - but maybe it\u0027s not.\nThere\u0027s a bug attached, so I guess it would make sense to make a reno - unless there\u0027s a ,,joint\u0027\u0027 reno of all those fixes in the chain.","accounts_in_message":[],"_revision_number":7},{"id":"1099226f6d4d99f0fa7f81b4e7d73d41cf3710bc","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-08-11 10:10:00.000000000","message":"Patch Set 7:\n\n\u003e Patch Set 7: Code-Review+2\n\u003e \n\u003e So, I feel it\u0027s one of those fixes we will need to fix later - but maybe it\u0027s not.\n\u003e There\u0027s a bug attached, so I guess it would make sense to make a reno - unless there\u0027s a ,,joint\u0027\u0027 reno of all those fixes in the chain.\n\nThis part is only a proactive application of the recommendation from manpages. It is not critical but ensures we actually don\u0027t step into something nasty along the way. 😊","accounts_in_message":[],"_revision_number":7},{"id":"8def8314a513aefae266448ec53bffae4c1d1679","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-08-25 16:07:55.000000000","message":"Patch Set 7: Workflow-1\n\nlet\u0027s wait for bifrost testing here as well","accounts_in_message":[],"_revision_number":7},{"id":"d1a375f863f29bf9c88007c7e687c3a0c875a88d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-09-27 07:24:47.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"3df21e4609a77daea83134e5ab55c1ff583514ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-27 08:59:59.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/6e7032294f76486389c5711467b27352 : SUCCESS in 29m 29s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/27c50cf921c34c3d87a1ed3dccb47678 : SUCCESS in 40m 41s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/d8ffdb680249415f853d3755aa2c2bd7 : SUCCESS in 1h 15m 24s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/d5d8457c19904898ac1ad02d21c75db6 : SUCCESS in 1h 15m 50s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/ccc75cecd6d84274928e1f75cb956300 : SUCCESS in 1h 33m 38s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/2d46c390076e4062a56150a0e2eedbf8 : SUCCESS in 32m 53s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/28059e3d2cfe464a814158bfa2aefe9e : SUCCESS in 19m 49s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/594bf5c85f3743feb537312548361e43 : SUCCESS in 59m 35s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c3cdedfc23c4e04af90e86e472aeb84 : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5913ab44863b4a9f8e101b0f0a41776b : SUCCESS in 5m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/71814d257c05405bacc119098ced92f2 : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c85365d7e6e54898a737826a7bd034a3 : SUCCESS in 4m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38a70c73f62c4a129d26968d283a551a : SUCCESS in 7m 11s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ba8320b079d3458791b88f0c035250b7 : SUCCESS in 20m 20s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/5c6d2f75846b4da0adda6c31c0cc8c58 : SUCCESS in 31m 05s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/cc57b743be7949fbb9f7a08b7e0cf889 : SUCCESS in 46m 12s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3dac2256878641f5a4bef936724e8992 : SUCCESS in 51m 21s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3a3158e8778a46d595e20060d4a735af : SUCCESS in 1h 08m 14s","accounts_in_message":[],"_revision_number":8},{"id":"79533d39ea27174f4169c65dfc668c641e75440e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-27 09:34:46.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/fd477140feb7443e9c3e098cb7c4d09f : SUCCESS in 1h 49m 33s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2aed8c8c6a4c4ae399a88ec663b72b57 : SUCCESS in 1h 12m 31s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/86192205dd674f01a281e1f04ce13b21 : SUCCESS in 1h 54m 54s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d099bbeb21994259b3b76253cab370ec : SUCCESS in 4m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/66ddd9b25cae48bca5e3ff0ced66fe73 : SUCCESS in 6m 07s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/a10b1b38bd154e55a26454dcee0be3a6 : SUCCESS in 39m 05s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"159ecb1b46c474463c0bfb06ba28a9b4226b4e14","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-09-27 12:19:52.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"679a7675af0400c4f2226edd0e1a9ff996b3b568","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-27 15:39:09.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"9d35261a62bed85640ddb119dd64b72e799154db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-27 15:39:45.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"c71729f379b3740afb9d7e6feaa655ee0171edeb","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-27 15:40:14.000000000","message":"Patch Set 9:\n\nok, ready now","accounts_in_message":[],"_revision_number":9},{"id":"db771e0fb9b04d0b53c50fd577bebde83e396a12","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-09-27 17:01:33.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"641febde6c473024526886eaa56954b7a9d41ad4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-27 17:06:03.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/7692f9888c0d429ca70b90a45b5a57c9 : SUCCESS in 24m 28s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/765de1f33669436087aa1bbccbfbe9b3 : SUCCESS in 34m 14s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/a55f91330a0945bdb9945d8bc9130f04 : SUCCESS in 1h 06m 50s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/d7141f465bb64c09bcfcec9535aecd82 : SUCCESS in 1h 03m 21s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/38df6fbcd2b3478381ec5665bd844631 : SUCCESS in 1h 21m 15s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f1a5ed1f198d4e8485894d21646febdb : SUCCESS in 31m 57s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/cb0af058f132406c9fbf3e748a75c343 : SUCCESS in 18m 51s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ef53bdc24def45e9aa87ed925ff4d091 : SUCCESS in 56m 50s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/add90c1f695a420db046e4fe0c190981 : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/48739f0ff2fd473fb091c566af5ed083 : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c35616e4530a4ed48b972c455d8ba4a3 : SUCCESS in 5m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d8afddbe72ff46b69b9ceb0f6126d460 : SUCCESS in 4m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5754bee2b2b24081a37f9a8dbb2cef22 : SUCCESS in 6m 30s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/808e97bad336436b8d403b727ac28918 : SUCCESS in 19m 32s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e7a3ee71400c45c790e073de79875fdb : SUCCESS in 34m 35s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2b266e039e7d4513a7d67d7d3d7f79c3 : SUCCESS in 49m 35s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6791049c3f054921bf45863b51605668 : SUCCESS in 49m 59s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5cf2ef0c914a4dc699a3f6627311e484 : SUCCESS in 1h 04m 07s","accounts_in_message":[],"_revision_number":9},{"id":"159c7ce774c76101ae79a248c81a769b3ccdca3e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-27 17:24:08.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"a3eb995cfb113b57b5d9eb71f5601fac8487001e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-27 17:51:14.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/10db7764a2b641028afbb35d986136e9 : SUCCESS in 2h 01m 14s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/966c1da110c14af391d0b81947757a2f : SUCCESS in 1h 38m 17s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/636080a0672c451ba7c52de9b622d40b : SUCCESS in 2h 08m 33s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/91dc7094656441999846a552331b8b9e : SUCCESS in 6m 07s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/36cb50c296434c068bc822f9606f20de : SUCCESS in 5m 26s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/a538176b04464fbcaea6dbb47fb4c35e : SUCCESS in 35m 44s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"2a3fed012a2cd1e27952cd4de3cc63aaca2746c0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-09-28 08:23:04.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"2feb1a6685e150bb06b99ce688846162bee7a9e0","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-09-30 18:46:02.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"620704b00eddca8fafa0e78426abb4a01f1d7d29","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 18:46:41.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"ebe8388d83060b8eadc3f02ac505aae8b6e093eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 20:22:29.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/b302fb505647452d88ee4ee2d1aa5692 : SUCCESS in 57m 24s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/9b8cc3a73f95460aa940536b055717a5 : SUCCESS in 1h 09m 45s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/4f635b777a104ace9f157b61d5741f5f : SUCCESS in 1h 34m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/822d296d9b62478caf7e5cb13b004443 : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/66f9057f9a0f4f4b9997807cb2479bed : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7edae2c0fb3c48a18de691d8205b6194 : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7e989f194bb46c281677171cfffd044 : SUCCESS in 6m 06s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/90b4f17d412e4af2943806b85b283223 : SUCCESS in 31m 56s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/6bc6222ccc4b457ea958246400d9d43d : SUCCESS in 55m 40s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0bd95ea8d8a640618bb2056de2631d75 : SUCCESS in 1h 04m 00s","accounts_in_message":[],"_revision_number":9},{"id":"1f79ef33f0b303a7d1f461b46ac2a07ac0f38b30","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 20:22:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9},{"id":"36480260f1e28639e20df6440cc9001359f402f6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-30 20:23:56.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a5c8800987f4becbed4aec08e1fdc02 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"6366dec51066189211c3e48c49cb5a13b32f99a5","revisions":{"80df56d55f59be3637ec7780af30be7cc6c11805":{"kind":"REWORK","_number":1,"created":"2021-08-05 15:00:52.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/17/803617/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/803617/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/803617/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/803617/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/803617/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/803617/1"}}},"commit":{"parents":[{"commit":"ea6ceef686c018d810a2ff20ca4ba81aa77164e5","subject":"Merge \"drop leftovers of RHEL support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ea6ceef686c018d810a2ff20ca4ba81aa77164e5"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-08-05 14:59:39.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-08-05 14:59:39.000000000","tz":0},"subject":"Avoid the use of --force in groupadd","message":"Avoid the use of --force in groupadd\n\nAccording to the man page, --force may ignore --gid and this is\nnot what we want.\nWith our ID system it is almost impossible that we hit this issue\nbut better be safe than sorry. ;-)\n\nChange-Id: I359982ddd144e07f8296b384638cf0b5331e495c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/80df56d55f59be3637ec7780af30be7cc6c11805"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/80df56d55f59be3637ec7780af30be7cc6c11805"}]},"branch":"refs/heads/master"},"7f84169669b0447095c811c87f44d8d8b84838a8":{"kind":"REWORK","_number":2,"created":"2021-08-05 15:05:44.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/17/803617/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/803617/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/803617/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/803617/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/803617/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/803617/2"}}},"commit":{"parents":[{"commit":"1d3303c067af927078a001aa8a73701601453d55","subject":"Config: Set default home to /var/lib/{user}","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1d3303c067af927078a001aa8a73701601453d55"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-08-05 14:59:39.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-08-05 15:05:28.000000000","tz":120},"subject":"Avoid the use of --force in groupadd","message":"Avoid the use of --force in groupadd\n\nAccording to the man page, --force may ignore --gid and this is\nnot what we want.\nWith our ID system it is almost impossible that we hit this issue\nbut better be safe than sorry. ;-)\n\nChange-Id: I359982ddd144e07f8296b384638cf0b5331e495c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7f84169669b0447095c811c87f44d8d8b84838a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7f84169669b0447095c811c87f44d8d8b84838a8"}]},"branch":"refs/heads/master"},"36d74c61230aac2037700aafe19b6d98cc482c5e":{"kind":"NO_CHANGE","_number":3,"created":"2021-08-05 15:07:44.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/17/803617/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/803617/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/803617/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/803617/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/803617/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/803617/3"}}},"commit":{"parents":[{"commit":"33da736668108ded943691d490cb72812c8b150d","subject":"Config: Set default home to /var/lib/{user}","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/33da736668108ded943691d490cb72812c8b150d"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-08-05 14:59:39.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-08-05 15:07:33.000000000","tz":120},"subject":"Avoid the use of --force in groupadd","message":"Avoid the use of --force in groupadd\n\nAccording to the man page, --force may ignore --gid and this is\nnot what we want.\nWith our ID system it is almost impossible that we hit this issue\nbut better be safe than sorry. ;-)\n\nChange-Id: I359982ddd144e07f8296b384638cf0b5331e495c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/36d74c61230aac2037700aafe19b6d98cc482c5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/36d74c61230aac2037700aafe19b6d98cc482c5e"}]},"branch":"refs/heads/master"},"a7a933c1f17d4ffb6e723679764996e44600dee7":{"kind":"REWORK","_number":4,"created":"2021-08-06 11:16:24.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/17/803617/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/803617/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/803617/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/803617/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/803617/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/803617/4"}}},"commit":{"parents":[{"commit":"33da736668108ded943691d490cb72812c8b150d","subject":"Config: Set default home to /var/lib/{user}","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/33da736668108ded943691d490cb72812c8b150d"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-08-05 14:59:39.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-08-06 11:14:58.000000000","tz":120},"subject":"Avoid the use of --force in groupadd","message":"Avoid the use of --force in groupadd\n\nAccording to the man page, --force may ignore --gid and this is\nnot what we want.\nWith our ID system it is almost impossible that we hit this issue\nbut better be safe than sorry. ;-)\n\nChange-Id: I359982ddd144e07f8296b384638cf0b5331e495c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a7a933c1f17d4ffb6e723679764996e44600dee7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a7a933c1f17d4ffb6e723679764996e44600dee7"}]},"branch":"refs/heads/master"},"db1e6c9774b21ca9c2ba30fe969929d616b4320f":{"kind":"REWORK","_number":5,"created":"2021-08-06 12:06:03.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/17/803617/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/803617/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/803617/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/803617/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/803617/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/803617/5"}}},"commit":{"parents":[{"commit":"33da736668108ded943691d490cb72812c8b150d","subject":"Config: Set default home to /var/lib/{user}","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/33da736668108ded943691d490cb72812c8b150d"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-08-05 14:59:39.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-08-06 12:04:54.000000000","tz":120},"subject":"Avoid the use of --force in groupadd","message":"Avoid the use of --force in groupadd\n\nAccording to the man page, --force may ignore --gid and this is\nnot what we want.\nWith our ID system it is almost impossible that we hit this issue\nbut better be safe than sorry. ;-)\n\nChange-Id: I359982ddd144e07f8296b384638cf0b5331e495c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/db1e6c9774b21ca9c2ba30fe969929d616b4320f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/db1e6c9774b21ca9c2ba30fe969929d616b4320f"}]},"branch":"refs/heads/master"},"a1db2f839cde5112a79a331d1b1b1f81ad2e8158":{"kind":"REWORK","_number":6,"created":"2021-08-09 12:25:31.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/17/803617/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/803617/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/803617/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/803617/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/803617/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/803617/6"}}},"commit":{"parents":[{"commit":"33da736668108ded943691d490cb72812c8b150d","subject":"Config: Set default home to /var/lib/{user}","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/33da736668108ded943691d490cb72812c8b150d"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-08-05 14:59:39.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-08-09 12:24:31.000000000","tz":120},"subject":"Avoid the use of --force in groupadd","message":"Avoid the use of --force in groupadd\n\nAccording to the man page, --force may ignore --gid and this is\nnot what we want.\nWith our ID system it is almost impossible that we hit this issue\nbut better be safe than sorry. ;-)\n\nChange-Id: I359982ddd144e07f8296b384638cf0b5331e495c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a1db2f839cde5112a79a331d1b1b1f81ad2e8158"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a1db2f839cde5112a79a331d1b1b1f81ad2e8158"}]},"branch":"refs/heads/master"},"996a83697dd3219043d296ee50e2d2baa4373a3b":{"kind":"REWORK","_number":7,"created":"2021-08-09 13:35:53.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/17/803617/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/803617/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/803617/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/803617/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/803617/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/803617/7"}}},"commit":{"parents":[{"commit":"33da736668108ded943691d490cb72812c8b150d","subject":"Config: Set default home to /var/lib/{user}","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/33da736668108ded943691d490cb72812c8b150d"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-08-05 14:59:39.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-08-09 13:35:44.000000000","tz":120},"subject":"Avoid the use of --force in groupadd","message":"Avoid the use of --force in groupadd\n\nAccording to the man page, --force may ignore --gid and this is\nnot what we want.\nWith our ID system it is almost impossible that we hit this issue\nbut better be safe than sorry. ;-)\n\nChange-Id: I359982ddd144e07f8296b384638cf0b5331e495c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/996a83697dd3219043d296ee50e2d2baa4373a3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/996a83697dd3219043d296ee50e2d2baa4373a3b"}]},"branch":"refs/heads/master"},"219766e58fe2e5202291ddb5c2ee9c891057bdfa":{"kind":"REWORK","_number":8,"created":"2021-09-27 07:24:47.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/17/803617/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/803617/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/803617/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/803617/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/803617/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/803617/8"}}},"commit":{"parents":[{"commit":"1dc1041cb6b0fe8f11bfe852a840dea6fdebb13f","subject":"Merge \"[debian] Use upstream MariaDB\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1dc1041cb6b0fe8f11bfe852a840dea6fdebb13f"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-08-05 14:59:39.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-09-27 07:24:44.000000000","tz":120},"subject":"Avoid the use of --force in groupadd","message":"Avoid the use of --force in groupadd\n\nAccording to the man page, --force may ignore --gid and this is\nnot what we want.\nWith our ID system it is almost impossible that we hit this issue\nbut better be safe than sorry. ;-)\n\nChange-Id: I359982ddd144e07f8296b384638cf0b5331e495c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/219766e58fe2e5202291ddb5c2ee9c891057bdfa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/219766e58fe2e5202291ddb5c2ee9c891057bdfa"}]},"branch":"refs/heads/master"},"6366dec51066189211c3e48c49cb5a13b32f99a5":{"kind":"REWORK","_number":9,"created":"2021-09-27 15:39:45.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/17/803617/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/17/803617/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/803617/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/803617/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/17/803617/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/17/803617/9"}}},"commit":{"parents":[{"commit":"1dc1041cb6b0fe8f11bfe852a840dea6fdebb13f","subject":"Merge \"[debian] Use upstream MariaDB\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1dc1041cb6b0fe8f11bfe852a840dea6fdebb13f"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-08-05 14:59:39.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-09-27 15:39:42.000000000","tz":0},"subject":"Avoid the use of --force in groupadd","message":"Avoid the use of --force in groupadd\n\nAccording to the man page, --force may ignore --gid and this is\nnot what we want.\nWith our ID system it is almost impossible that we hit this issue\nbut better be safe than sorry. ;-)\n\nChange-Id: I359982ddd144e07f8296b384638cf0b5331e495c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6366dec51066189211c3e48c49cb5a13b32f99a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6366dec51066189211c3e48c49cb5a13b32f99a5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
