)]}'
{"id":"openstack%2Fkolla~804018","triplet_id":"openstack%2Fkolla~stable%2Fwallaby~If8f0730031435dda4235b7f2d2c23e5f5f767f87","project":"openstack/kolla","branch":"stable/wallaby","topic":"bug/1808805-stable/wallaby","hashtags":[],"change_id":"If8f0730031435dda4235b7f2d2c23e5f5f767f87","subject":"monasca-thresh: Allow topology check and removal in storm","status":"MERGED","created":"2021-08-10 13:04:38.000000000","updated":"2021-08-11 08:01:14.000000000","submitted":"2021-08-11 07:59:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"804018-bug/1808805-stable/wallaby","cherry_pick_of_change":792751,"cherry_pick_of_patch_set":5,"meta_rev_id":"120979f5d1cdbfd5183defe50ea5e4fa2d2137b9","_number":804018,"virtual_id_number":804018,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-11 07:59:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-08-10 15:15:17.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2021-08-11 06:45:48.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-08-10 19:42:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2021-08-11 06:45:48.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-10 13:04:38.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"},"state":"REVIEWER"},{"updated":"2021-08-10 14:18:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-11 06:45:48.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"031162e979671385e59dff40badf7e4138c65918","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-08-10 13:04:38.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"40e271800e8857f6d33dadc4ef814d81a5cd8b1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 14:18:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/4598d095f76d4813a212ecb1cadd49bf : SUCCESS in 23m 23s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/6126429aee494780afc23b48affc728d : SUCCESS in 41m 05s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/c374a1ec705f4a61969a3b82c0364a4c : SUCCESS in 1h 08m 20s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/41f6b45435d24876bcc07bef9f98de07 : SUCCESS in 1h 02m 50s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/e5f6739e35774273868dff4b2b7f3fc5 : SUCCESS in 1h 10m 04s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/632f5fc34814450fbf030648b3c69427 : SUCCESS in 31m 47s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/299e69eac7f74ebb979d49051f69fce4 : SUCCESS in 18m 00s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/be0498f6b78d4d7d9ca8a66ed94a70d5 : SUCCESS in 56m 01s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7f4e73a0632405593c32dd476860335 : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/acb18a486ee841fc94b22adc4eb224b9 : SUCCESS in 4m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/26fe66670dbe451f9b12ea1a062a5fd8 : SUCCESS in 4m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d82b8c057e58428786375cdc96b6a8b7 : SUCCESS in 4m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b894c5d7d19e45e4aed8a92c5796d674 : SUCCESS in 6m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aa7304747d4f434080a992be875cadec : SUCCESS in 4m 24s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6b7a0ed40e0d44df8ac1995c8548f535 : SUCCESS in 19m 02s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b866b182c106462db2d2d6bfcc1e0dc9 : SUCCESS in 31m 58s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7c738a71154244abbd4d69bc3ab06d49 : SUCCESS in 49m 08s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c296a0f0469c4f2da6fd9e122fea8320 : SUCCESS in 50m 11s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/82085bf94f6d46a882a18459578c76c3 : SUCCESS in 1h 05m 56s","accounts_in_message":[],"_revision_number":1},{"id":"43e6d780d1ddbc93a0c3db55aec6cd7aee97fe80","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-08-10 15:15:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a0505c14708e10414e8b23e4fb523845e79f8cdc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 15:19:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/3d5bfa44d790462ea2c12e2aae9db5db : SUCCESS in 1h 45m 13s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3d1f6336be6f4279bee23d4d36a4b6c7 : SUCCESS in 1h 14m 15s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/50aa9fe8e90e45ff8f02f6815012684b : SUCCESS in 2h 04m 12s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/42d6e24bb56445a39d9cbb0cc8b86ed3 : SUCCESS in 1h 14m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"719d8e1e49167a20e4ed0258793d7ce24d0d0f08","author":{"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"},"date":"2021-08-10 19:42:19.000000000","message":"Patch Set 1: Code-Review+1\n\nLooks great","accounts_in_message":[],"_revision_number":1},{"id":"7628975ee76a82942d7a452da38222d02d6cc550","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-08-11 06:45:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\napproving clean backport (contents approved by others on master)","accounts_in_message":[],"_revision_number":1},{"id":"7b914683d1351ab357a68720ccf32dc50b1dcd18","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 06:46:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5288aacb6bdb3263fa9256cfbcde507d165fe82b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 07:59:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/51b36e25f02f49fa95e0d3d3470c25c6 : SUCCESS in 38m 47s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/333ad55b206a431e95eee1c0a5da5481 : SUCCESS in 1h 04m 03s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/a117d4272a934b438c3a8e222bd6fd99 : SUCCESS in 1h 09m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45c0dad9e2c448cabca6b25dc25266d8 : SUCCESS in 6m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2fd9edaa4ca34dce9c4eebcde9627e89 : SUCCESS in 6m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e7bafc82662842539955c6dc035710bf : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5416804d347458fad46850615bc1693 : SUCCESS in 8m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ecd232ebc07c4ed58c315872ce27d7ed : SUCCESS in 6m 35s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4a709cc5aa4d4a77a107802e8d918969 : SUCCESS in 36m 56s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9c8d4f04f2fd48abacda604fbc639e90 : SUCCESS in 59m 01s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7db585977f2349dfaf6662c2ed09bbcc : SUCCESS in 58m 28s","accounts_in_message":[],"_revision_number":1},{"id":"9b36efa8021d108188cac30ac608dc46204a4362","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 07:59:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"120979f5d1cdbfd5183defe50ea5e4fa2d2137b9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 08:01:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfc2fbaf90fe4de188cd1ee4e3da1ad1 : SUCCESS in 1m 12s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a7d91fbcdb59405f9a16685b66ba4d8f : SUCCESS in 1m 26s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4a0d2ce116e7fc7da1361e9b8c4855075fc83270","revisions":{"4a0d2ce116e7fc7da1361e9b8c4855075fc83270":{"kind":"REWORK","_number":1,"created":"2021-08-10 13:04:38.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/18/804018/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/18/804018/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/804018/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/804018/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/804018/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/18/804018/1"}}},"commit":{"parents":[{"commit":"ac18704dd53e92b5ce5fb4312b46e06da155e6f8","subject":"Fix missing pacemaker-cli-utils in Debian hacluster images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ac18704dd53e92b5ce5fb4312b46e06da155e6f8"}]}],"author":{"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","date":"2021-05-24 03:34:46.000000000","tz":-420},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-08-10 13:04:38.000000000","tz":0},"subject":"monasca-thresh: Allow topology check and removal in storm","message":"monasca-thresh: Allow topology check and removal in storm\n\nPatch adds a script in the monasca-thresh image that can be used\nto check if a topology exists in Storm, and optionally kill it.\n\nThis is part of a bug in kolla-ansible where topologies were\nnot submitted to Storm, but run locally.  This patch includes\na topology check script enabled by KOLLA_BOOTSTRAP which will exit\nkolla_start if the topology exists, and optionally enables topology\nremoval (to allow replacement) enabled by TOPOLOGY_REPLACE.\n\nTopology names and various timeouts may be customized.  If the\nnew env variables are not set, existing behavior is unchanged.\n\nPartial-Bug: #1808805\nChange-Id: If8f0730031435dda4235b7f2d2c23e5f5f767f87\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4a0d2ce116e7fc7da1361e9b8c4855075fc83270"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4a0d2ce116e7fc7da1361e9b8c4855075fc83270"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
