)]}'
{"id":"openstack%2Fkolla~804019","triplet_id":"openstack%2Fkolla~stable%2Fvictoria~If8f0730031435dda4235b7f2d2c23e5f5f767f87","project":"openstack/kolla","branch":"stable/victoria","topic":"bug/1808805-stable/victoria","hashtags":[],"change_id":"If8f0730031435dda4235b7f2d2c23e5f5f767f87","subject":"monasca-thresh: Allow topology check and removal in storm","status":"MERGED","created":"2021-08-10 13:04:41.000000000","updated":"2021-08-11 08:06:57.000000000","submitted":"2021-08-11 08:04:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"804019-bug/1808805-stable/victoria","cherry_pick_of_change":792751,"cherry_pick_of_patch_set":5,"meta_rev_id":"1ece97b85f782458116572c7556d4c4b7b1e0eb2","_number":804019,"virtual_id_number":804019,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-11 08:04:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-08-10 15:15:07.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2021-08-11 06:46:01.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-08-10 19:44:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2021-08-11 06:46:01.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2021-08-10 19:44:17.000000000","_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-10 13:04:41.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"},"state":"REVIEWER"},{"updated":"2021-08-10 14:18:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-11 06:46:01.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"409887bf311dc1bad059df7f61a5eaac798ba3c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-08-10 13:04:41.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"df0b1ea847f4629362f091804dcf948ac3858d12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 14:18:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/29a8be42bc2b449fa5a750e5e0632ffe : SUCCESS in 28m 36s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/34a18bfaa76a474cbde0d585179703d2 : SUCCESS in 26m 41s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/94c518be58494e649b0ace3e84373ad1 : SUCCESS in 1h 00m 39s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/6301d2f6e5d949339d465a7803bdeef5 : SUCCESS in 58m 28s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/ab730dced12f4eb3aac2796fb7c46759 : SUCCESS in 1h 09m 03s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/887eb55f8184492a9d0b8f51664ca8e9 : SUCCESS in 22m 48s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/39f45d789bd14cc2affbfd3d12a9acb4 : FAILURE in 19m 41s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/9669b65282914e5881c73ee132275f1b : SUCCESS in 55m 34s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b810129bc7dd46c88f497effa275e920 : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/131e58c108c7498e81a9ade9d0f69189 : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/db2c662d384f43778e7bc301b722eaae : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4714c70d0ee94627b34bd00d4905713a : SUCCESS in 6m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5dd74701b7e04f0c95b27150972ac2be : SUCCESS in 6m 08s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9e82b6841bd14950911896cd28313238 : SUCCESS in 27m 26s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9a2b05e82d9b42f0a8701326e6fdd664 : SUCCESS in 21m 08s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5971cac9ae094d5e9a596d534435c836 : SUCCESS in 53m 45s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a498a46f8e784353ad1d6537ea872cb5 : SUCCESS in 56m 56s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e9ce0014f8d54a479447637d5e391c20 : SUCCESS in 55m 57s","accounts_in_message":[],"_revision_number":1},{"id":"ecc980e86ec4ee3f8a9834aba6fc4a0ab1a3009d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 15:08:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/f176d89710894acfa271ca9ba464f9b3 : SUCCESS in 1h 14m 13s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/55d78ace48164642811a45c1f7a485d6 : SUCCESS in 1h 06m 08s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/314578947999461ea71083cae95f8777 : SUCCESS in 1h 54m 17s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/6661ff965fe444a0b020fa51a02900ce : SUCCESS in 1h 02m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"265882cd6ebb2e514fdbb3c4a2ed5126aca681ac","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-08-10 15:15:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"95be8008caacc1f38dd31df805da9f52497db552","author":{"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"},"date":"2021-08-10 19:43:14.000000000","message":"Patch Set 1: Backport-Candidate+1","accounts_in_message":[],"_revision_number":1},{"id":"9841059a1a4f0503098302cadc316bf9776e1f21","author":{"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"},"date":"2021-08-10 19:44:17.000000000","message":"Patch Set 1: -Backport-Candidate Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b53153ab5cc5b829953ed49e6970ada8e92922e1","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-08-11 06:46:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\napproving clean backport (contents approved by others on master)","accounts_in_message":[],"_revision_number":1},{"id":"53ed8c447e5d4e663644d3e4370b1dfa51651cc2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 06:46:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"abcc10b37aeacb2215cef43ff574372f81d9211f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 08:04:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/6ccea68ce22a46bf947d452f76e86024 : SUCCESS in 35m 47s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/1e5733cce58941e9a81023c10b62a92b : SUCCESS in 1h 10m 52s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/5833a3b7bf4f4edba87966649362c6d4 : SUCCESS in 1h 16m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ea5c1a2bdab4a82b4575cf58c997e22 : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/923797fac3a24e78a5c6f0387f1990b5 : SUCCESS in 5m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9e8eb153de4c4f178c6e8c2ef51e7247 : SUCCESS in 6m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b360d7ecd1a640c999d535b5b5a9fcb9 : SUCCESS in 8m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/535eb2c0775c4e5faf23358568edb54f : SUCCESS in 5m 50s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ef747a342e5b41c5991ad8fae3719e2e : SUCCESS in 32m 04s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9f12571aaeb34d70a8dec99ac4dbedc2 : SUCCESS in 1h 01m 39s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f3318ff15ac64e86a00c0b04211444ba : SUCCESS in 1h 11m 23s","accounts_in_message":[],"_revision_number":1},{"id":"b2aa90a8ff68189249b6308b88b151eace440bb8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 08:04:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1ece97b85f782458116572c7556d4c4b7b1e0eb2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 08:06:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f1fefd6423d42ceb53855989d66dd73 : SUCCESS in 1m 06s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b84bf575cb69400ca458a12b23820315 : SUCCESS in 2m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3978b25cfcb0f871ed6dde718f3b8d76f0d3f0b5","revisions":{"3978b25cfcb0f871ed6dde718f3b8d76f0d3f0b5":{"kind":"REWORK","_number":1,"created":"2021-08-10 13:04:41.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/19/804019/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/19/804019/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/804019/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/804019/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/19/804019/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/19/804019/1"}}},"commit":{"parents":[{"commit":"99e48490044ca0c0591748153b76444a2989c18f","subject":"Merge \"Add boto3 as s3 dependency for Glance container\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/99e48490044ca0c0591748153b76444a2989c18f"}]}],"author":{"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","date":"2021-05-24 03:34:46.000000000","tz":-420},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-08-10 13:04:41.000000000","tz":0},"subject":"monasca-thresh: Allow topology check and removal in storm","message":"monasca-thresh: Allow topology check and removal in storm\n\nPatch adds a script in the monasca-thresh image that can be used\nto check if a topology exists in Storm, and optionally kill it.\n\nThis is part of a bug in kolla-ansible where topologies were\nnot submitted to Storm, but run locally.  This patch includes\na topology check script enabled by KOLLA_BOOTSTRAP which will exit\nkolla_start if the topology exists, and optionally enables topology\nremoval (to allow replacement) enabled by TOPOLOGY_REPLACE.\n\nTopology names and various timeouts may be customized.  If the\nnew env variables are not set, existing behavior is unchanged.\n\nPartial-Bug: #1808805\nChange-Id: If8f0730031435dda4235b7f2d2c23e5f5f767f87\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3978b25cfcb0f871ed6dde718f3b8d76f0d3f0b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3978b25cfcb0f871ed6dde718f3b8d76f0d3f0b5"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
