)]}'
{"id":"openstack%2Fkolla~804020","triplet_id":"openstack%2Fkolla~stable%2Fussuri~If8f0730031435dda4235b7f2d2c23e5f5f767f87","project":"openstack/kolla","branch":"stable/ussuri","topic":"bug/1808805-stable/ussuri","hashtags":[],"change_id":"If8f0730031435dda4235b7f2d2c23e5f5f767f87","subject":"monasca-thresh: Allow topology check and removal in storm","status":"MERGED","created":"2021-08-10 13:04:44.000000000","updated":"2021-08-11 08:06:16.000000000","submitted":"2021-08-11 08:04:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"804020-bug/1808805-stable/ussuri","cherry_pick_of_change":792751,"cherry_pick_of_patch_set":5,"meta_rev_id":"4a26cfb8bd20ef3e7aca72b6cae411b6953c0f60","_number":804020,"virtual_id_number":804020,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-11 08:04:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-08-10 15:14:54.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2021-08-11 06:46:08.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-08-10 19:43:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2021-08-11 06:46:08.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-10 13:04:44.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"},"state":"REVIEWER"},{"updated":"2021-08-10 14:18:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-11 06:46:08.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"32a0426a3d2f5f1deb60401a225ab92c991b4a08","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-08-10 13:04:44.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"5049cb5f3a8d9ac88fb4ea91c140f14618568e49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 14:18:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/af6fcdc3e65c48179dd4c048eadcc19f : SUCCESS in 27m 10s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/d2ab65dc9bc04527976bdc9c2f164620 : SUCCESS in 31m 28s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/1b00f31125a144a1b4618c2122ebbd57 : SUCCESS in 1h 03m 06s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/672ec8e7017f406c84def35892188454 : SUCCESS in 55m 02s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/93ea59f9dcc84d78bcbb88c27b49d57e : SUCCESS in 1h 12m 35s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/34c685eb894e4f88ab8280c287839e2a : SUCCESS in 20m 47s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/fae0f01b117e4917af74bc93c8ad2d49 : SUCCESS in 20m 39s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e740eb2c83d2484da9f2e46c77cfdceb : SUCCESS in 1h 00m 30s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8eb5b85f2098406c90a28e6525dbc063 : SUCCESS in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/935dc1d4a5314cb4aa7bd1082f12d502 : SUCCESS in 4m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c3a986d08ce24e1ea9bcc7ee3df8533a : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19f2c0071fe64bceb6883fd64b22940f : SUCCESS in 7m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/61e7ce7ed4f44b39911cbe5858df1cc2 : SUCCESS in 5m 01s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/54535fd1cccb459d979ea6ccda30e550 : SUCCESS in 22m 20s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/149aa234553b4ca6959eb6ef1f2023fc : SUCCESS in 28m 51s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9e0ac066246c416097b138979506307c : SUCCESS in 58m 43s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/bcc3a7d066b14852896770c5db41ece2 : SUCCESS in 53m 39s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f26df4b02bd54eadb4e128f8fc7edb0e : SUCCESS in 1h 01m 28s","accounts_in_message":[],"_revision_number":1},{"id":"84ac7ffd30f530526838c03ef1afdacf91b6b146","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 15:13:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/525b9d2a07a94ec79c92780cb40a9591 : SUCCESS in 1h 41m 35s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d2f0965676c9457f8416f112c340d0a4 : SUCCESS in 1h 12m 48s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1664d115fdd4471faa14e5273b0e53bc : SUCCESS in 2h 06m 30s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/35fc0ca67d2340d29ce5e7db8147736c : SUCCESS in 1h 11m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ee4f7f16fa31562141efec698464470c765e183c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-08-10 15:14:54.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"07f41dbee4fda6ef093d904f300f9f1247a13423","author":{"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"},"date":"2021-08-10 19:43:51.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"3c4b569ee650f97ec656ed5f0eade92ed3fd4f86","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-08-11 06:46:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\napproving clean backport (contents approved by others on master)","accounts_in_message":[],"_revision_number":1},{"id":"790c66ba3e0b4f8a70fb3881af8b9269e9f6698a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 06:46:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ef97926f35abbf3ab2164081f7516ac4482acdea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 08:04:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/05f586c0b9d94675b0f1a8ab21b0e7a2 : SUCCESS in 39m 05s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/fad9801d4e7942eaa9999a1ecbc6616a : SUCCESS in 1h 07m 09s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/21c78efcdbbe4ffa87df6fa1b56a10a8 : SUCCESS in 1h 06m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f68b4303648b4e71b74c7f0ce35ec765 : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/332641d77ed44e5883c570e3cd2130a7 : SUCCESS in 6m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/924056b0a5f4424b8e874b756901e36c : SUCCESS in 6m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ef96dc4f3824700baddc9e073c14272 : SUCCESS in 8m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/10e04c7c78e547d08b53bd634dbb978d : SUCCESS in 5m 15s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/89bd23e200ab4006afa6bb9a7355e92e : SUCCESS in 26m 37s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8f9dc1f3ffaf4e5d9dc4d939b932fb59 : SUCCESS in 55m 49s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d706fb587940474d8ee20e5b1f575048 : SUCCESS in 1h 11m 13s","accounts_in_message":[],"_revision_number":1},{"id":"0e731ef7e52ce3f050926b31325090774e7caeef","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 08:04:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4a26cfb8bd20ef3e7aca72b6cae411b6953c0f60","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 08:06:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0aed1c214e854a2895c7d82e37e7f71b : SUCCESS in 1m 22s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c3c28da00ee64a63a02efe850db8a955 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"36e4f077ee96b303f0acb1542f8abf1f05324755","revisions":{"36e4f077ee96b303f0acb1542f8abf1f05324755":{"kind":"REWORK","_number":1,"created":"2021-08-10 13:04:44.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/20/804020/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/20/804020/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/20/804020/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/20/804020/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/20/804020/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/20/804020/1"}}},"commit":{"parents":[{"commit":"e0fa1aa51ee3d50f4b9e577a1fe6a726e7fcebdb","subject":"masakari: Add crm_mon to masakarimonitors sudoers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e0fa1aa51ee3d50f4b9e577a1fe6a726e7fcebdb"}]}],"author":{"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","date":"2021-05-24 03:34:46.000000000","tz":-420},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-08-10 13:04:44.000000000","tz":0},"subject":"monasca-thresh: Allow topology check and removal in storm","message":"monasca-thresh: Allow topology check and removal in storm\n\nPatch adds a script in the monasca-thresh image that can be used\nto check if a topology exists in Storm, and optionally kill it.\n\nThis is part of a bug in kolla-ansible where topologies were\nnot submitted to Storm, but run locally.  This patch includes\na topology check script enabled by KOLLA_BOOTSTRAP which will exit\nkolla_start if the topology exists, and optionally enables topology\nremoval (to allow replacement) enabled by TOPOLOGY_REPLACE.\n\nTopology names and various timeouts may be customized.  If the\nnew env variables are not set, existing behavior is unchanged.\n\nPartial-Bug: #1808805\nChange-Id: If8f0730031435dda4235b7f2d2c23e5f5f767f87\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/36e4f077ee96b303f0acb1542f8abf1f05324755"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/36e4f077ee96b303f0acb1542f8abf1f05324755"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
