)]}'
{"id":"openstack%2Fkolla~804855","triplet_id":"openstack%2Fkolla~stable%2Fwallaby~I670fa84476e35e55cbed2d22e48836ed5e28495f","project":"openstack/kolla","branch":"stable/wallaby","topic":"bug/1934753-stable/wallaby","hashtags":[],"change_id":"I670fa84476e35e55cbed2d22e48836ed5e28495f","subject":"Config: allow to set group for user","status":"MERGED","created":"2021-08-17 12:01:06.000000000","updated":"2021-08-18 15:26:14.000000000","submitted":"2021-08-18 15:24:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"804855-bug/1934753-stable/wallaby","cherry_pick_of_change":802663,"cherry_pick_of_patch_set":4,"meta_rev_id":"4fd9d8c6fc487dfb9eeb338a0c207d3147e0ac21","_number":804855,"virtual_id_number":804855,"owner":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-18 15:24:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-08-18 10:41:46.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2021-08-17 12:07:07.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2021-08-18 10:41:46.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-17 12:01:06.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2021-08-17 12:07:07.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2021-08-17 13:23:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-18 10:41:46.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"3f3ff28d386a40108d31c231af4844852f99b28c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-08-17 12:01:06.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"d684725fe29cb5eac8d73520b2fe01436faac0a6","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-08-17 12:07:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ed45ca8b81d2c6c2059fb65ea3403f5f964e5093","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 13:23:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/a5deb8d8ecd64bd7bb54e93f0c835cf6 : SUCCESS in 24m 54s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/ed7369414e3c475a80b50868096a55e6 : SUCCESS in 39m 13s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/0ea65a381c074172bc0154b0a7663dc6 : SUCCESS in 1h 03m 49s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/7e70990723c049a2a5d0320ae09ceb6b : SUCCESS in 59m 49s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/8810cc21bade4e5fbd79c73582c2e3cd : SUCCESS in 1h 16m 47s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/01b5df4f3bea4c7c88c39a812c06abb0 : SUCCESS in 29m 21s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/66cdc77ca2304ab787abc36d632aa681 : SUCCESS in 17m 35s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/bd820f8a84064c4384b425c44ad4ba2c : SUCCESS in 57m 58s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba77a933f39b44b7bced5390dac72776 : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/74a083fc80774822b0ef3434916eb57f : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7b3e37d5b7364fa19bc82e20734fcfec : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/faf4507aaa164891b3ccbb91375943f2 : SUCCESS in 4m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bbc15b805ab484e81db9c3dc15aecd5 : SUCCESS in 5m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2687079d29bb4135a3c263b24b89dec9 : SUCCESS in 4m 38s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/77d145c91fd9471194a91f62fa3b7c79 : SUCCESS in 18m 44s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/896b37bb957c4b8a847a93aca27cef01 : SUCCESS in 26m 24s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/84c08f17260148018a34293bd352ed14 : SUCCESS in 52m 24s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/69b80a4277c74ae6b95d3041fca0c16f : SUCCESS in 50m 22s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b93bdcdc69074b8c90e9db1b902972c6 : SUCCESS in 56m 50s","accounts_in_message":[],"_revision_number":1},{"id":"800ae02579ac87ca38301bd5b6bf3015a80e3abf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 14:07:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/1335135b7cdf4a6bb055107d1a458190 : SUCCESS in 2h 00m 34s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f582516cee934c7eb17e95f112ecb014 : SUCCESS in 1h 43m 48s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8b57667ce5c840e5bcf1148e0724d9ce : SUCCESS in 2h 04m 22s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/be086bac5cd744ddbd8cefb7506ebcb9 : SUCCESS in 1h 37m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"eedb0cd4d8306fac9f54d663d2f5f454ef1bf290","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-08-18 10:41:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"48273cc8e0b3acf2365d1ee24ec134974283e6f0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-18 10:42:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"269b228b8dc376db3d2e301d9a4e6df6709a1b82","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-18 12:40:43.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/ecd42df319cb4656b595f1d255296393 : SUCCESS in 52m 41s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/ce69f6ddf32f4ac2b74416e033f30735 : SUCCESS in 58m 54s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/ebf9b51ff46b4ad79874efff4d3cbd3d : FAILURE in 43m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efdbdbff435241ee94fa4aee549fc559 : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/57ddb637f5094f21886d2708ab89043f : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b82402ca9a664db8814127655ce0bcc6 : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1db0728a0c884890a22ffa62532a925c : SUCCESS in 6m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5dc603b6597c4f129a17cc92695c4655 : SUCCESS in 5m 54s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/17ec5d30073447bdab191693344894a6 : SUCCESS in 1h 53m 59s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/674b891ef1024000abcda1215d0471ef : SUCCESS in 58m 30s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/134e2b4fa8024a3ab9834c70f305102d : FAILURE in 25m 16s","accounts_in_message":[],"_revision_number":1},{"id":"0291f25b9c35cb991327f3259753b0a0920ebba4","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-08-18 12:45:18.000000000","message":"Patch Set 1:\n\nrecheck\n\nunlucky pull","accounts_in_message":[],"_revision_number":1},{"id":"f33f2de9bce4d358b9a4c5b0ba38a12a9efaf835","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-18 14:01:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/abfabfd7822541368e4eac3930d05585 : SUCCESS in 40m 50s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/c394a481a84a406295405e30181e98fa : SUCCESS in 53m 27s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/d2993548054d4c65b1b1084a47bd7f8b : SUCCESS in 1h 03m 30s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/4cd4827a032d47e6a29434989212e47a : SUCCESS in 1h 09m 32s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/a75c259c20004c5dbe2b9469e8f51f0f : SUCCESS in 1h 13m 45s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/655b91159acf4bf88d0167028653b4e2 : SUCCESS in 47m 08s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/faade379ce3544449600b9a89d439b95 : SUCCESS in 36m 43s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/98c7b9c35928438bb1ef7afc6ddcdf28 : SUCCESS in 1h 06m 36s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5e3d9ec19c341f38e8479f2106e2ba5 : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d6eb2fd9c3cb4d97bc99cbc422f67650 : SUCCESS in 6m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ae46c1e2656a4f8a973ebfe92b91f9a9 : SUCCESS in 4m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/899783a2715c49f6adf769637a9fcdd2 : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6462bf20df3d414abb5480510cd4fae0 : SUCCESS in 6m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b3515ca440f74e729ac46ee155a9873c : SUCCESS in 4m 50s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a6ed2621186c4164a0c4dff4ef65a19d : SUCCESS in 39m 26s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/fd465fad91cb4e9dac20d7abf3e69bd0 : SUCCESS in 50m 06s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d1f63cb0805540eab2f24c283f6533ff : SUCCESS in 53m 08s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/91596361b5a342098cd4141dd9251eb8 : SUCCESS in 59m 48s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/35ebf74330ff44c5a95a9a0c2e2f4d70 : SUCCESS in 59m 12s","accounts_in_message":[],"_revision_number":1},{"id":"bce86e36100f9c55876498a19fac419e47a7fb82","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-18 14:01:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d38c46ce8dd104164600f0c4446687410ab8ff65","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-18 14:17:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/463b07fa2ab345c38a03a1dc8a0d6397 : SUCCESS in 59m 26s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c3781ae13a8742fbb64aaf85512d0ba4 : SUCCESS in 46m 17s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/708ef42ee4b14b5fb048c31e10511475 : SUCCESS in 1h 26m 28s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/3e5fa1d34255498f8709ecfc1adda1b0 : SUCCESS in 1h 22m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ad30962e3236abb6939cf008a27d452cd3bd5f08","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-18 15:24:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/e0f10a74f04042c6a3d58fe828d9e9dd : SUCCESS in 51m 58s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/522a741d95734fa1960b6b74e2b035c7 : SUCCESS in 1h 22m 08s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/48b9551df4624bbdb64e6806f5ef905e : SUCCESS in 1h 15m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2226131618542b9b3439fc030d03c3d : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e7d73e628aa74f04856cd8e7e8b00baf : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c6c910f9e2404781858900293e72196e : SUCCESS in 6m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81a31d7f76794e28a2109d7bf1011948 : SUCCESS in 7m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6ab86386f5c044ce974de57b4af0f561 : SUCCESS in 5m 16s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9158631b5e004e12a7eaf8d403e60809 : SUCCESS in 41m 18s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d5aa8d84d7f048189533d3da5c76c423 : SUCCESS in 52m 43s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9e721b10112d4be2a730f1d072087a36 : SUCCESS in 1h 01m 52s","accounts_in_message":[],"_revision_number":1},{"id":"89d8bf8e0849b0a4b2e9a41579a1764519239ef9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-18 15:24:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4fd9d8c6fc487dfb9eeb338a0c207d3147e0ac21","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-18 15:26:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34be654c655e4c44a417d9462961f04e : SUCCESS in 1m 23s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bedc0276459a4c99b6bb2a40e37c767d : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"69a6809b1e708111d7e27f302a44a1c2515bd5d8","revisions":{"69a6809b1e708111d7e27f302a44a1c2515bd5d8":{"kind":"REWORK","_number":1,"created":"2021-08-17 12:01:06.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/55/804855/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/804855/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/804855/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/804855/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/804855/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/804855/1"}}},"commit":{"parents":[{"commit":"d8eed6c0c94624dc8d4f4f71a8d973313462cbb5","subject":"Fix naming of CentOS Stream in reno","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d8eed6c0c94624dc8d4f4f71a8d973313462cbb5"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin@juszkiewicz.com.pl","date":"2021-07-28 07:25:45.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-08-17 12:01:06.000000000","tz":0},"subject":"Config: allow to set group for user","message":"Config: allow to set group for user\n\nAccording to recent IRC discussions is could be useful for haclient.\nPartial-Bug: #1934753\n\nChange-Id: I670fa84476e35e55cbed2d22e48836ed5e28495f\n(cherry picked from commit 65684250175117b47d41e29d1bc3f9116c0084b5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/69a6809b1e708111d7e27f302a44a1c2515bd5d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/69a6809b1e708111d7e27f302a44a1c2515bd5d8"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
