)]}'
{"id":"openstack%2Fkolla~804856","triplet_id":"openstack%2Fkolla~stable%2Fwallaby~I4a79815f4e19d82fa77341672cd26d2282800f6a","project":"openstack/kolla","branch":"stable/wallaby","topic":"bug/1934753-stable/wallaby","hashtags":[],"change_id":"I4a79815f4e19d82fa77341672cd26d2282800f6a","subject":"base/deb: bump \u0027system\u0027 groups limit beyond Kolla ones","status":"MERGED","created":"2021-08-17 12:01:40.000000000","updated":"2021-08-18 12:42:22.000000000","submitted":"2021-08-18 12:40:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"804856-bug/1934753-stable/wallaby","cherry_pick_of_change":802676,"cherry_pick_of_patch_set":9,"meta_rev_id":"040286d963bf38f4294cbb266af936c0465571e3","_number":804856,"virtual_id_number":804856,"owner":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-18 12:40:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-08-18 10:41:34.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2021-08-17 12:07:11.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2021-08-18 10:41:34.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-17 12:07:11.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2021-08-17 13:22:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-18 10:41:34.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"e326a1f47e533734a2af6d1dc22acdf5be02bc13","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-08-17 12:01:40.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"80a8b6464ee595f912b1e7223115fd089c829a30","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-08-17 12:07:11.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2eec77f1ec10fd4670c32b522f95d10f559484d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 13:22:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/57a4cefc88b6496a9c421c276bb50cba : SUCCESS in 31m 11s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/319c5bc6e1ed4008b8c4391ba1a1248c : SUCCESS in 48m 11s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/90ff92102a8646de912c6b733548206c : SUCCESS in 1h 09m 18s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/122fa6d0cb7d4b04bd76af5e87b1a090 : SUCCESS in 1h 08m 21s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/b6196862113c4bce9101781c5dabb6aa : SUCCESS in 1h 16m 39s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/d3f6786680854c2b99954697e5f9eaa1 : SUCCESS in 38m 11s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/99f1c5bf17454cac9281c634da156e3d : SUCCESS in 19m 58s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/62c8cfa16b664749b72ac04de55ad290 : SUCCESS in 54m 52s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ea4264de735471eb1285ec7caf21f1a : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/30c4f686038d488f8e40022468389fdd : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b5b05ad062d04f6c88d5e85ae3ad99a4 : SUCCESS in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b33694c1e4024d74a9c964b0eb347d85 : SUCCESS in 4m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58f9fdf398204fd0b6a6d0b8ac5ac8fc : SUCCESS in 6m 10s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7c2df005a9e54311836065b22b160c23 : SUCCESS in 20m 38s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4d81d8a470504f24b6c05dddf28119c8 : SUCCESS in 35m 17s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/872bb5edab934b08af089c241dfe6738 : SUCCESS in 55m 01s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9ddd2844b9324f85b0b93c69a66f3d56 : SUCCESS in 52m 00s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8ffaeda26c2946cfaa660c474864b50a : SUCCESS in 1h 05m 30s","accounts_in_message":[],"_revision_number":1},{"id":"b01783ec14acd98e64fc6b64fc9a6ce5369170f4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 13:59:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/97b76faa97b24820a33b7979cb52b2a6 : SUCCESS in 1h 55m 21s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f15d77bc6eec4b56931c6d75514ca3df : SUCCESS in 1h 53m 05s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6b4f6bec9a17480e8e3a0183d01d5033 : SUCCESS in 1h 37m 18s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/2ade65ab0c3c4ed4b3da752216bdb476 : SUCCESS in 1h 35m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"63421f3c9b84235ccd98247327e9dd761a828f43","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-08-18 10:41:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"164d0131c19520a1f024e96ab860ce24b08ebf88","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-18 10:41:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e6ed1545b1411479cc94f28178477ce5970e6549","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-18 12:40:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/b989e8e8936744ae934c23ee20c28052 : SUCCESS in 47m 13s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/0ebf204888494fbab1c390dcd0bc40c9 : SUCCESS in 1h 39m 05s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/371acaee62224bd9a3d31bff81399f66 : SUCCESS in 1h 54m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88a81bbc0d3a4a059327b9fadd51e871 : SUCCESS in 6m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aca382bf3e364f55b97cec73359ef172 : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/08257c6c19df4f58973663a465cd12c4 : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d71f702aa0ea4b608e6e604444673ddd : SUCCESS in 6m 27s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4e0d3d1f2e3e426f90a285957c9e0320 : SUCCESS in 39m 40s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b66240f070f1465098cad4752e173aac : SUCCESS in 1h 05m 49s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3e4360d59718443d80b53d4110941856 : SUCCESS in 1h 06m 02s","accounts_in_message":[],"_revision_number":1},{"id":"859bc224b3825f3a53f0a715e03cb5576dc75dad","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-18 12:40:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"040286d963bf38f4294cbb266af936c0465571e3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-18 12:42:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/554d260217e64d128af3ee752e4e2e57 : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b7b9d3855df90db76f0418378bdb9c69053b5eb9","revisions":{"b7b9d3855df90db76f0418378bdb9c69053b5eb9":{"kind":"REWORK","_number":1,"created":"2021-08-17 12:01:40.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/56/804856/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/804856/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/804856/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/804856/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/804856/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/804856/1"}}},"commit":{"parents":[{"commit":"d8eed6c0c94624dc8d4f4f71a8d973313462cbb5","subject":"Fix naming of CentOS Stream in reno","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d8eed6c0c94624dc8d4f4f71a8d973313462cbb5"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-07-28 09:48:17.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-08-17 12:01:40.000000000","tz":0},"subject":"base/deb: bump \u0027system\u0027 groups limit beyond Kolla ones","message":"base/deb: bump \u0027system\u0027 groups limit beyond Kolla ones\n\nPacemaker complains that group \u0027haclient\u0027 already exists\nduring installation and is not \u0027system group\u0027. So let bump\nrange for \u0027system\u0027 groups to make it happier.\n\nOur images do not have normal groups so it should not affect.\nPartial-Bug: #1934753\n\nChange-Id: I4a79815f4e19d82fa77341672cd26d2282800f6a\n(cherry picked from commit 20d8b83ecb99669caf501cd54a38eaa8ffebb0bd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b7b9d3855df90db76f0418378bdb9c69053b5eb9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b7b9d3855df90db76f0418378bdb9c69053b5eb9"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
