)]}'
{"id":"openstack%2Fkolla~805134","triplet_id":"openstack%2Fkolla~stable%2Fwallaby~I614548b43de316d86bd53cec5e6b563204f9eff0","project":"openstack/kolla","branch":"stable/wallaby","topic":"bug/1934753-stable/wallaby","hashtags":[],"change_id":"I614548b43de316d86bd53cec5e6b563204f9eff0","subject":"Add some system users to fix user uid inconsistencies","status":"MERGED","created":"2021-08-19 09:25:55.000000000","updated":"2021-08-23 12:35:08.000000000","submitted":"2021-08-23 12:33:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"805134-bug/1934753-stable/wallaby","cherry_pick_of_change":802671,"cherry_pick_of_patch_set":11,"meta_rev_id":"adc4dd6ad2af8270acf66b85df82f1feae25f133","_number":805134,"virtual_id_number":805134,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-23 12:33:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-08-23 08:32:18.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2021-08-19 09:26:00.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":1,"date":"2021-08-20 09:04:41.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2021-08-23 08:32:18.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-19 09:25:55.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2021-08-19 10:43:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-23 08:32:18.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"3aa47bc3f2ab74b16de7d45ae8961edd5a9be6f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-08-19 09:25:55.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"1f9a08da57037e60241f5b66aa376bfa28cc9c49","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-08-19 09:26:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"11e6fa3f7643ebff7afd0b90d993d51cfa65c260","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 10:43:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/edacb132ec644ad99ffe08c7bdc68b21 : SUCCESS in 26m 47s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/2777dc1ba4e44116bb988d626a751b17 : SUCCESS in 52m 07s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/349288d1f4e4452085b86237c1414ce1 : SUCCESS in 1h 02m 43s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/8ab99a2992e94dd9831db6bc974a0da6 : SUCCESS in 1h 00m 57s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/d6ab32e90532476da798657127c7257a : SUCCESS in 1h 14m 08s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/2ef64ebeeca24f01abe3ae98f43ffdee : SUCCESS in 38m 44s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/4fce9f3731404aada2665eedfee8d8da : SUCCESS in 24m 15s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/49170067312c40f8ba8fd850a2658184 : SUCCESS in 1h 01m 49s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb43b28c6c7a4d42bf41b30170623c22 : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fdbbac25d0f74a34ab755686c4f9dd63 : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/951d70175d99402ea553bd332c8e5434 : SUCCESS in 4m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/72122bca9f584e10881c730c9396dc2c : SUCCESS in 4m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4228267e0b0244229f02c28f78fc7fe3 : SUCCESS in 5m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9bcd1a525dfa4098b9ce35f0ff8e078f : SUCCESS in 4m 29s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1b10661c01fc4d4a9fbacb0738b99ea3 : SUCCESS in 25m 35s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/51ad95b91c6f4332aa18fbf561823530 : SUCCESS in 35m 45s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/545e3a0384f34f448f966b943f34e5be : SUCCESS in 54m 36s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4afe0a6d241c4d15aea966a20bdb6bd2 : SUCCESS in 52m 21s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/845f39a9658249d58667da8e3270984d : SUCCESS in 58m 09s","accounts_in_message":[],"_revision_number":1},{"id":"6a1dd0da4ae4cd7fdf52f1c674f7a22fa50d007c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 11:11:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/c7927ca2621441e19753bc2ebe59b470 : SUCCESS in 49m 38s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a5195ec3816f44d38c72aedd0da64bbf : SUCCESS in 1h 11m 48s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ce77a76bca664000adc1a8c7b80eee0b : SUCCESS in 1h 41m 58s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/47894a853bb24ff69860b92bb5eac1bf : SUCCESS in 1h 07m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9d3614a9b846268fda46787b12e8a502b602d226","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-08-20 09:04:41.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"57fd78847aee362e1ab39cf7ce9d485b9fcdbf22","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-08-23 08:32:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6379eeea7808acb8d5515069821d70b2d8763aaf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 08:32:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"93eef24b4c6ceb24188d965a2dedd513aa93453f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 09:43:43.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/d0235cece3f54fe5a1d74d41d35e0b58 : SUCCESS in 39m 23s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/d1632c1ff27e499b83273ebe49cb7ef5 : SUCCESS in 1h 00m 10s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/fbe9519b4c064c098b6f8e4338968530 : SUCCESS in 1h 07m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b6c461ca00d4736913ccb3fca2bd2cb : SUCCESS in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e3026b3de8de4bd39fc6d21ed1d0b230 : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/abe29145a7554709aa1154c23c6450a8 : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c27e9c2b27ec4a69aab6100197ab4946 : SUCCESS in 6m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/14f729c712814291ae829600c8d16c05 : SUCCESS in 4m 36s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a2bfba65acb4490b80f4493d3ed074c5 : FAILURE in 36m 05s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e5067f82efd54819aad5f06fadda8252 : SUCCESS in 47m 45s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/24e90e3ab2434c06a289d2122b54c029 : SUCCESS in 55m 51s","accounts_in_message":[],"_revision_number":1},{"id":"85ec7888d6a7d38976aecac85a71dfed08acd1e8","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-08-23 09:59:08.000000000","message":"Patch Set 1:\n\nrecheck\n\nmirror sync in progress","accounts_in_message":[],"_revision_number":1},{"id":"482636e34fe9bbd8326c9ac5f28564e7e26f1656","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 11:17:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/483b27f211b84fa0bd8a163bbd1445f2 : SUCCESS in 23m 52s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/c563bef1216b4a50a90e25fe2808a1b7 : SUCCESS in 40m 36s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/112525a8fca74aba9543b0146651811b : SUCCESS in 1h 02m 56s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/4b69d207d63c449a84d9f9acc5bd6ea3 : SUCCESS in 59m 52s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/6bf4a282ae88418c958337da621e7cff : SUCCESS in 1h 14m 07s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/8d9d42c6d9be4f2d9158cde3126c45dd : SUCCESS in 32m 52s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/add474763dab47a6a7b41b7296348ab7 : SUCCESS in 20m 34s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e968750cc7f04057b0de50ecffc2ff62 : SUCCESS in 1h 00m 19s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67d26421cf3b451c922f2826ae7042e8 : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1c45774344c14aa58c14db6ff28efdbd : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e76ccd17a5e741ccaf2e9f97628ea532 : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5165eb13ba634531b49bbe4b0337a807 : SUCCESS in 4m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f50bfb6127a64eb397307f31f93c27b6 : SUCCESS in 5m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b8550f6a9ca043eeba7eda4c69a71b2f : SUCCESS in 4m 24s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b72ba2cdb31a476b99a3f98f0fa8b6f9 : SUCCESS in 21m 40s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/076761b7879f4ceeb47daddc503f1a8a : SUCCESS in 33m 34s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5cacb06547414465a13ee96a7ff14ea3 : SUCCESS in 47m 39s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ef1d211ab4d8428597266f0d272cc950 : SUCCESS in 49m 28s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e7d3fe76f43b4e49b6770c39b6179eb9 : SUCCESS in 55m 02s","accounts_in_message":[],"_revision_number":1},{"id":"220645afab5a37d35e2ecc5635246de13b7e43fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 11:17:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"512430899f21acd34ed0b99d0d70aceed864a043","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 12:19:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/661ba95feb34416986ce061ce9448ffa : SUCCESS in 2h 04m 01s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/45b0e96438334201b38d61598c7de168 : SUCCESS in 57m 51s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f1184fae72514f85bc5077ae86b5f5d1 : SUCCESS in 2h 05m 33s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/47fc3efc881649fdad2c29908c2199c0 : SUCCESS in 1h 28m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"880195c0b5c3c3b3a8e0ad55dad319116ac2b875","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 12:33:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/02d683ce68c947658a20adf326f5ddc3 : SUCCESS in 45m 57s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/54161976ff33422eaccc43a6444598f7 : SUCCESS in 1h 12m 35s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/60a8301790be4f3db97c23491387942f : SUCCESS in 1h 14m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d4194e16ca74fc7b8010fb6220fc28a : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc860ec67ca4492ca9e3d2448fffda1d : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/313caf8770c049f985504d9346fb0919 : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06d935c24a84486db7b8bbef347d1a99 : SUCCESS in 8m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/99f9499038504c71b09672f5590eca14 : SUCCESS in 5m 04s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b3b98576e2cc45899a34626aadd77f06 : SUCCESS in 41m 43s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/aa35ae5e367243a59502ee2186c672f7 : SUCCESS in 1h 00m 09s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d1f55c9d853d4fa4b0bb435994350b86 : SUCCESS in 1h 07m 01s","accounts_in_message":[],"_revision_number":1},{"id":"7ff1a65dd1181bebd31bde34eb54fc6dcb8ef669","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 12:33:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"adc4dd6ad2af8270acf66b85df82f1feae25f133","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 12:35:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f06058a412fa485e9d2b4f5d304193a9 : SUCCESS in 1m 06s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d4bd8e1fb3684ba198f93c9ed647d917 : SUCCESS in 1m 20s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"571572a2537344186d8b14465c396183b9ccc5a7","revisions":{"571572a2537344186d8b14465c396183b9ccc5a7":{"kind":"REWORK","_number":1,"created":"2021-08-19 09:25:55.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/34/805134/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/34/805134/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/805134/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/805134/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/34/805134/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/34/805134/1"}}},"commit":{"parents":[{"commit":"da6317031df2de9dead8754de0efb203426edc98","subject":"Merge \"Config: allow to set group for user\" into stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/da6317031df2de9dead8754de0efb203426edc98"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-07-28 09:09:51.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-08-19 09:25:55.000000000","tz":0},"subject":"Add some system users to fix user uid inconsistencies","message":"Add some system users to fix user uid inconsistencies\n\nThis patch is fixing user uids inconsistency between\nbase and openstack-base images.\n\nCloses-Bug: #1934753\nChange-Id: I614548b43de316d86bd53cec5e6b563204f9eff0\n(cherry picked from commit 0799569109007beb6c24dce1ee906c7c8779a560)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/571572a2537344186d8b14465c396183b9ccc5a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/571572a2537344186d8b14465c396183b9ccc5a7"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
