)]}'
{"id":"openstack%2Fkolla~805644","triplet_id":"openstack%2Fkolla~stable%2Fussuri~I46b77978926fc2b578a68d1aaa944b2198af0685","project":"openstack/kolla","branch":"stable/ussuri","topic":"unify-curl-options","hashtags":[],"change_id":"I46b77978926fc2b578a68d1aaa944b2198af0685","subject":"Unify curl options","status":"MERGED","created":"2021-08-23 13:46:00.000000000","updated":"2021-08-30 18:27:06.000000000","submitted":"2021-08-30 18:24:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"805644-unify-curl-options","meta_rev_id":"426fbbb0a6bc2e1e45be1987717f6cca40b37a2c","_number":805644,"virtual_id_number":805644,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":0,"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-30 18:24:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33855,"name":"Andrei Yachmenev","email":"andweanew@gmail.com","username":"weanew"},{"value":0,"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"recommended":{"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},"all":[{"value":1,"date":"2021-08-25 06:18:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":1,"date":"2021-08-24 20:29:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},{"value":2,"date":"2021-08-30 16:38:42.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2021-08-24 17:56:29.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-08-25 06:07:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33855,"name":"Andrei Yachmenev","email":"andweanew@gmail.com","username":"weanew"},{"value":1,"date":"2021-08-25 07:10:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":0,"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},{"value":1,"date":"2021-08-30 16:38:42.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33855,"name":"Andrei Yachmenev","email":"andweanew@gmail.com","username":"weanew"},{"value":0,"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":0,"date":"2021-08-24 20:31:21.000000000","_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33855,"name":"Andrei Yachmenev","email":"andweanew@gmail.com","username":"weanew"},{"value":0,"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"value":0,"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33855,"name":"Andrei Yachmenev","email":"andweanew@gmail.com","username":"weanew"},{"value":0,"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"reviewers":{"REVIEWER":[{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},{"_account_id":33855,"name":"Andrei Yachmenev","email":"andweanew@gmail.com","username":"weanew"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-23 13:46:00.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-08-23 15:16:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-24 07:29:14.000000000","updated_by":{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},"reviewer":{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},"state":"REVIEWER"},{"updated":"2021-08-24 07:31:30.000000000","updated_by":{"_account_id":33855,"name":"Andrei Yachmenev","email":"andweanew@gmail.com","username":"weanew"},"reviewer":{"_account_id":33855,"name":"Andrei Yachmenev","email":"andweanew@gmail.com","username":"weanew"},"state":"REVIEWER"},{"updated":"2021-08-24 08:03:19.000000000","updated_by":{"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},"reviewer":{"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},"state":"REVIEWER"},{"updated":"2021-08-24 08:04:10.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2021-08-25 07:10:58.000000000","updated_by":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"reviewer":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"state":"REVIEWER"},{"updated":"2021-08-30 16:38:42.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"}],"messages":[{"id":"cd0ef70a8d1a480041bb23e1d5395f539d1e5aa1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-08-23 13:46:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5520a04a03c1976544d0f20e7f077f2d4c4a4f91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 15:16:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/b1d8dd56d07a453c81f3b251d439e1e6 : SUCCESS in 26m 21s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/740ecaf24a4a42b9a1cac78df55a45cb : SUCCESS in 35m 56s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/1894e90505f9450da76bfac39dbac37b : SUCCESS in 1h 11m 17s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/6c949c72da8e4188ad130ba8a5be2121 : SUCCESS in 1h 06m 36s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/4ff4d7952cfd49a78833fd473ffcb5f6 : SUCCESS in 1h 14m 18s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/6a3d74ad43c9437abe1e0938c87c71ae : SUCCESS in 27m 53s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/ac81aa7986b042d58fc9e3a34efb6b7d : SUCCESS in 23m 53s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7d565b8977ca461a9c2b7ba23c453b7b : SUCCESS in 1h 02m 19s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72553f5424814aa2ae9bfdc146abd2cd : SUCCESS in 6m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e6a60ef0d7664b18a2fe197f639f2264 : SUCCESS in 7m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/edaec4c6004f4c47a97a9b66bb20a200 : SUCCESS in 6m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d32414010ae84c0280c5334452013c15 : SUCCESS in 7m 49s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6d801975bb9242ee98d9a727131b5d4f : SUCCESS in 29m 25s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a58d175c5ed44e14b694c301b546dc25 : SUCCESS in 35m 32s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b316d54619594b9ca74f5f90ee5b709f : SUCCESS in 58m 22s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/714f7399b2074722999efd3da2b9d616 : SUCCESS in 56m 41s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9a1d965317c14a59a9f5899063515139 : SUCCESS in 1h 07m 52s","accounts_in_message":[],"_revision_number":1},{"id":"20f59e0fbf04852e135c41a4fdc93c0e1d443bcb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-23 16:03:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/dbbab4c315894b25be2b4a47e3c60570 : SUCCESS in 58m 59s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/fff207022cce41d3bc026bc47792def2 : SUCCESS in 1h 04m 22s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/10894b1d066a43528311af6c2816db51 : SUCCESS in 1h 59m 28s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/783d7b476c3e438198015a1003b3c2e1 : SUCCESS in 2h 08m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5ae9e4dfee5a8996357b3fc8a55a6c21dc14fa79","author":{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},"date":"2021-08-24 07:29:14.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"30039be036027009973701e9887800ee0c396b6f","author":{"_account_id":33855,"name":"Andrei Yachmenev","email":"andweanew@gmail.com","username":"weanew"},"date":"2021-08-24 07:31:30.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f2b77cb16590e0744c2230c911c0197b0cc76526","author":{"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},"date":"2021-08-24 08:03:19.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"dd09030d59af2b714bbd4f7ba346411ff2fcdf14","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-08-24 13:06:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3e6bd3fb72cf8707e6e42633f5a5620349475566","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-08-24 13:11:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ce71c42db3b26ac8c7d9ee2c0db7374cc488848d","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-08-24 13:21:18.000000000","message":"Patch Set 1:\n\nwill remove the orphan note about the conflicts after CI job https://zuul.openstack.org/status#799511 finished","accounts_in_message":[],"_revision_number":1},{"id":"e447d9fd9700d2f37e1434f55ac97a7a58d4cd12","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-08-24 13:22:51.000000000","message":"Patch Set 1:\n\ncorrect CI job: https://zuul.openstack.org/status#805640","accounts_in_message":[],"_revision_number":1},{"id":"3f68c4f9278508df1ca447b701046bff824db754","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-08-24 17:39:29.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"cfed7c1a0b4434bb4640e37e9b7e969cecd3d3eb","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-08-24 17:56:29.000000000","message":"Patch Set 2: Code-Review+2\n\nthanks","accounts_in_message":[],"_revision_number":2},{"id":"2009d20ee623a97c33f806719b5600bf117cae3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 18:57:48.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/b5c55043ed214f5796b30332fd62f6ac : SUCCESS in 28m 55s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/e180fead63284a4f9b71a1ce06d91352 : SUCCESS in 40m 39s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/8ef94386f8dd4a97be4aa673d7745ed1 : SUCCESS in 1h 09m 35s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/8f95a366edb546f6aa1ffb10ac56416d : SUCCESS in 57m 34s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/eff0c528731d4fea93d6fdd4730190f1 : SUCCESS in 1h 14m 02s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/3c3c4028f5914c8983ef211ca58104c8 : SUCCESS in 27m 17s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/b1745e612e7842b0a00819a40769954e : SUCCESS in 22m 15s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7db21eede1c045baafa3fbf9373cda00 : SUCCESS in 55m 38s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a3e5803005a41eda256ceca1bed95b7 : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3dafe273a81544aaa75a6a37f0c96288 : SUCCESS in 4m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f452ad5973d047d18c31f3d9c8c66b17 : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80d821f5153e4d6db76ac0fe64eb1ec6 : SUCCESS in 8m 11s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9005c746c5704cce8d89ef92c02fe910 : FAILURE in 26m 44s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/aab1f830ec4140d9ae2d1f5565e831a6 : FAILURE in 21m 16s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/889500dafb27418f9b1f4ee699a49f46 : FAILURE in 13m 34s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/04c542f12bc44c13b37d59a608c421db : FAILURE in 18m 59s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b82b4b58fee54bee805640ebda61b576 : FAILURE in 15m 17s","accounts_in_message":[],"_revision_number":2},{"id":"ae9c6a75d280ab4d8bb764dfa7d05861992d29aa","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-08-24 19:04:26.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"74c60d1cbf547b228177878a33d5fd9d9e30dab3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 19:31:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/8231cb307ec44d98bc8f895ff452a35b : SUCCESS in 1h 49m 14s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e18c990b69e04e8cacea199a47727f6b : SUCCESS in 38m 21s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e88dc17bd4724ae3a2193c7901e1332f : SUCCESS in 1h 21m 05s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/b3893501fd6c44c694aa1f4b55fe18f5 : SUCCESS in 41m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"92ba98a5cc8a9269ba589df68bb63bb999d3f009","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 20:27:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/13ffc88d73ad488fb6bae8452427b23a : SUCCESS in 26m 18s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/21c361f53f214155a5e2001723276805 : SUCCESS in 31m 57s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/27e857a089cd4d9a8c3103b3b4e1c37e : SUCCESS in 1h 03m 05s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/01eb283ab9c84177a44d6b5935d6bb0e : SUCCESS in 58m 37s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/ed99a211f4544e92b87a7a5664853c7d : SUCCESS in 1h 12m 51s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/d1203356e7b349e48f187206f975288e : SUCCESS in 30m 18s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/a5811b6b0aa84dd4b7f4a926eb75f594 : SUCCESS in 21m 31s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/097c82ab79ad45ca89f173bdf002722a : SUCCESS in 1h 14m 58s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/edb874439cee46c7965f733349e12016 : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9f41f86a5af2405788983bf290cb13d8 : SUCCESS in 5m 35s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8304ae7fb1b14a00bd788bc59ebc3cef : SUCCESS in 6m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a07e11a406bd4c23aeaee5895825d1a3 : SUCCESS in 8m 07s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c81c1d4548684fcc8cc1142255bd4c89 : SUCCESS in 20m 44s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/fa6392817500454fb1fa965adc3073c4 : SUCCESS in 29m 08s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ce1bccab8f5e44529026a7d938934d2f : SUCCESS in 59m 25s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/28c631d6fe5c4d12a974bedee3e32398 : SUCCESS in 55m 27s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7c98674bfaad4f48ac5ce18aa6235826 : SUCCESS in 1h 06m 03s","accounts_in_message":[],"_revision_number":2},{"id":"3319f75845a8813fad711f93abe0c3e97ff6c217","author":{"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},"date":"2021-08-24 20:29:35.000000000","message":"Patch Set 2: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"bf05cb3ae9c0a759dbb75dffcf42b25a538a06c1","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-08-24 20:30:06.000000000","message":"Removed Backport-Candidate+1 by Sergei Raiskii \u003cconnoreika@gmail.com\u003e\n","accounts_in_message":[],"_revision_number":2},{"id":"ec27d4964373996dbc6fe9a5fa06a16de44a7dc4","author":{"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},"date":"2021-08-24 20:31:15.000000000","message":"Patch Set 2: Backport-Candidate+1","accounts_in_message":[],"_revision_number":2},{"id":"4c03a3bfe93783148cbb2b5c289729ec34825e95","author":{"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},"date":"2021-08-24 20:31:21.000000000","message":"Patch Set 2: -Backport-Candidate","accounts_in_message":[],"_revision_number":2},{"id":"63250a37716476d6a9a7d17f51c0d4542d3f7ca9","author":{"_account_id":33855,"name":"Andrei Yachmenev","email":"andweanew@gmail.com","username":"weanew"},"date":"2021-08-25 06:07:36.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"7a1d4a38a8eb8627eb3c485be0d12ddd06115aaa","author":{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},"date":"2021-08-25 06:18:23.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"435c3130ed971401a9d8913d09848b7c57fdd2c2","author":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"date":"2021-08-25 07:10:58.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"60f11f344b0c31cf17be9a4a2072a7f7d04c9a4a","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-08-30 16:38:42.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"cf1758212ec98a9da0b16abc7c15f257ef099591","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-30 16:39:09.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5137ad1d47a1d924e17301b99f6d173b0f6748a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-30 18:24:57.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/f0bcbe5e16144ed0872da3a68b9a7ce0 : SUCCESS in 30m 55s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/647065e1b7cb4a3db584ebdb474c0c0b : SUCCESS in 59m 14s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/72b8a4bca541439c8e087389c77ca5cc : SUCCESS in 1h 06m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad2966a557cb475f8bb15f8a1f61d4df : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/10fcaf032cef4f6cb0e7823fa5da909a : SUCCESS in 4m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0c4ef9bb40e94d6c9afa12de530f3040 : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed6e546642614766ad7c4bef3ac160ed : SUCCESS in 6m 37s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/60aa39a68eb94824a66a8ace92461ca3 : SUCCESS in 26m 04s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5260e1f161c142079ad0efe5a55ff427 : SUCCESS in 57m 01s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/aef9923e6f81426696aeeeca0b1f21e6 : SUCCESS in 1h 06m 04s","accounts_in_message":[],"_revision_number":2},{"id":"780197ee71666b5ff0720fb8f09bcf7be7277614","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-30 18:24:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"426fbbb0a6bc2e1e45be1987717f6cca40b37a2c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-30 18:27:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f44cc3d58e549ab8f271079625dcd75 : SUCCESS in 1m 27s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2637b43281ffbe6258655ea21b3c27fa03e4f064","revisions":{"5085e7b927b0cb08dc4e60cb3ac585be1762001f":{"kind":"REWORK","_number":1,"created":"2021-08-23 13:46:00.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/44/805644/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/805644/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/805644/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/805644/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/805644/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/805644/1"}}},"commit":{"parents":[{"commit":"36e4f077ee96b303f0acb1542f8abf1f05324755","subject":"monasca-thresh: Allow topology check and removal in storm","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/36e4f077ee96b303f0acb1542f8abf1f05324755"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2020-12-02 17:46:53.000000000","tz":60},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-08-23 13:38:27.000000000","tz":180},"subject":"Unify curl options","message":"Unify curl options\n\nCurrently we use couple of curl options throughout Dockerfiles, this change\nadds all common options to curlrc (-sSLf) and removes usage of those in\nDockerfiles.\n\nAlso fixed conflicts added by I71ba982339d5dd4306b2166d346e3e6b1f3af60e\n\nChange-Id: I46b77978926fc2b578a68d1aaa944b2198af0685\n(cherry picked from commit 22c32a30e62d42c3631306775c0cd029eb659e9a)\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5085e7b927b0cb08dc4e60cb3ac585be1762001f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5085e7b927b0cb08dc4e60cb3ac585be1762001f"}]},"branch":"refs/heads/stable/ussuri"},"2637b43281ffbe6258655ea21b3c27fa03e4f064":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-08-24 17:39:29.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/44/805644/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/805644/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/805644/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/805644/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/805644/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/805644/2"}}},"commit":{"parents":[{"commit":"36e4f077ee96b303f0acb1542f8abf1f05324755","subject":"monasca-thresh: Allow topology check and removal in storm","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/36e4f077ee96b303f0acb1542f8abf1f05324755"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2020-12-02 17:46:53.000000000","tz":60},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2021-08-24 17:38:36.000000000","tz":180},"subject":"Unify curl options","message":"Unify curl options\n\nCurrently we use couple of curl options throughout Dockerfiles, this change\nadds all common options to curlrc (-sSLf) and removes usage of those in\nDockerfiles.\n\nAlso fixed conflicts added by I75927b6016bdb7577fbff354632bdab6407f359c\nand I0bb8b2429906e698858f6b699e4c4d8bacc2dd74\n\nChange-Id: I46b77978926fc2b578a68d1aaa944b2198af0685\n(cherry picked from commit 22c32a30e62d42c3631306775c0cd029eb659e9a)\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2637b43281ffbe6258655ea21b3c27fa03e4f064"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2637b43281ffbe6258655ea21b3c27fa03e4f064"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
