)]}'
{"id":"openstack%2Fkolla~805761","triplet_id":"openstack%2Fkolla~stable%2Fwallaby~I61623748d1149adb1b096adf2fffb5a902df3181","project":"openstack/kolla","branch":"stable/wallaby","topic":"debuntu-ironic-neutron-agent-stable/wallaby","hashtags":[],"change_id":"I61623748d1149adb1b096adf2fffb5a902df3181","subject":"ironic-neutron-agent: enable for Debuntu binary","status":"MERGED","created":"2021-08-24 08:57:57.000000000","updated":"2021-08-24 17:27:04.000000000","submitted":"2021-08-24 17:25:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"805761-debuntu-ironic-neutron-agent-stable/wallaby","cherry_pick_of_change":799611,"cherry_pick_of_patch_set":9,"meta_rev_id":"eeae5fc3f7fbaa70201936a50ada7b81e0554e67","_number":805761,"virtual_id_number":805761,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-24 17:25:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-08-24 11:23:59.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2021-08-24 12:57:46.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2021-08-24 12:57:46.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-24 10:24:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-24 12:57:46.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"8dfe1b41ef99e1f944c98a97febeebf16732c277","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-08-24 08:57:57.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* kolla/image/build.py","accounts_in_message":[],"_revision_number":1},{"id":"7d562ff4a732fde8d14506795e0d9203454b5a5a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-08-24 08:58:33.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"06aa0af685db7d11c172fad2e84e3d3559f25b5f","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-08-24 08:58:37.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"1042ff03f563a1eb4d6ac232b413b63d69c6da4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 10:24:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/9815a8589ad24b10819dfdac7962b9b6 : SUCCESS in 30m 02s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/ee80a3717228435f8d780c9c1d472f05 : SUCCESS in 41m 39s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/aa92bcf32f8c4ff49c5aa21e48a4e5df : SUCCESS in 1h 08m 00s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/b7c8ed4333dd4c7c81a76c9335c297a6 : SUCCESS in 56m 51s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/7c55e96072bf4ddc90e10c6c500522f1 : SUCCESS in 1h 13m 42s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/92857048fa8249a3ba21d26921dac17f : SUCCESS in 33m 22s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/c439bcc013194735883ba957c1ad659b : SUCCESS in 25m 22s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/1bd89e7debe345908fc62fcf3ca6c91b : SUCCESS in 57m 25s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3da68a365474fa6bf2b7723215dddef : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a83b2bd9551443c88b3a0ecfaad7007 : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9a571a9e82ab49b397759fea33985569 : SUCCESS in 4m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/56a98e29a96a4d7ab709dcb14093545f : SUCCESS in 6m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10dfa13fb92449fe8e4faa5984f21cec : SUCCESS in 6m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/53c75749959f449f983dca40fb619cab : SUCCESS in 4m 28s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f119fa989501458a9a5a4a8b52727098 : SUCCESS in 19m 48s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/11dd6ec73e9642158eabedd7b5b5e3a5 : SUCCESS in 41m 24s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2336c0ac55154a748ba63ca68d347626 : SUCCESS in 59m 14s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3b383e14387441a895b72a7e62019b2f : SUCCESS in 55m 11s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7d1d6be2ca1e4b79973284e2acc8db39 : SUCCESS in 1h 09m 19s","accounts_in_message":[],"_revision_number":2},{"id":"64afcac11961b5bb9dce317570370d5757804a73","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 10:43:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/b62c5a3d416948f7adb1a3c3798fb586 : SUCCESS in 1h 17m 50s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3b292468aeb94df282d67ab47f9c03f6 : SUCCESS in 49m 51s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6b4994c23df740fc88f768597e02fa8a : SUCCESS in 1h 32m 40s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/fc12176cc0e34d029509e4b43c3e3536 : SUCCESS in 1h 25m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2b09dcfd10b7ad6d3b3df8a038cd5453476d30b2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-08-24 11:23:59.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"051f7d26d2cbd588f81eff1e16ae13995e9cbda7","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-08-24 12:57:46.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d0a4d1463a3db755cfb26733c39debc85b5aa0ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 12:58:12.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d7ea8afc98915d6670e84689ebdcb314900c20b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 14:35:41.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/f399c73379df4233a4eeebbdb0b10b4f : SUCCESS in 48m 19s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/34d4ad74b169426882ae8c66674a1962 : FAILURE in 32m 08s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/c8e3e36e083a4347991d2559a22f976b : SUCCESS in 1h 30m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ad8129538c14907a2826e1044d3b1a4 : SUCCESS in 7m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d75acc04cfdc43ab9f101781d6e52cc6 : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/390f4afcfb4e4ecca48c19a488d7f04f : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6236815452c4290b845b2d44a124e68 : SUCCESS in 7m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fb39b8bf498e4c7d894a55f8f9f054c0 : SUCCESS in 5m 51s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/30f011d4caae45db91148b87266a1cb3 : SUCCESS in 40m 27s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/2e636761e1e4418bb04afef43b27380e : SUCCESS in 52m 49s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/58245b6b8dc4410b97a589ecbc292ea5 : SUCCESS in 57m 30s","accounts_in_message":[],"_revision_number":2},{"id":"8a01d671df08257a93a4daac83004248f0c36f9a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-08-24 14:48:27.000000000","message":"Patch Set 2:\n\nrecheck\n\nslow node?","accounts_in_message":[],"_revision_number":2},{"id":"a471975657772ba5f9f1c4b9982e8db4a44c77bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 16:06:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/b61cff37845a485eb18319dbcb086b66 : SUCCESS in 24m 51s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/80eea231edff4ce0a46f21cc04259944 : SUCCESS in 36m 42s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/b567b38111c245b68c97cc0d9307b044 : SUCCESS in 1h 01m 11s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/23f08aa0fc2543b09323c100c5a6229f : SUCCESS in 1h 03m 50s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/bc9e0db267ce48ebb2ae4807a2593169 : SUCCESS in 1h 13m 28s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/89acefba5de94505bcf8a9f2f68b2139 : SUCCESS in 31m 11s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/d09d2c902f5a43c69496eddeb6808533 : SUCCESS in 18m 30s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/cecbf84aeee34c43905cddea9cc27550 : SUCCESS in 54m 33s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f77362cc334d4650a15673efaabaeb53 : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3cd8415f365247339c71caae2aa55ff9 : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/709a3f6012b343139161a3b0c1719278 : SUCCESS in 4m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b3e53279de5d43a1b7030574cb68aa0c : SUCCESS in 4m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2747c6251d7460d96209eebbef43d0c : SUCCESS in 6m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9dcf6e8a4a2d49d0b53aa17f6a94cf2b : SUCCESS in 4m 18s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3bb39fda6b3746c68cce524882eb2e45 : SUCCESS in 22m 51s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/cb204f5df6534df9a4e85f5af288056d : SUCCESS in 47m 46s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c24a8d68971c4b99bd270420557ccb3b : SUCCESS in 56m 54s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/64ffc3256b624726a938b4b8ef388eb1 : SUCCESS in 48m 43s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8c846e48ffde40efb4c5006bf7f166f8 : SUCCESS in 1h 04m 08s","accounts_in_message":[],"_revision_number":2},{"id":"bb9a90180a7299531ac1ef5c6c7cec38ccc4a70c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 16:07:22.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9ef01056696acaca1254d368ae3bb4c08723dcf9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 16:37:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/5980c5d023cd4a4f846224c7a9a80466 : SUCCESS in 1h 11m 46s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/590284831aba40dab68bd89a7e731bc7 : FAILURE in 55m 10s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/228d3fd146364e7e978a0d16218b59fe : FAILURE in 1h 39m 41s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/df62857f63d7477c8f66f0bba3bbe271 : FAILURE in 1h 05m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"22b3c85814c198eef30a225987b65b062c523d42","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 17:25:21.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/676b38fb61ca4986b16526c183dda6cd : SUCCESS in 42m 32s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/e221335bfe8e4e718d3c05096a6bb907 : SUCCESS in 1h 01m 39s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/2f5b24563396470f9deb868e6c9b0eff : SUCCESS in 1h 13m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45a036a3f9924b4e985259b7802b9e30 : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dc706d62644744b09bb7ff1e4709eceb : SUCCESS in 6m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4c82a72a13ec48869055bc14446f0e44 : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8af7f081f364024914455c39ce80dce : SUCCESS in 6m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d0a4a1aa31c34710a9abd7cdb3e53652 : SUCCESS in 5m 13s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/347127c619544d2c8f073a36e72f1a29 : SUCCESS in 45m 19s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/45c5d34c333a4d92809e90fccb36ed58 : SUCCESS in 51m 43s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f263fc224ab94eddb27a2404510075cc : SUCCESS in 57m 12s","accounts_in_message":[],"_revision_number":2},{"id":"1897f83773819df0ae59ba24b0db64ee5d6bd927","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 17:25:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"eeae5fc3f7fbaa70201936a50ada7b81e0554e67","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 17:27:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6c4d3ee052e41268ed09baac9250f55 : SUCCESS in 1m 13s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ac0fb677c05a425ab3e1af49b7dcc198 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3567cffb2ed0b3079570c058c764fc8c9f58ec21","revisions":{"b101f2ba9aaaf98ba56bd32cbb7841a59719c938":{"kind":"REWORK","_number":1,"created":"2021-08-24 08:57:57.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/61/805761/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/61/805761/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/805761/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/805761/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/805761/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/61/805761/1"}}},"commit":{"parents":[{"commit":"571572a2537344186d8b14465c396183b9ccc5a7","subject":"Add some system users to fix user uid inconsistencies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/571572a2537344186d8b14465c396183b9ccc5a7"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-07-06 08:56:29.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-08-24 08:57:57.000000000","tz":0},"subject":"ironic-neutron-agent: enable for Debuntu binary","message":"ironic-neutron-agent: enable for Debuntu binary\n\nPackages are available for focal and bullseye.\n\nChange-Id: I61623748d1149adb1b096adf2fffb5a902df3181\n(cherry picked from commit 928c4648a4c2cf08c616d74f340c675418ba3385)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b101f2ba9aaaf98ba56bd32cbb7841a59719c938"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b101f2ba9aaaf98ba56bd32cbb7841a59719c938"}]},"branch":"refs/heads/stable/wallaby"},"3567cffb2ed0b3079570c058c764fc8c9f58ec21":{"kind":"REWORK","_number":2,"created":"2021-08-24 08:58:33.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/61/805761/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/61/805761/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/805761/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/805761/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/805761/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/61/805761/2"}}},"commit":{"parents":[{"commit":"571572a2537344186d8b14465c396183b9ccc5a7","subject":"Add some system users to fix user uid inconsistencies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/571572a2537344186d8b14465c396183b9ccc5a7"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-07-06 08:56:29.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-08-24 08:58:31.000000000","tz":0},"subject":"ironic-neutron-agent: enable for Debuntu binary","message":"ironic-neutron-agent: enable for Debuntu binary\n\nPackages are available for focal and bullseye.\n\nChange-Id: I61623748d1149adb1b096adf2fffb5a902df3181\n(cherry picked from commit 928c4648a4c2cf08c616d74f340c675418ba3385)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3567cffb2ed0b3079570c058c764fc8c9f58ec21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3567cffb2ed0b3079570c058c764fc8c9f58ec21"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
