)]}'
{"id":"openstack%2Fkolla~805762","triplet_id":"openstack%2Fkolla~stable%2Fwallaby~I01356184b8e0ef2b5341ec74a2c434b09a50766d","project":"openstack/kolla","branch":"stable/wallaby","topic":"debuntu-binary-cloudkitty-stable/wallaby","hashtags":[],"change_id":"I01356184b8e0ef2b5341ec74a2c434b09a50766d","subject":"Allow build debian/ubuntu-binary cloudkitty images","status":"MERGED","created":"2021-08-24 09:00:15.000000000","updated":"2021-08-24 17:28:16.000000000","submitted":"2021-08-24 17:25:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"805762-debuntu-binary-cloudkitty-stable/wallaby","cherry_pick_of_change":797433,"cherry_pick_of_patch_set":5,"meta_rev_id":"9e292f66ca9f7c5fe4326a61a53f420895204a0f","_number":805762,"virtual_id_number":805762,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-24 17:25:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-08-24 11:23:45.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2021-08-24 12:57:19.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2021-08-24 12:57:19.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-24 09:00:15.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2021-08-24 10:30:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-24 12:57:19.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"338479dd7d42d1656d2d9d8108ccfaac1c2c180e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-08-24 09:00:15.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* kolla/image/build.py","accounts_in_message":[],"_revision_number":1},{"id":"a88bd5784d575efeb84436de8465bee858a1b8eb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-08-24 09:00:43.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"3a6a3c3373dc694e3898dcbce0a4992a98a298fe","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-08-24 09:00:47.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"6231102573614fadbffa8df79c01eeaeaf46400e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 10:30:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/f6d4c83f2c1444918078c94f6ba61b2e : SUCCESS in 26m 46s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/f0fe98b749544e30a7c057f81a22787a : SUCCESS in 50m 41s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/aa2300772e244a019fa2964163914614 : SUCCESS in 1h 06m 24s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/cccf08aacf7a4888a12bd552fbb0787f : SUCCESS in 1h 10m 59s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/d5a302074f15453f9d242c4a4ed62f88 : SUCCESS in 1h 23m 25s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/02af152316ff4a53ba107d7cce5a2a21 : SUCCESS in 42m 26s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/0c79d43f60c54a20bf7ee80a936e24de : SUCCESS in 20m 05s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/8f1bc7e2e50a49ca8a8a91c4d31bec67 : SUCCESS in 1h 03m 07s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f77471c21654062a512aca6eb36ef1b : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5b7bd17f79544f0895eefcdec4d964ae : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3b9ff09cc98e4c75b499cc79e0ebacd8 : SUCCESS in 4m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/970e39bc393b41588b3a580e9e90a2d9 : SUCCESS in 4m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5612bc2cc4284cbe87f45facec6a2ffa : SUCCESS in 5m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dafe7ad572754bdda8889888bbcd425a : SUCCESS in 4m 24s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d94bf8cc459f44b1a7e8f3ab612593dc : SUCCESS in 22m 46s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8e92a23173164705b95559e5552b945d : SUCCESS in 33m 35s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ad9fdc3e10da41948e33c840416ace49 : SUCCESS in 57m 48s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/41a1aa6e4aa24e1e8eba279fe17caae0 : SUCCESS in 49m 30s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/579ec90360174a97bb87f33897464361 : SUCCESS in 58m 31s","accounts_in_message":[],"_revision_number":2},{"id":"08a6050f9acac2fe3082c26c4baa1d6f038a50e0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 11:13:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/b0634f75d9464a979beeac3085db82dd : SUCCESS in 2h 05m 43s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/da38da55c82946549222cba0191b7335 : SUCCESS in 1h 26m 58s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f43dedff235544ad8956e9e4f9d2ca6b : SUCCESS in 1h 25m 03s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/83fda97ce8104a84a4963692aae8bb21 : SUCCESS in 56m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bdb30e351f19549eb232736e47b4ed5c0bd27230","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-08-24 11:23:45.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"c947867702b5f3891afd876b2e74014b17de2e4c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-08-24 12:57:19.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"45b8df848c35b283829f1ff4d2137235631acb5a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 12:58:13.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ff29774b20195b8644d25d5fa76df273f3e1d395","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 14:35:42.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"85fcf42f0bf9c66089457836df633e65226e9a5b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-08-24 14:47:29.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"68620e3972fa54786f890b4cfe528a5b4e2c5e9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 16:04:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/76d8a699081d4572aa9b929e357b58b5 : SUCCESS in 29m 37s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/48cf0f7138994d5ca9e079ab6889fef2 : SUCCESS in 45m 14s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/f50b6b3d1e2a4123be42ded8dd1de4a8 : SUCCESS in 1h 08m 16s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/eee93db6d54d4a478a0429275c84b5ca : SUCCESS in 1h 06m 30s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/8f1a84c5397747cdb7d502feca973911 : SUCCESS in 1h 14m 57s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/5e487eed43f74fbda12190b4c71c14d5 : SUCCESS in 30m 13s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/d298229a929f4edfa5bef7ea8ecc4b95 : SUCCESS in 20m 45s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/beb8ea5f87f6412a98a0f4c4de130d2e : SUCCESS in 1h 00m 48s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86ddb8b3677c47fc912ddeaa2af2f0a2 : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6ef25f0d9d0e42ae9a07f03bba845d46 : SUCCESS in 5m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d54f7d2e3570487dbef99e0efcaa304e : SUCCESS in 5m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e1547c10888c43e7a176615cd9547f2f : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1acddb97bf98401eac62b908a9455a1f : SUCCESS in 7m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e0e193c8ae0b4172a503012168ce80bb : SUCCESS in 4m 31s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a4fa9b77634e40dd9410cdf2f2431a66 : SUCCESS in 21m 28s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a6eeac2ac53c4c828de06af1e9841b19 : SUCCESS in 40m 59s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/96bd0b62731c427f9f8324cbe82ff9d8 : SUCCESS in 54m 10s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/52fa344826074c5fa2e48f0dc26ce1a9 : SUCCESS in 53m 52s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/012b089f9f0e4724b5ae7f62718375e9 : SUCCESS in 1h 04m 28s","accounts_in_message":[],"_revision_number":2},{"id":"21f6f9cd728e2355991205409365f0fc022555e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 16:07:20.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9e86227a250d02113425c53a7bcb6e83412e4963","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 17:00:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/65914ebd4f2247d4b96084ffedad0d65 : SUCCESS in 1h 48m 48s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7afbf55c09df4b319374f780f26c9b63 : SUCCESS in 1h 30m 31s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a3a4090866234ff0aa3816ab086c0b02 : SUCCESS in 2h 10m 54s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/e79ee80327334e348927f2389d56076e : SUCCESS in 1h 19m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e685f3df52eb4565791cc19299dd28a1392faeec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 17:25:25.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/4f0a9a10dc6a4347bac97b41f4bc88d9 : SUCCESS in 44m 27s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/680dd29b3d4e4f86af1378cf5f6e5282 : SUCCESS in 1h 03m 28s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/674d9f653277472eba2bebcd7b6e4343 : SUCCESS in 1h 11m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6cad6b8840414e72ab219f4562749f2e : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6686a3a2f7df4e30812a099fcc8d9b92 : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dc7a79653a344f20b5789774923db54a : SUCCESS in 6m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/640c95eae6fc445a8b20cd98fe7bbccc : SUCCESS in 7m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/037e9eb6173248e0b25f8c9284de5bdc : SUCCESS in 5m 40s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/aa44e9162ff446c7ab9545711e90c399 : SUCCESS in 30m 42s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/36bbb3b42fc14c30b82df1ba76c75bbf : SUCCESS in 52m 36s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ae14097932c54f6bbb0c2a0baf491360 : SUCCESS in 1h 02m 08s","accounts_in_message":[],"_revision_number":2},{"id":"4a7790661b2ccb0519e9dd1856a96e594f200192","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 17:25:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"9e292f66ca9f7c5fe4326a61a53f420895204a0f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-24 17:28:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03401acb7d534eec8e4b4c65959a6190 : SUCCESS in 59s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7e023f48770e42c58f82f0c54d11b543 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"333f02a0725adfeb000e2fe99b2c07749c9186ad","revisions":{"51b8d40a400d438a648a461470e7e31e32a7cd54":{"kind":"REWORK","_number":1,"created":"2021-08-24 09:00:15.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/62/805762/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/62/805762/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/805762/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/805762/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/805762/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/62/805762/1"}}},"commit":{"parents":[{"commit":"3567cffb2ed0b3079570c058c764fc8c9f58ec21","subject":"ironic-neutron-agent: enable for Debuntu binary","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3567cffb2ed0b3079570c058c764fc8c9f58ec21"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-06-22 11:36:24.000000000","tz":120},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-08-24 09:00:15.000000000","tz":0},"subject":"Allow build debian/ubuntu-binary cloudkitty images","message":"Allow build debian/ubuntu-binary cloudkitty images\n\nChange-Id: I01356184b8e0ef2b5341ec74a2c434b09a50766d\n(cherry picked from commit 11b445126093cc323e90cade9435f0467737b85d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/51b8d40a400d438a648a461470e7e31e32a7cd54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/51b8d40a400d438a648a461470e7e31e32a7cd54"}]},"branch":"refs/heads/stable/wallaby"},"333f02a0725adfeb000e2fe99b2c07749c9186ad":{"kind":"REWORK","_number":2,"created":"2021-08-24 09:00:43.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/62/805762/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/62/805762/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/805762/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/805762/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/805762/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/62/805762/2"}}},"commit":{"parents":[{"commit":"3567cffb2ed0b3079570c058c764fc8c9f58ec21","subject":"ironic-neutron-agent: enable for Debuntu binary","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3567cffb2ed0b3079570c058c764fc8c9f58ec21"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-06-22 11:36:24.000000000","tz":120},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-08-24 09:00:40.000000000","tz":0},"subject":"Allow build debian/ubuntu-binary cloudkitty images","message":"Allow build debian/ubuntu-binary cloudkitty images\n\nChange-Id: I01356184b8e0ef2b5341ec74a2c434b09a50766d\n(cherry picked from commit 11b445126093cc323e90cade9435f0467737b85d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/333f02a0725adfeb000e2fe99b2c07749c9186ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/333f02a0725adfeb000e2fe99b2c07749c9186ad"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
