)]}'
{"id":"openstack%2Fkolla~805900","triplet_id":"openstack%2Fkolla~master~I5cb327578f5caf73801d1eedf252bd7ee246dee4","project":"openstack/kolla","branch":"master","topic":"bug/1934753","hashtags":[],"change_id":"I5cb327578f5caf73801d1eedf252bd7ee246dee4","subject":"Revert \"Config: Set default home to /var/lib/{user}\"","status":"MERGED","created":"2021-08-25 16:07:13.000000000","updated":"2021-08-26 06:25:58.000000000","submitted":"2021-08-25 20:34:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"revert_of":802685,"submission_id":"805900-bug/1934753","meta_rev_id":"af10c39bac0336ea66b981b861a41f2801232981","_number":805900,"virtual_id_number":805900,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-25 20:34:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-08-26 06:25:58.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-08-25 16:32:36.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2021-08-25 19:24:19.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-25 16:07:13.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-08-25 16:07:13.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2021-08-25 16:07:13.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2021-08-25 16:07:13.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-25 16:09:53.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"5113cbe577bdd21c7002376d7dc0f05ab6d82d43","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-08-25 16:07:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b52d622fb3d6d7e04d2e92594f877295546e3679","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-08-25 16:32:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f0d969cbb0f9f9eded2f443e170be08e5a986571","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 17:33:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/20e588f8588d44308401aacec08afdbb : SUCCESS in 22m 01s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/99b82d10b6df44d4bcdc0f054c6608c1 : SUCCESS in 39m 50s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/de2211ac9a0648e2a7d2e0520d8bf12d : SUCCESS in 1h 00m 32s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/bbbb9568fad74354912a4e6db7cc3ad9 : SUCCESS in 1h 16m 48s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/ba852be6b228467b8157d9f4d3406625 : SUCCESS in 1h 12m 16s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/dc896582340b47f18623201ad8b9f2b8 : SUCCESS in 25m 03s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/37f83a56b1574f6c94bace410fb7a383 : SUCCESS in 20m 21s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/80b50c61d15c46a7b953fa546b317ba9 : SUCCESS in 56m 14s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37a77cc5f69b44eba83d3003c62d2686 : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be563644d8f24ae698cc3e5cdb83f744 : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cd65e03cbf934d7f9bea3a907fd916a4 : SUCCESS in 6m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/36360a66154f47809d1e889788c37302 : SUCCESS in 5m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0372f52ff4304805a9a14e6d2aa30ce9 : SUCCESS in 8m 18s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/78ba8ce892d240c5882bfbc67835e9bb : SUCCESS in 24m 11s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3ff2527d885943569e8eb155abfd0a9a : SUCCESS in 31m 46s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0508274a3ec748c79d8b34b90402a31a : SUCCESS in 55m 29s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2ed88bb864dd4cdaaa4fac3a406b7b84 : SUCCESS in 51m 50s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2e802e4f82e9412cbd75baa73cec0ef5 : SUCCESS in 1h 03m 05s","accounts_in_message":[],"_revision_number":1},{"id":"6acc25b09112abaf79195e85d9d31a7cd2d6b12b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 17:46:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/5e49fb2dcb0c4185ad33f403ccb37f20 : SUCCESS in 1h 33m 31s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/13665dfe7c27473facc6694370f7a3da : SUCCESS in 1h 05m 51s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3a752e0971b041688aff8fb0761df915 : SUCCESS in 1h 30m 17s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cafd7559b9834bb8a394f43fac65a94b : SUCCESS in 9m 45s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/246ac2f0508946a9a24fbb5d438ad4d9 : SUCCESS in 8m 47s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/e0df88702d8e4ae7a857e510ad9f3a6a : SUCCESS in 58m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"119ca0636630851f27971e5a1806eabdca547496","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-08-25 19:24:19.000000000","message":"Patch Set 1: Workflow+1\n\nmerging early because it has shown to be fixing the bifrost job in https://review.opendev.org/c/openstack/kolla-ansible/+/806040","accounts_in_message":[],"_revision_number":1},{"id":"6ad2b390bc2297a67ee278f482f17d96a971ab76","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 19:24:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cbe3dda2988e8894cb6cbb7a797d9212b11b1b0a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 20:34:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/659a00a1b46a4307abc85f515939589b : SUCCESS in 36m 43s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/bd27d699433348e4a869175386b7c854 : SUCCESS in 1h 03m 22s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/44f085c4304b468888833a5dd2ac7545 : SUCCESS in 1h 08m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1af2b75b5ec84faaafa7b1a89a10beb3 : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a56fade7707646aa8c4034c8ef87e908 : SUCCESS in 5m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bdd8984dd89342fb9facb6290a72cc58 : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b65e33b01aeb49abb3426fb2464d5865 : SUCCESS in 8m 05s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9b1cdd085ce746188c17037c3a5ebb54 : SUCCESS in 28m 18s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/6705714d7aa244b3a0456eaeb163b225 : SUCCESS in 59m 34s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3be8077d931d4a40b9c250a33582b516 : SUCCESS in 57m 49s","accounts_in_message":[],"_revision_number":1},{"id":"ba67b2abafcdf9be365fa5a03eb745e2718e6aa4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 20:34:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9b3bc4a4031f22648274ea944a99c3d818d4b705","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 20:36:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f7ad249cd414f08999cf0ff0bed9294 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":1},{"id":"af10c39bac0336ea66b981b861a41f2801232981","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-08-26 06:25:58.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4f07c285c1bb80381f41d7f9e33ce3e8483350df","revisions":{"4f07c285c1bb80381f41d7f9e33ce3e8483350df":{"kind":"REWORK","_number":1,"created":"2021-08-25 16:07:13.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/00/805900/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/805900/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/805900/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/805900/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/805900/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/805900/1"}}},"commit":{"parents":[{"commit":"33da736668108ded943691d490cb72812c8b150d","subject":"Config: Set default home to /var/lib/{user}","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/33da736668108ded943691d490cb72812c8b150d"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-08-25 16:07:13.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-08-25 16:07:13.000000000","tz":0},"subject":"Revert \"Config: Set default home to /var/lib/{user}\"","message":"Revert \"Config: Set default home to /var/lib/{user}\"\n\nThis reverts commit 33da736668108ded943691d490cb72812c8b150d.\n\nReason for revert: That commit broke bifrost.\nIt needs to be redone and tested with bifrost.\n\nChange-Id: I5cb327578f5caf73801d1eedf252bd7ee246dee4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4f07c285c1bb80381f41d7f9e33ce3e8483350df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4f07c285c1bb80381f41d7f9e33ce3e8483350df"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
