)]}'
{"id":"openstack%2Fkolla~805990","triplet_id":"openstack%2Fkolla~master~Ic8f0554c95c1903640c98a7831b829c1f88f49ff","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"Ic8f0554c95c1903640c98a7831b829c1f88f49ff","subject":"Fix Elasticsearch Curator","status":"MERGED","created":"2021-08-25 13:04:33.000000000","updated":"2021-09-01 10:31:47.000000000","submitted":"2021-09-01 10:29:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"805990","meta_rev_id":"f4005faf73ef8edd1d0dcd44541678690b32749a","_number":805990,"virtual_id_number":805990,"owner":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-01 10:29:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-09-01 09:01:41.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2021-08-31 13:31:36.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":1,"date":"2021-08-31 11:02:40.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2021-09-01 09:01:41.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":1,"date":"2021-08-31 11:02:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-25 14:25:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-26 08:44:26.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2021-08-27 18:02:49.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"CC"},{"updated":"2021-08-30 14:27:27.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2021-08-30 14:32:16.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2021-09-01 09:01:41.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"d69717b46ac77e4a0775ed21f7c575bae1e13628","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"date":"2021-08-25 13:04:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1d2f9850ea8ca5e94086c4ea637e75788cdd4f34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 14:25:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/faf5ccce7509446fb9dc2b990591ee4f : FAILURE in 32m 38s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/95fc4a096b1541d993fca11fcc35c103 : FAILURE in 42m 23s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/72dbbedc03654a45b0f11558c14c4bd7 : SUCCESS in 1h 10m 41s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/e5cddd5ad7904df380a0ecf92c4be2b1 : SUCCESS in 1h 05m 16s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/ae56315c4e4f4f8aa54e2446a42a50e5 : SUCCESS in 1h 19m 03s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/73ef6edf45564a1f990ba823b346a258 : FAILURE in 30m 20s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/55c5033f893d459ba2cb6a5df902d4da : FAILURE in 22m 18s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/fc9b7a17b0a04a3a9d8ac6ada54132e3 : SUCCESS in 59m 12s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebe639b539a1488395cd38f996744dca : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7939dbde7937466cbc28cd28f1449d28 : SUCCESS in 6m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1e9370dc6fe64f488af04f9c434d7367 : SUCCESS in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bd91177aaf424a468dac47f3e6942c4e : SUCCESS in 8m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ab67806a9d14d02b3f9b1903e471203 : SUCCESS in 7m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7cc7cb6e926145da8c8199d02c5e5fc8 : SUCCESS in 5m 40s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b7c0598e3c1649de85136ebfa483999f : FAILURE in 21m 16s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/11a9fc9f58394dbd88c72b80f9f1e950 : FAILURE in 32m 26s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ba2eeb6766d54f61bc3fbf35da090a9f : SUCCESS in 53m 14s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b235efc56c61482a877b78ff4ac53c7f : SUCCESS in 51m 40s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e3078e2f68ee430488cb7790644794e1 : SUCCESS in 57m 31s","accounts_in_message":[],"_revision_number":1},{"id":"6b517983ec38435a76bf710406fa7d19e8d6afd1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"date":"2021-08-25 14:30:35.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c600466c8e6663ef51ecf2db32e42d1cef7d85c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 15:53:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/0ac9ee4752804ac7a66468a5f4f5ccd6 : SUCCESS in 28m 26s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/00fcdde33ce941609547b807295e8e77 : SUCCESS in 52m 49s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/92f6da119e4743c6828a7f26e714b957 : SUCCESS in 1h 03m 39s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/3bb97ea51485430caee27fc240a4c32f : SUCCESS in 1h 06m 36s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/4a957687298a483f9e96232f711ca9eb : SUCCESS in 1h 14m 30s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/d7e1a918f9ff4bb48d63be089bd35175 : SUCCESS in 31m 07s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/c462f1f1effc426c8fe315cb5bf2b2ba : SUCCESS in 22m 30s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/220b299f292049c89e20529f5f381b95 : SUCCESS in 1h 06m 45s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2af5ffc6d6f94edcb5a829e3a5cfcae3 : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2ff704a3dfde4b159685775393c8f885 : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0003d2fac10a40a79a1a75785154b288 : SUCCESS in 5m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6953ff0e17774bbcb3358c7cbbef5a98 : SUCCESS in 5m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e77a39332964e49a3cc07c427c334a7 : SUCCESS in 7m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/792e781d7cde40648292f35cfcaa2fa4 : SUCCESS in 4m 43s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/bbe0e711754348ec881c8f85b518cc69 : SUCCESS in 18m 45s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/eb8904e9aafa497fbd9cfc37025ee1b1 : SUCCESS in 28m 56s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/51ab1bc929434996aa15c5ac59a9cd8a : SUCCESS in 51m 14s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f4c3ebb3aa114c24810f22ecf5d9d24c : SUCCESS in 55m 04s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/94a95c155571499ebf59c8e091a045cf : SUCCESS in 1h 10m 45s","accounts_in_message":[],"_revision_number":2},{"id":"7b9862a207deb2548f9f006c59185d253ca7db89","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-25 16:07:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/1ad2897bc5b54dcb8ace2a3e0539dacb : SUCCESS in 1h 28m 32s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3b01b665768e4a8ea179f1553d066b41 : SUCCESS in 1h 16m 16s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e1f898602f3549b4b42a0855388c2397 : SUCCESS in 1h 31m 50s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8b99667277154527b1e43d04aa5f4ffb : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a564126fad6e49ddbd470f5aa337d80c : SUCCESS in 6m 23s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/d0888f7e01414b93ab7452401040663f : SUCCESS in 36m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f9c3273f149db55edba253102c40b71e6dfef4b9","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-08-26 08:44:26.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"038cfe1e2ccff78cb0d326df27d32bf12ebac63e","author":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"date":"2021-08-26 08:49:55.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5bd81ca2969c7d1a617f604eef88daf8f73e2a7e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-08-26 10:12:34.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"049b03c900b7f2f4393f2594c74b1f8eadd857c7","author":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"date":"2021-08-26 12:08:04.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"952ef1c0e2000847406865b23b0f28ce4a981cd6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"date":"2021-08-27 14:02:36.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b06d45556122230af223dbd9f387dd2417a136ae","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-27 14:26:45.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/28319e080bc14fb6aa99dc456614b309 : FAILURE in 9m 48s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/71ca65394923438d9d7a98f8f906216d : FAILURE in 7m 57s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3d13f2f07ec1496cad4f7da51883ba59 : FAILURE in 18m 06s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c8225c8c932047a0b9ddae7b65905836 : SUCCESS in 6m 09s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c6e25e3ff6ff4d1eb75041308679af62 : SUCCESS in 6m 32s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/2caf011e244c4d62b5c0e71d88d63477 : FAILURE in 6m 33s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"340fc1eedddacb9088dfd2565d8f55738140f99c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-27 14:30:15.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/3df9e6f07b7e44548f7c8db97fb922bd : FAILURE in 6m 29s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/9919486a0ad143b6822900a4d3e25e5b : FAILURE in 8m 24s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/7bbdccc5e9f94048975023dc29059592 : FAILURE in 18m 28s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/9c5f45964b2f4c70b4da28d8098777ea : FAILURE in 13m 38s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/d15cb6260ab34324a9b408f63eac9d9d : FAILURE in 14m 46s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/697975072a4947faad10a84906747b66 : FAILURE in 5m 34s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/1ddc04f25e1e43c9986d435037d60b9b : FAILURE in 6m 25s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a154a2f0210142e7833bf40250d92e48 : FAILURE in 19m 19s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7e5f0cfeae14b768570a4427416ed5c : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f02ae1c38ae44f4e8ee2c9a173129d67 : SUCCESS in 6m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3a1f735bd2aa439c9c9925b6d84bb0b1 : SUCCESS in 4m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1837c6fc258e4fdfad14b75344649507 : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90abcf074d824a51b4761767e5c88d66 : SUCCESS in 7m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a0625c30c28145f9a3e62d550a1510d9 : SUCCESS in 5m 59s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2f62753c21284cadac86875da3a2e8fa : FAILURE in 4m 36s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9b362104199f4919a25e288bd47c68a5 : FAILURE in 5m 20s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4c4f0588db70419688813b3ca005804b : FAILURE in 12m 24s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ccb144b1381c4ae4bbbf6bf18575fec1 : FAILURE in 12m 46s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8c43f770a48a4734bf72d4cccf5f2947 : FAILURE in 12m 06s","accounts_in_message":[],"_revision_number":3},{"id":"9c02680f3fdad8a8e9891f09ca2b41732d8df12e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"date":"2021-08-27 14:51:08.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"949b3fe7220e805880a1278247e3265c07e082af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-27 16:23:46.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/95240aa6c0884cfc9f07b75fb0f6cb5b : SUCCESS in 38m 04s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/df1598f9e5914fc4a21d5d03d1769f6e : SUCCESS in 1h 00m 35s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/951401a04dd149e1b584429e24190fb2 : SUCCESS in 1h 11m 18s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/30a5ef62024f49aba48929ab9b12e296 : SUCCESS in 1h 30m 27s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/0d1e2007f0a34c278cceecde345ff393 : SUCCESS in 1h 25m 17s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f3a3f156895648b4a945faa34f2f1e3e : SUCCESS in 30m 19s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/0a5d39f0679c46a898a6d99fa6421971 : SUCCESS in 21m 14s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f645a4f08fa641cc839c20c4b4193c6f : SUCCESS in 1h 02m 52s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28a7189187eb4da89a2f36f8ef470bef : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a8e80bb148714ec681cedf5ebb673e99 : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/97e61dea2ff9408b8816dd3c1252a2bf : SUCCESS in 6m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fb7887b9e3864b63947fd1c3bd359d3d : SUCCESS in 4m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28be8af3df0e4cc28ad65ba152fa5f8a : SUCCESS in 6m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dd3af1bef3614ae7a313420bb0578863 : SUCCESS in 4m 55s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/fce991583f334e56851853f1be4de8d9 : SUCCESS in 21m 41s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7bc0be91ead447f3a39bac8c89d4ec32 : SUCCESS in 27m 32s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5ec0ecb2d0de4532b7868c3c59bbd2fe : SUCCESS in 51m 14s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1261be0e2ac24ff6977305ec462e03c5 : SUCCESS in 52m 54s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7bb53a6974ec4b12b7e6d9035d33d4eb : SUCCESS in 59m 16s","accounts_in_message":[],"_revision_number":4},{"id":"e71ab7e8322d299bdfe010eada246894dd048b23","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-08-27 18:02:49.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2eaa676f0705fb6a6da6883acbbc71aeb7da0bfe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-27 18:24:30.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/f3770fdf73694f85a9878abc181f2304 : SUCCESS in 1h 51m 24s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ea7e03260e034a94a34d58e45cdcfccf : SUCCESS in 1h 46m 22s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/27d81751fa584a0cae45004b8a1fe681 : TIMED_OUT in 3h 31m 22s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/60f0b364548c4da4a8d84fb0132f7377 : SUCCESS in 5m 40s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2427cf1fcfc240998c79c712ffe2357c : SUCCESS in 4m 37s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/bea7237325c6401ab27809c481d77382 : SUCCESS in 41m 16s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9998c038634d4766e108db4274cd940f0ab7dcf7","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-08-30 14:27:27.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"2bff261996f25e8a3760d1326cccc0312bb4de72","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-08-30 14:32:16.000000000","message":"Patch Set 4: Backport-Candidate+1 Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0543c26cad7358b2baf37d68a4c7647d01b2a2e4","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-08-31 09:23:33.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"88e05182b505dde5883078a65bbfb0b9d041f109","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"date":"2021-08-31 11:01:08.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"5ef27716b0df3f317bd7636eccb9a736a7ff77af","author":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"date":"2021-08-31 11:01:50.000000000","message":"Patch Set 4:\n\n(2 comments)\n\nThanks, addressed the comments","accounts_in_message":[],"_revision_number":4},{"id":"5ff1aae9eb1d8783fe9019e2ed127220d2952610","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-08-31 11:02:40.000000000","message":"Patch Set 5: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"e69a7acc007158a86746393294d2e7e9dc7785b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 12:21:50.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/d86715d1067546fd9f59f3202051a398 : SUCCESS in 25m 03s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/591f478d4cec4c05bd29dbf4783cf838 : SUCCESS in 38m 29s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/b00b7631a0ef4876854e4168810c1fc5 : SUCCESS in 1h 11m 24s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/d462ffaad56c4000ba0df019ea4a6ee9 : SUCCESS in 1h 06m 40s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/f9cdb84c5e9144b2ae2cbf1c012a6ee4 : SUCCESS in 1h 15m 00s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/bf8d083cb1b044fcb8226974960571e6 : SUCCESS in 27m 51s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/f9ce533c1a484eba862260f555140d2d : SUCCESS in 19m 52s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/4d88694a90894618aa9f7dc93d536ebd : SUCCESS in 58m 22s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab690135f9c443a0bff4a4064a4a9d89 : SUCCESS in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ef69379881ce4afe956f1061d841721a : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f962d49f9c33428fbb21e3731e027375 : SUCCESS in 4m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ab10cefa88f84ceebc15e9642562bab3 : SUCCESS in 6m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0534ae00543423d91c1f1cb3a389fbc : SUCCESS in 8m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/acbe26d06e4f42508ac5c40c73aaa671 : SUCCESS in 5m 25s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/99f8cb8017904303be9046ec3127d694 : SUCCESS in 25m 18s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/20741e3fd77145fcbb64cc0e10abfc97 : SUCCESS in 35m 03s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/fcda34f7979441bda4e0e4e88dd770e8 : SUCCESS in 56m 43s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/800e1184382447448442251a21ea68dd : SUCCESS in 53m 22s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/010fc28e1d11404787483fdd88e96fa0 : SUCCESS in 1h 04m 55s","accounts_in_message":[],"_revision_number":5},{"id":"7c89f97d625dbcd09c9290c16d4b899b70fac5e2","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-08-31 13:31:36.000000000","message":"Patch Set 5: Code-Review+2\n\nthanks","accounts_in_message":[],"_revision_number":5},{"id":"9e43f13140168b8f454a1166869ddfaf6ca2d208","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-31 14:21:42.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/a130ac3aa1a0438abcff46b28c52e02e : TIMED_OUT in 3h 17m 24s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/666c39138fb5455b84613e8a0d078554 : SUCCESS in 3h 09m 37s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6ce13899d59246d0acc868fc10c66060 : TIMED_OUT in 3h 04m 17s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d9495df7a45f4f3fbc215c0ba38c0372 : SUCCESS in 6m 24s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4171c998356943848217c03a74645faa : SUCCESS in 5m 47s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/ed9f317790d048ca84c77ad7ce44bf19 : TIMED_OUT in 3h 15m 47s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"b97053ba2553211ae2fc356191fe6f7b196f3f01","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-09-01 09:01:41.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"01917dd9ff9fba8e9c406d5ef61127d21e3b7ddb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-01 09:03:46.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"5a413cfda48ded0971846ebc15e2f2a7bc4fe1a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-01 10:29:32.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/35f61083b6fa4c468fd9743534455d40 : SUCCESS in 36m 53s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/8211fc2cd18e4a3fa2214117e72d9885 : SUCCESS in 1h 01m 53s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/f068650a01114ca4b3b4afdae3c80243 : SUCCESS in 1h 23m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2298765f594414fbe3fffe5567ce8f6 : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/24e9842409334414827a884893122ebb : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/01fa4d3d8eaa476195a7a4f83231b1be : SUCCESS in 6m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20698ecebad04893884a49f5a1c9e819 : SUCCESS in 6m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3484b9253f3a4af8848d133997e953b5 : SUCCESS in 4m 59s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/7bc1b61e8b9b4304a9d4e942ff39a71d : SUCCESS in 28m 31s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8bdb0951a59b42be8ec36efeafee7726 : SUCCESS in 53m 53s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c998b4583c9b4ca996f2f944c981f727 : SUCCESS in 1h 03m 27s","accounts_in_message":[],"_revision_number":5},{"id":"95f887eb23a340be3583e77c8e1278b766012d24","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-01 10:29:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"f4005faf73ef8edd1d0dcd44541678690b32749a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-01 10:31:47.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20036baa20c5403bb30a7a412a0bba8f : SUCCESS in 1m 50s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c6b96f1aceb84b2b96a79de5750b3ed5 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"f31fbab84f215e12e5042a948a81cde6b5831678","revisions":{"b290b16789d3584add66db1e776889ef9985e235":{"kind":"REWORK","_number":1,"created":"2021-08-25 13:04:33.000000000","uploader":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"ref":"refs/changes/90/805990/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/90/805990/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/805990/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/805990/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/805990/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/90/805990/1"}}},"commit":{"parents":[{"commit":"ee84d3b8b9c3ac68a6f33a55e17acd17c0850805","subject":"Merge \"Config: Set default home to /var/lib/{user}\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ee84d3b8b9c3ac68a6f33a55e17acd17c0850805"}]}],"author":{"name":"Piotr Parczewski","email":"piotr@stackhpc.com","date":"2021-08-25 12:59:53.000000000","tz":120},"committer":{"name":"Piotr Parczewski","email":"piotr@stackhpc.com","date":"2021-08-25 13:04:26.000000000","tz":120},"subject":"Fix Elasticsearch curator","message":"Fix Elasticsearch curator\n\nDue to licensing, Kolla is stuck with the 7.10.2 version of Elasticsearch. This\ncommit pins python library version to be compatible with version of the cluster.\n\nCloses-Bug: #1941073\nChange-Id: Ic8f0554c95c1903640c98a7831b829c1f88f49ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b290b16789d3584add66db1e776889ef9985e235"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b290b16789d3584add66db1e776889ef9985e235"}]},"branch":"refs/heads/master"},"4be766a78083236a7766b0c174d370b4f08a8ba9":{"kind":"REWORK","_number":2,"created":"2021-08-25 14:30:35.000000000","uploader":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"ref":"refs/changes/90/805990/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/90/805990/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/805990/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/805990/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/805990/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/90/805990/2"}}},"commit":{"parents":[{"commit":"ee84d3b8b9c3ac68a6f33a55e17acd17c0850805","subject":"Merge \"Config: Set default home to /var/lib/{user}\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ee84d3b8b9c3ac68a6f33a55e17acd17c0850805"}]}],"author":{"name":"Piotr Parczewski","email":"piotr@stackhpc.com","date":"2021-08-25 12:59:53.000000000","tz":120},"committer":{"name":"Piotr Parczewski","email":"piotr@stackhpc.com","date":"2021-08-25 14:28:18.000000000","tz":120},"subject":"Fix Elasticsearch Curator","message":"Fix Elasticsearch Curator\n\nCurator does not seem to work with latest elasticsearch library. Pin it\nto the version from project\u0027s requirements [1].\n\n[1] https://github.com/elastic/curator/blob/v5.8.4/requirements.txt\n\nCloses-Bug: #1941073\nChange-Id: Ic8f0554c95c1903640c98a7831b829c1f88f49ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4be766a78083236a7766b0c174d370b4f08a8ba9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4be766a78083236a7766b0c174d370b4f08a8ba9"}]},"branch":"refs/heads/master"},"79250c167be6d553c5320b05ec274ae6c9897fcb":{"kind":"REWORK","_number":3,"created":"2021-08-27 14:02:36.000000000","uploader":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"ref":"refs/changes/90/805990/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/90/805990/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/805990/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/805990/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/805990/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/90/805990/3"}}},"commit":{"parents":[{"commit":"ee84d3b8b9c3ac68a6f33a55e17acd17c0850805","subject":"Merge \"Config: Set default home to /var/lib/{user}\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ee84d3b8b9c3ac68a6f33a55e17acd17c0850805"}]}],"author":{"name":"Piotr Parczewski","email":"piotr@stackhpc.com","date":"2021-08-25 12:59:53.000000000","tz":120},"committer":{"name":"Piotr Parczewski","email":"piotr@stackhpc.com","date":"2021-08-27 13:56:42.000000000","tz":120},"subject":"Fix Elasticsearch Curator","message":"Fix Elasticsearch Curator\n\nNewest elasticsearch python library required by Curator does no longer\nwork against the last OSS version of Elasticsearch (7.10.2). Pin it\nto the last known working version.\n\nCloses-Bug: #1941073\nChange-Id: Ic8f0554c95c1903640c98a7831b829c1f88f49ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/79250c167be6d553c5320b05ec274ae6c9897fcb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/79250c167be6d553c5320b05ec274ae6c9897fcb"}]},"branch":"refs/heads/master"},"041da935292472adcd1a1343d9c282e27bdc9b5d":{"kind":"REWORK","_number":4,"created":"2021-08-27 14:51:08.000000000","uploader":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"ref":"refs/changes/90/805990/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/90/805990/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/805990/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/805990/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/805990/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/90/805990/4"}}},"commit":{"parents":[{"commit":"ee84d3b8b9c3ac68a6f33a55e17acd17c0850805","subject":"Merge \"Config: Set default home to /var/lib/{user}\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ee84d3b8b9c3ac68a6f33a55e17acd17c0850805"}]}],"author":{"name":"Piotr Parczewski","email":"piotr@stackhpc.com","date":"2021-08-25 12:59:53.000000000","tz":120},"committer":{"name":"Piotr Parczewski","email":"piotr@stackhpc.com","date":"2021-08-27 14:50:56.000000000","tz":120},"subject":"Fix Elasticsearch Curator","message":"Fix Elasticsearch Curator\n\nNewest elasticsearch python library required by Curator does no longer\nwork against the last OSS version of Elasticsearch (7.10.2). Pin it\nto the last known working version.\n\nCloses-Bug: #1941073\nChange-Id: Ic8f0554c95c1903640c98a7831b829c1f88f49ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/041da935292472adcd1a1343d9c282e27bdc9b5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/041da935292472adcd1a1343d9c282e27bdc9b5d"}]},"branch":"refs/heads/master"},"f31fbab84f215e12e5042a948a81cde6b5831678":{"kind":"REWORK","_number":5,"created":"2021-08-31 11:01:08.000000000","uploader":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"ref":"refs/changes/90/805990/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/90/805990/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/805990/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/805990/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/805990/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/90/805990/5"}}},"commit":{"parents":[{"commit":"ee84d3b8b9c3ac68a6f33a55e17acd17c0850805","subject":"Merge \"Config: Set default home to /var/lib/{user}\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ee84d3b8b9c3ac68a6f33a55e17acd17c0850805"}]}],"author":{"name":"Piotr Parczewski","email":"piotr@stackhpc.com","date":"2021-08-25 12:59:53.000000000","tz":120},"committer":{"name":"Piotr Parczewski","email":"piotr@stackhpc.com","date":"2021-08-31 11:01:00.000000000","tz":120},"subject":"Fix Elasticsearch Curator","message":"Fix Elasticsearch Curator\n\nNewest elasticsearch python library required by Curator does no longer\nwork against the last OSS version of Elasticsearch (7.10.2). Pin it\nto the last known working version.\n\nCloses-Bug: #1941073\nChange-Id: Ic8f0554c95c1903640c98a7831b829c1f88f49ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f31fbab84f215e12e5042a948a81cde6b5831678"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f31fbab84f215e12e5042a948a81cde6b5831678"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
