)]}'
{"id":"openstack%2Fkolla~806511","triplet_id":"openstack%2Fkolla~stable%2Fwallaby~Idda2ff4a5fea56e4e17a9dbf05141572671b6afc","project":"openstack/kolla","branch":"stable/wallaby","topic":"bug/1934753-stable/wallaby","hashtags":[],"change_id":"Idda2ff4a5fea56e4e17a9dbf05141572671b6afc","subject":"Config: Set default home to /var/lib/{user}","status":"ABANDONED","created":"2021-08-29 11:19:54.000000000","updated":"2021-08-29 12:54:38.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":802685,"cherry_pick_of_patch_set":9,"meta_rev_id":"8b767e093051f13afa63f31aa29c42f5caab094a","_number":806511,"virtual_id_number":806511,"owner":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-08-29 12:40:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":-2,"date":"2021-08-29 12:28:42.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-29 12:28:42.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2021-08-29 12:40:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"535ceb6f0ade92b698c4890e92e84e1c460047ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-08-29 11:19:54.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"4e13f22b9f06c64e36867587e2b0fa6358f79c77","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-08-29 12:28:42.000000000","message":"Patch Set 1: Code-Review-2\n\nit has been reverted on master due to breakage\n\nhttps://review.opendev.org/c/openstack/kolla/+/805900","accounts_in_message":[],"_revision_number":1},{"id":"38049e8664882a14f6bd0c6972f40b600d95620b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-29 12:40:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/65d3acec251e4356975791c1d8efcdb7 : SUCCESS in 27m 45s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/c58b44c5424b4e1d973d971bb99c2059 : SUCCESS in 40m 19s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/8ae4e6c9f28f4904b2ca50cf3f936968 : SUCCESS in 1h 06m 37s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/d3bc6de976a74b4dafc26689e7738c58 : SUCCESS in 1h 06m 44s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/509c794d35f14708bed3a062b67099de : SUCCESS in 1h 19m 48s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/bf9e9b620b50455580cc5aeb71c65161 : SUCCESS in 32m 53s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/f584466253514ae38be639895ea63e7a : SUCCESS in 21m 44s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/431a3560d81d404498943cf8e1f4c4e0 : SUCCESS in 1h 01m 50s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71f832b8ce0b4033856a41941b83d0db : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/36ab6a343c5b44da8873a8c3684b5e14 : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d5240e7fafd94d6fad92f191658ee525 : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2d4b67a3cc95472da92b6918482f8b18 : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd7ea967e11842bbb9113cb9896bda6a : SUCCESS in 6m 45s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4968125fd2f1411fa25a89d1a68c9df9 : SUCCESS in 27m 15s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/cdd8c62e39ee4940ad0edd1f316703a4 : SUCCESS in 35m 02s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/97f590af1700442796cd565df8d055bd : SUCCESS in 55m 40s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/270f1623fbb24b06b624d5d44fbeb14f : SUCCESS in 52m 55s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c9e8ac72f56b4157b88757959ead1870 : SUCCESS in 59m 26s","accounts_in_message":[],"_revision_number":1},{"id":"98e02a473650622ecdea4b88f594bf2a1791c421","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-08-29 12:54:19.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1: Code-Review-2\n\u003e \n\u003e it has been reverted on master due to breakage\n\u003e \n\u003e https://review.opendev.org/c/openstack/kolla/+/805900\n\nAh, ok, so i\u0027m going to abandon.","accounts_in_message":[],"_revision_number":1},{"id":"8b767e093051f13afa63f31aa29c42f5caab094a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-08-29 12:54:38.000000000","message":"Abandoned\n\nbreaking bifrost","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"28630d7d10efa21fa46c9064087f5e3962f2cc01","revisions":{"28630d7d10efa21fa46c9064087f5e3962f2cc01":{"kind":"REWORK","_number":1,"created":"2021-08-29 11:19:54.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/11/806511/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/11/806511/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/806511/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/806511/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/806511/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/11/806511/1"}}},"commit":{"parents":[{"commit":"333f02a0725adfeb000e2fe99b2c07749c9186ad","subject":"Allow build debian/ubuntu-binary cloudkitty images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/333f02a0725adfeb000e2fe99b2c07749c9186ad"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-07-28 10:50:48.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-08-29 11:19:54.000000000","tz":0},"subject":"Config: Set default home to /var/lib/{user}","message":"Config: Set default home to /var/lib/{user}\n\nThis patch is setting users home dirs to /var/lib/{user}.\nThis is the usual default but it\u0027s good to be explicit.\n\nChange-Id: Idda2ff4a5fea56e4e17a9dbf05141572671b6afc\n(cherry picked from commit 33da736668108ded943691d490cb72812c8b150d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/28630d7d10efa21fa46c9064087f5e3962f2cc01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/28630d7d10efa21fa46c9064087f5e3962f2cc01"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
