)]}'
{"id":"openstack%2Fkolla~807043","triplet_id":"openstack%2Fkolla~master~Ie1e77f9c7d83538c051362a8972745f184c72019","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"Ie1e77f9c7d83538c051362a8972745f184c72019","subject":"ceph: Update CentOS packages to Pacific","status":"MERGED","created":"2021-09-02 08:20:23.000000000","updated":"2021-09-08 14:52:17.000000000","submitted":"2021-09-08 14:50:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"807043","meta_rev_id":"22bb52cfe70af88fb93cfc7130f18ddda7a0d0f0","_number":807043,"virtual_id_number":807043,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-08 14:50:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-09-08 13:30:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-09-03 12:29:58.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-09-08 13:30:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-02 08:43:20.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2021-09-02 08:46:08.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2021-09-02 09:37:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-08 13:30:40.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"9aec4ebb02ed1936a55e72444d22de7e32e4db01","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-09-02 08:20:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0cc2ac79116e11cb248c2d539992279de15442a0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-02 08:43:20.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5a0fc527648ec9deb80cccf921ea4c8f4d42810c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-09-02 08:46:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ae09a93e3adf6cbe6fd9c644d062ecd8cc3a3cd3","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-02 09:21:33.000000000","message":"Patch Set 1: -Code-Review\n\nI agree, let\u0027s get Pacific.\nFixing docs in https://review.opendev.org/c/openstack/kolla/+/807085","accounts_in_message":[],"_revision_number":1},{"id":"5e9357c3b655a55e877749e3dbf630f434b3d7f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 09:37:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/f95cfa9f6de34fe7886098f5066d3fd7 : SUCCESS in 24m 06s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/5e37089d831d480583735d91eeb27fe2 : SUCCESS in 47m 18s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/025f1a5d672c4872a640624f7c8c6b8e : SUCCESS in 1h 03m 35s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/c54c712b96484c3c8747d1fe8488f50f : SUCCESS in 1h 09m 38s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/fc553bc37723444198e801a027b7e589 : SUCCESS in 1h 11m 39s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/3366cb2a571645bcb60f32d52cc4efd9 : SUCCESS in 39m 29s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/81936a595be745b3baf859d25d2f79a2 : SUCCESS in 20m 12s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e681f680e5734d7c9f697c53c3b39ebd : SUCCESS in 59m 37s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c8152d6ab63468691e92ba3fa0d2afa : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a8bb5f5620e84297ad31c3c1ebafcff0 : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/651459efa4e54175b80f0194c3effee9 : SUCCESS in 4m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c1d6c84b17104556a620a64e02c5d1bb : SUCCESS in 4m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28db7c766bc540bfb194e9193f30d592 : SUCCESS in 6m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8356b7abb1f04569b60a7bdbd65820f0 : SUCCESS in 4m 51s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/fc1089c14a3f4d40b91a280baa3c2377 : SUCCESS in 19m 49s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/735ce1cf60f54cc695362c90a5c4dcc3 : SUCCESS in 31m 55s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f8e1afa8718e45c6938dcaa104791182 : SUCCESS in 48m 46s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/e5334f642ebe4faa8fd005d2a0a85652 : SUCCESS in 48m 38s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/14a1516e89a24dbf8f1c9cc997c8dbc5 : SUCCESS in 59m 01s","accounts_in_message":[],"_revision_number":1},{"id":"3db8befec154efcabdc6bd05f8df2a9d7e684c9c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-02 09:40:49.000000000","message":"Patch Set 1:\n\ncheck experimental\n\nfor ceph","accounts_in_message":[],"_revision_number":1},{"id":"79ec84b1f94fb8d5c439563e2c7cd037b0a1359c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 10:43:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/ed0eb739d60849a9b6cac2ab38621c22 : SUCCESS in 49m 30s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5ae649dd5ed84b8092d2c706e27a6534 : SUCCESS in 1h 44m 55s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/43910645b87445539e980492f486860a : SUCCESS in 2h 17m 59s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ffeef944e990472a86fc563000d92619 : SUCCESS in 6m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/405f1a7910cc49c799402045ce596bf2 : SUCCESS in 6m 22s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/b0832c421ce6418e83914d782abe98bc : SUCCESS in 1h 50m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d4a5e8bf3ad097f1300f60c7b36e9706e16e72b6","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-02 15:55:28.000000000","message":"Patch Set 1:\n\ncheck experimental","accounts_in_message":[],"_revision_number":1},{"id":"2a9504abbeca63afd8c3de0fe3dac808d41b4107","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-02 16:03:49.000000000","message":"Patch Set 1:\n\nyeah, trailing characters are not allowed except for rechecks","accounts_in_message":[],"_revision_number":1},{"id":"c0a598b214efdf260114167f366d1fba51f23c08","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 18:23:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-no-infra-wheels-centos8s-source https://zuul.opendev.org/t/openstack/build/262ccc2435734f45aef47d662d581870 : SUCCESS in 37m 26s\n- kolla-ansible-centos8s-source-bifrost https://zuul.opendev.org/t/openstack/build/b526641972af4e8ab18877f098d56724 : SUCCESS in 38m 30s (non-voting)\n- kolla-ansible-centos8s-source-cephadm https://zuul.opendev.org/t/openstack/build/9c0e6cee6ffa4405a6eb6bbf00fef06f : SUCCESS in 1h 22m 58s (non-voting)\n- kolla-ansible-centos8s-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/3f25564fade84ce38e19f10c516837d2 : FAILURE in 1h 35m 30s (non-voting)\n- kolla-ansible-centos8s-source-zun https://zuul.opendev.org/t/openstack/build/a75ba37f6feb423da2bee4c923706879 : SUCCESS in 1h 31m 56s (non-voting)\n- kolla-ansible-centos8s-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/eeece01b15a54ac5b60d67ebe9d24c91 : SUCCESS in 1h 32m 01s (non-voting)\n- kolla-ansible-centos8s-source-ironic https://zuul.opendev.org/t/openstack/build/4c5ab6800aa04a8bb929850428dacc6a : SUCCESS in 1h 06m 04s (non-voting)\n- kolla-ansible-centos8s-source-swift https://zuul.opendev.org/t/openstack/build/97c23718aa1f4460adad06b42f0363fc : SUCCESS in 1h 21m 40s (non-voting)\n- kolla-ansible-centos8s-source-mariadb https://zuul.opendev.org/t/openstack/build/fecb74f7a0cb4d4cb031f7939008060c : SUCCESS in 38m 01s (non-voting)\n- kolla-ansible-centos8s-source-masakari https://zuul.opendev.org/t/openstack/build/9be7ca1abb6446c5b0e8802313995d48 : SUCCESS in 1h 12m 18s (non-voting)\n- kolla-ansible-centos8s-source-ovn https://zuul.opendev.org/t/openstack/build/1f885dcb0b6744a681f96a387938c646 : SUCCESS in 1h 21m 14s (non-voting)\n- kolla-ansible-centos8s-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/8c2274e3bea04595ac2ece5e1231dd96 : SUCCESS in 36m 04s (non-voting)\n- kolla-build-no-infra-wheels-debian-source https://zuul.opendev.org/t/openstack/build/5af2a5d83c114d1f990eb90578ccb71d : SUCCESS in 32m 10s\n- kolla-build-no-infra-wheels-ubuntu-source https://zuul.opendev.org/t/openstack/build/bf94ada8dc6c41d0825634c4675b6b56 : SUCCESS in 37m 14s\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/c0629f3848d9477089fe2ed91f0a5053 : SUCCESS in 1h 21m 24s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/c6fd75ad4b4241cd8b05b6743562515b : FAILURE in 1h 10m 36s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/a4245b1c1fb940d488b674d0ae9fd443 : POST_FAILURE in 1h 28m 22s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/d76c505a0c1a443192c9d74aa0da48c2 : SUCCESS in 51m 04s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/54c1b18690d54e0b8782a697bab367bc : SUCCESS in 1h 03m 30s (non-voting)\n- kolla-ansible-ubuntu-source-mariadb https://zuul.opendev.org/t/openstack/build/ea31192477a1491b978e48c6bcfba5b7 : SUCCESS in 30m 59s (non-voting)\n- kolla-ansible-ubuntu-source-masakari https://zuul.opendev.org/t/openstack/build/b6d0733ac2f24d7d856730f7e0f3dde1 : SUCCESS in 58m 40s (non-voting)\n- kolla-ansible-ubuntu-source-linuxbridge https://zuul.opendev.org/t/openstack/build/ac2aa81894d741bd8fa786a546e3819e : SUCCESS in 49m 01s (non-voting)\n- kolla-ansible-ubuntu-source-ovn https://zuul.opendev.org/t/openstack/build/bb27376471914835a9da45d4373194a7 : SUCCESS in 1h 09m 56s (non-voting)\n- kolla-ansible-ubuntu-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/162ca964fc064bdcb02211c0ee5f9ba5 : SUCCESS in 27m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"63a3f63d58a1eee328f459fa29a5a7530f032291","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-02 18:30:00.000000000","message":"Patch Set 1:\n\nlooking good; just bump to Pacific (and rebase on docs fix)","accounts_in_message":[],"_revision_number":1},{"id":"6f625c1b5642048ea6be7b1d0327a723a5b0850b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-09-03 08:17:07.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4904fb23f0111cad1dd3210e3a18000ee34ccce2","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-03 08:20:38.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"def8160fe7f30df668f7ada4d70580956c3bce5f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-03 08:21:22.000000000","message":"Patch Set 2:\n\ncheck experimental","accounts_in_message":[],"_revision_number":2},{"id":"7a6a38b974963289f7af43c48d2a7c3fea1ad107","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-09-03 08:32:54.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f5943fc7aa5eb8b3e4b1728c2257b45f7c84dd7a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-03 08:35:49.000000000","message":"Patch Set 3:\n\ncheck experimental","accounts_in_message":[],"_revision_number":3},{"id":"fe4d0941e67519b90661b9d103092e5209a3b255","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-09-03 09:16:49.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0577ff63433184db154f92185c5a28142ade9e6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 09:46:12.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/d8748e350e0b4112866c175f6a4fc3a0 : SUCCESS in 25m 09s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/eb6458dff7b14bfdbcfb5c41dea37127 : SUCCESS in 38m 16s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/ff6c7f8ac4514a0098ac64bf1ce5a5a7 : SUCCESS in 1h 11m 21s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/6b818d0473ad4229ab9b47f1cc604256 : SUCCESS in 1h 07m 29s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/b1a1cbf096f14cf3a0eb02be3f507e02 : FAILURE in 1h 03m 13s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/206f5e611a0a48f39f2c1c4d3ec2eb3f : SUCCESS in 29m 53s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/685c047639cc4fff9ce0125b40f1e89b : SUCCESS in 25m 03s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7dcd587e151140dda92150bc6dcb9537 : SUCCESS in 56m 53s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7619f15d31d41259a7fb17a4f118730 : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f8803e2561dd423f995b2dd3241544a0 : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dbe8d92766ea450d9252481e1307e430 : SUCCESS in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/864dc11125484204bb875199fd07b865 : SUCCESS in 4m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a668343e052f45bb9ecbf908b221340a : SUCCESS in 7m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7b0a6db4377549bf81ec1e78130010f4 : SUCCESS in 4m 57s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5132e54f343f4ca7afd777429fa58745 : SUCCESS in 30m 56s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2844ed4ce5be400fbf259297b4f81e56 : SUCCESS in 39m 52s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/78b1851b052f41abac41bb691c97eb84 : SUCCESS in 1h 01m 42s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/12fa5c0fc18f40ec9be4295766c0a942 : SUCCESS in 50m 33s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ad19bc1fde8545cb8d50345485787920 : SUCCESS in 1h 05m 47s","accounts_in_message":[],"_revision_number":3},{"id":"b9e5e11d649524d027744a94e1da0f5aedd53926","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 10:49:49.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/9f48e0c90ae24b5c93e0166755078429 : SUCCESS in 58m 40s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/affe7bb734c34ed08cac89b09a886ebb : SUCCESS in 53m 47s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/05dbc416500940cb8c2da32413aec0c0 : SUCCESS in 2h 14m 47s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5ef787122ae24b86b76669ac5db06d05 : SUCCESS in 7m 25s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b0ad8494456a479d9c7cf5a8cd57893d : SUCCESS in 7m 32s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/92652e48e95842338224a40beea5da64 : SUCCESS in 1h 12m 13s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c9cec235dfa473743b4907457408d3d9e5ba16ed","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 11:12:36.000000000","message":"Patch Set 3:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-no-infra-wheels-centos8s-source https://zuul.opendev.org/t/openstack/build/ebbfbbdf0af3403b8bb66f1a3d741663 : SUCCESS in 37m 40s\n- kolla-ansible-centos8s-source-bifrost https://zuul.opendev.org/t/openstack/build/a3d53b2dc9684869ab7e83e6ae930cd0 : SUCCESS in 35m 33s (non-voting)\n- kolla-ansible-centos8s-source-cephadm https://zuul.opendev.org/t/openstack/build/9cce690bfc7f44939a64c8ae608dff9f : SUCCESS in 1h 32m 17s (non-voting)\n- kolla-ansible-centos8s-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/abf959ec5cc14207b856e255bb56a08c : FAILURE in 1h 24m 14s (non-voting)\n- kolla-ansible-centos8s-source-zun https://zuul.opendev.org/t/openstack/build/92fb4b89879c43f5a8b4029419383f89 : SUCCESS in 1h 34m 07s (non-voting)\n- kolla-ansible-centos8s-source-scenario-nfv https://zuul.opendev.org/t/openstack/build/ff937c4664e94a5298292bfa1e33ca22 : TIMED_OUT in 2h 31m 17s (non-voting)\n- kolla-ansible-centos8s-source-ironic https://zuul.opendev.org/t/openstack/build/1496e55ee4744f7c9bd33fd76e404c66 : SUCCESS in 1h 05m 52s (non-voting)\n- kolla-ansible-centos8s-source-swift https://zuul.opendev.org/t/openstack/build/5840de7b3d5143d0af8e565834967be6 : SUCCESS in 1h 28m 00s (non-voting)\n- kolla-ansible-centos8s-source-mariadb https://zuul.opendev.org/t/openstack/build/2ccd2f8fdcb14178a22f4c2b0df3d12d : SUCCESS in 35m 15s (non-voting)\n- kolla-ansible-centos8s-source-masakari https://zuul.opendev.org/t/openstack/build/a945311c61154578a158f66d7fe2fb90 : SUCCESS in 1h 09m 30s (non-voting)\n- kolla-ansible-centos8s-source-ovn https://zuul.opendev.org/t/openstack/build/0508177c74a74f47941940f687b9a8ea : FAILURE in 59m 18s (non-voting)\n- kolla-ansible-centos8s-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/6717516775f6476caccad294d2e6ffea : SUCCESS in 36m 19s (non-voting)\n- kolla-build-no-infra-wheels-debian-source https://zuul.opendev.org/t/openstack/build/a902ae0193344c289265dcfe426fa004 : SUCCESS in 34m 10s\n- kolla-build-no-infra-wheels-ubuntu-source https://zuul.opendev.org/t/openstack/build/8a3c16fb52674ad9a98a109e5699f458 : SUCCESS in 32m 17s\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/9fcc341568fe4924bd01fe55ff8477a5 : SUCCESS in 1h 07m 29s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/babfff4f0c5b40fe8a352d461dce4832 : FAILURE in 1h 11m 10s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/64dbc8ff846c4895a82fa7943561b4a0 : POST_FAILURE in 1h 28m 31s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/d02c1c10a22c4cd4825bfd23c8c5f981 : SUCCESS in 55m 39s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/366b0a409544487c9c243e950581027d : SUCCESS in 1h 16m 31s (non-voting)\n- kolla-ansible-ubuntu-source-mariadb https://zuul.opendev.org/t/openstack/build/15b07885618c4e9699043557ec416bfa : SUCCESS in 31m 57s (non-voting)\n- kolla-ansible-ubuntu-source-masakari https://zuul.opendev.org/t/openstack/build/0b10763327004665857cc20208b9d368 : SUCCESS in 51m 36s (non-voting)\n- kolla-ansible-ubuntu-source-linuxbridge https://zuul.opendev.org/t/openstack/build/83876bc1f7c548a1b65f53f0d852e94d : SUCCESS in 53m 30s (non-voting)\n- kolla-ansible-ubuntu-source-ovn https://zuul.opendev.org/t/openstack/build/33e13fa09da9409dae742f95359c237e : SUCCESS in 1h 01m 50s (non-voting)\n- kolla-ansible-ubuntu-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/519485459a1d460ba14d54fc4773a549 : SUCCESS in 28m 51s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"014b8f067539d534882736138dc7c522cb862337","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-03 12:10:47.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b85a46e93e436db3abf734035663aa990c0a6178","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-03 12:29:51.000000000","message":"Patch Set 3:\n\nrecheck\n\n    \"msg\": \"unable to find /var/lib/ansible/.my.cnf. Exception message: (2002, \\\"Can\u0027t connect to MySQL server on \u0027192.0.2.10\u0027 (115)\\\")\"\n\n\nlong time no see that error","accounts_in_message":[],"_revision_number":3},{"id":"e8e77271ef454aa14f25854101e080cb8284b274","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-03 12:29:58.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"c1bb4454478f5de846c1017f375df6440cd1d57e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 13:50:22.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/fe30853a1f8f4a0fa29f9d12f42f5ceb : SUCCESS in 27m 26s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/c0b3882550104edb95b870456ea2af9f : SUCCESS in 35m 17s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/2fa34d456b2c46fdae2558c2f8f0d63b : SUCCESS in 1h 06m 25s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/286d004ac2c940d38dbd439045316503 : SUCCESS in 1h 06m 37s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/f58cd5e9dfe84dea9e4d7b802c472b0e : SUCCESS in 1h 16m 13s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/0f31aa7e97a245bbade985391f7416ec : SUCCESS in 30m 00s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/93f910e22790449482a8c27d7eb25531 : SUCCESS in 24m 04s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a9415e10ab8f45c5871226e70cd9b5b4 : SUCCESS in 1h 06m 10s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dae8dbdfad15447690bfef2893f81d19 : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a9786680c6824892b2a645e337525330 : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5151de64b14d43e59191fe082d8f4ba3 : SUCCESS in 4m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/80c1f36894084776adcef251a44883dd : SUCCESS in 3m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb23552c1827491185e1988019d0103b : SUCCESS in 6m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fa38e26f8ef64f20bd412114516ec078 : SUCCESS in 4m 32s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/430a4dd2873746b2b982b7865c8542da : SUCCESS in 22m 28s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8c07645744564650a102147e4f5462f6 : SUCCESS in 30m 17s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c9e63384d8984a78a2f4d4d36722ab1d : SUCCESS in 57m 19s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2a35cee653dd4732922fd9d37d09e83a : SUCCESS in 1h 01m 02s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/53fb76d0c90a4974ba278381d1cf322e : SUCCESS in 1h 00m 10s","accounts_in_message":[],"_revision_number":3},{"id":"33f611037a72a97534bb504a2f6d6f1c542d1f66","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 13:52:37.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/3622287f10be42d6a96dcdacf70d7f99 : SUCCESS in 58m 23s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/65fd32054adf46e687167b4ccbf56c93 : SUCCESS in 1h 06m 30s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f9394c20c9c44d31b4c0e17ce1f51f77 : SUCCESS in 1h 16m 38s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ce574529826a4912b96e303e388e6311 : SUCCESS in 5m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e8e1e0327d7844cf891948cc8ec6ae3d : SUCCESS in 5m 15s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/1b596b449dbb4da78efee928d7d0debd : SUCCESS in 32m 58s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"99b7fc5849883fe7413ae61d03328517f7c40d8c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-09-08 13:30:40.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"e9671250e39e7a3a35094efcbc57207a10c4ead7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-08 13:31:10.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5c89c24753fdf135acf1e04be25b016994c70475","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-08 14:50:31.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/82c221b9703444f3b5b91078018005f3 : SUCCESS in 41m 12s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/7eb3ba7fd3084b6f9897874961afb5e8 : SUCCESS in 1h 01m 31s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/2050cff4c18c4f6f94f82dbda48c1037 : SUCCESS in 1h 14m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9d816e71dff418aaf1b6399d107d24c : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2f96bd7f281e4d95b647fc0f95504171 : SUCCESS in 5m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/83163448415440a09ea1285cedda46ab : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/752a6e7214c641dcaf4d2bd78b5f13e8 : SUCCESS in 7m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ee267dc83b404ef0acdc4e3e65659ba3 : SUCCESS in 5m 29s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3ecc28ebed154c35b99c06938b680d7e : SUCCESS in 33m 29s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d7863069c5ca4c2a9a3abc60cae1b282 : SUCCESS in 56m 53s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6df29d255aa241c4816014f891230a06 : SUCCESS in 1h 02m 36s","accounts_in_message":[],"_revision_number":3},{"id":"8cf5aabb6051b29b24f87c838fab7627da30b988","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-08 14:50:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"22bb52cfe70af88fb93cfc7130f18ddda7a0d0f0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-08 14:52:17.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a9da9cd308b49f4b0acd97d88f0410e : SUCCESS in 1m 09s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a438b4c17a047f28806167cc32b5f5d : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c6cba08314816346a9b759bb5da0a2c4ae1a41a6","revisions":{"c60c8b8500d66d7aa9886e8e48b6353d124750af":{"kind":"REWORK","_number":1,"created":"2021-09-02 08:20:23.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/43/807043/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/43/807043/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/43/807043/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/43/807043/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/43/807043/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/43/807043/1"}}},"commit":{"parents":[{"commit":"630476445cbc1b3b8a4987f1071b3b69e492166d","subject":"Merge \"Fix Elasticsearch Curator\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/630476445cbc1b3b8a4987f1071b3b69e492166d"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-09-02 08:19:05.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-09-02 08:19:05.000000000","tz":60},"subject":"ceph: Update CentOS packages to Octopus","message":"ceph: Update CentOS packages to Octopus\n\nUpdates the CentOS images to use centos-release-ceph-octopus. This is\nin line with Ubuntu which gets Octopus from UCA. Debian remains on\nNautilus.\n\nChange-Id: Ie1e77f9c7d83538c051362a8972745f184c72019\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c60c8b8500d66d7aa9886e8e48b6353d124750af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c60c8b8500d66d7aa9886e8e48b6353d124750af"}]},"branch":"refs/heads/master"},"933965f1098024212fcd059e352fe8c515ee1291":{"kind":"REWORK","_number":2,"created":"2021-09-03 08:17:07.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/43/807043/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/43/807043/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/43/807043/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/43/807043/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/43/807043/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/43/807043/2"}}},"commit":{"parents":[{"commit":"d03d5aaee5e8e42e1a050e46c3d681b74173646a","subject":"[doc] Ubuntu has Ceph Pacific in Wallaby+","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d03d5aaee5e8e42e1a050e46c3d681b74173646a"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-09-02 08:19:05.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-09-03 08:16:15.000000000","tz":60},"subject":"ceph: Update CentOS packages to Pacific","message":"ceph: Update CentOS packages to Pacific\n\nUpdates the CentOS images to use centos-release-ceph-pacific. This is\nin line with Ubuntu which gets Pacific from UCA. Debian remains on\nNautilus.\n\nChange-Id: Ie1e77f9c7d83538c051362a8972745f184c72019\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/933965f1098024212fcd059e352fe8c515ee1291"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/933965f1098024212fcd059e352fe8c515ee1291"}]},"branch":"refs/heads/master"},"c6cba08314816346a9b759bb5da0a2c4ae1a41a6":{"kind":"REWORK","_number":3,"created":"2021-09-03 08:32:54.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/43/807043/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/43/807043/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/43/807043/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/43/807043/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/43/807043/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/43/807043/3"}}},"commit":{"parents":[{"commit":"d03d5aaee5e8e42e1a050e46c3d681b74173646a","subject":"[doc] Ubuntu has Ceph Pacific in Wallaby+","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d03d5aaee5e8e42e1a050e46c3d681b74173646a"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-09-02 08:19:05.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-09-03 08:32:14.000000000","tz":60},"subject":"ceph: Update CentOS packages to Pacific","message":"ceph: Update CentOS packages to Pacific\n\nUpdates the CentOS images to use centos-release-ceph-pacific. This is\nin line with Ubuntu which gets Pacific from UCA. Debian remains on\nNautilus.\n\nChange-Id: Ie1e77f9c7d83538c051362a8972745f184c72019\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c6cba08314816346a9b759bb5da0a2c4ae1a41a6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c6cba08314816346a9b759bb5da0a2c4ae1a41a6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
