)]}'
{"id":"openstack%2Fkolla~807177","triplet_id":"openstack%2Fkolla~stable%2Fvictoria~Iad118c9885b2134a7d071a6629a2f3e3ae3d101e","project":"openstack/kolla","branch":"stable/victoria","hashtags":[],"change_id":"Iad118c9885b2134a7d071a6629a2f3e3ae3d101e","subject":"[CI] Fix periodic publish of centos8 images","status":"MERGED","created":"2021-09-02 16:11:15.000000000","updated":"2021-09-02 22:34:25.000000000","submitted":"2021-09-02 22:12:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"807177","meta_rev_id":"6ff0b77f8a38dd4af6f93ed285e2e5692d6a760b","_number":807177,"virtual_id_number":807177,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-02 22:12:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":2,"date":"2021-09-02 16:21:02.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":1,"date":"2021-09-02 16:21:02.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-02 16:15:20.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2021-09-02 17:39:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b362b03ebfc00277d85f8776f1bc8fe303c107e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-02 16:11:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1ff4dedd8cfbe806aae364cac6fa0c9114ecb9bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-02 16:12:50.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"102a8739c111f98153a74276869a2fcc51cefd90","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-02 16:15:20.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nimportant CI fix","accounts_in_message":[],"_revision_number":2},{"id":"10d298976ad1dc0f045eadcd6404194c2dfae07e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-02 16:15:43.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"eed238ef7e77de249f58454a7e7dc8413541f318","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-09-02 16:19:33.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"ec88dae8516961cef5edc5c8201b8a123e7d1cbc","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-02 16:21:02.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"86e179c6d16cdeb15ff6363f7245a3af27177e68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 17:39:44.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/bc1b745a236948379a999ba2d61405f3 : SUCCESS in 26m 29s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/ec7672bf11af4ad0942f023aa1705f9d : SUCCESS in 35m 48s\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/cd5371f096c74a40a35805f92a9de741 : SUCCESS in 30m 41s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/efd51eb62aa6428e947b1f6bb38ee1db : SUCCESS in 30m 39s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/4f9f375962f84d2e96c81f4a4fc174e6 : SUCCESS in 1h 12m 30s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/9c2c3207baf24a68959966367729a683 : SUCCESS in 1h 08m 03s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/2ffdc365eed844bdb7cbaabf6c439abf : FAILURE in 12m 33s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/7dd3fa73ad2e4d0c866017433b2314e0 : SUCCESS in 1h 10m 12s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/62f1344dfb2e4e389f23d9d260fda219 : SUCCESS in 1h 08m 18s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/e5effc8e30114d61be55c54afd2a6189 : SUCCESS in 1h 17m 11s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ceb93564f6cc4b62b9f20af701629f58 : SUCCESS in 29m 05s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/ebe60566c8e444a5ab16263edeb94020 : FAILURE in 25m 05s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/bd7f9bba9e47403f89f9762aa0adc587 : SUCCESS in 1h 01m 38s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec84dde3180e40579fb4f0a5e0626da6 : SUCCESS in 6m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8b8e60675dce4320b72e2446d3ba146f : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d87d664c7b96437db176287553a659cd : SUCCESS in 8m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8d47553bee74b1499c40067c5821980 : SUCCESS in 7m 14s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/07e09e06f77344c9ba5a31a5e6877a3b : SUCCESS in 25m 38s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/74bf445db102424e98cba9ea8e22a78e : SUCCESS in 29m 23s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/49843dd45fb54b4f8dccb2cf2af0db0b : SUCCESS in 56m 17s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/852c60a5238147598dda3998895f1e73 : SUCCESS in 53m 13s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e904e74f63c641b1b99f14de1bed0b2e : FAILURE in 32m 08s","accounts_in_message":[],"_revision_number":3},{"id":"bc43933401b1012a2b12448351ce8bf897ef6c97","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 18:32:12.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/3a2a9ed4805d4705b3b3bac77f450ade : SUCCESS in 2h 10m 11s (non-voting)\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/48c6354388654b199776984318b3f425 : SUCCESS in 44m 51s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/03e8519351d84b568e1e2fa108056803 : SUCCESS in 38m 39s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4a2ddd4579b04c2688b210bddada80fd : FAILURE in 15m 29s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/9b3c023574054ac8b3eb7e900dd80865 : SUCCESS in 1h 31m 19s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"92635a7ed1701afdd94c539853ff99ed9b73f34a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-02 19:40:07.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"d07686bdcce00f068f944ce7f9b3eb8b586f307c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 21:00:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8-binary https://zuul.opendev.org/t/openstack/build/7e5c9671c9514dd39d12883bdad02736 : SUCCESS in 22m 51s (non-voting)\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/9fa8d81417274b0499ff39ac48901c1b : SUCCESS in 24m 51s\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/49c352c948624c439b2f25697083b920 : SUCCESS in 24m 32s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/525ac9044d2045bd956bd9b1153d56ba : SUCCESS in 26m 54s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/2d51ad3acdee4dac8e16be128add4b85 : SUCCESS in 1h 01m 08s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/0030ed07f8d340ad81f9ce16726a3edc : SUCCESS in 59m 13s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/9c94204497524df5aa2a674eb39a2dec : SUCCESS in 1h 10m 47s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/0dfc53a40baa48fb83eb0ac6dec12b7b : SUCCESS in 1h 15m 30s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/2ed14688c9554169b50596f8ba22378c : SUCCESS in 59m 22s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/648f14adcc324d15bf0b8cd4bf98441a : SUCCESS in 1h 06m 43s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/835d1458a40b471593d7c32dfda60144 : SUCCESS in 19m 51s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/e91d18b7df4148569ac052b6b3969f3c : FAILURE in 18m 05s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/0aaff5d412aa4cfe8a8db57d1966f0d4 : SUCCESS in 52m 29s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/586efcc1bce54d1887856b5c14bb0bd3 : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/189c53d556d247a9bd63e51630e256ae : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2e4c8405c7aa489e9d34fe9ef9b93c67 : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e8bb7c0f063416c96320c05ade6017a : SUCCESS in 6m 14s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4f7c67f2ad4d490abdd442237384545c : SUCCESS in 23m 56s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c96341e8ef40450dba22705371858ad7 : SUCCESS in 19m 44s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5439ce8cfb65421aa9350581dd71cc42 : SUCCESS in 55m 27s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3b0b3ac0f4df46f5a0ce0976d53b38c0 : SUCCESS in 50m 49s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/01a9901788dc48588150ea5edcc36216 : SUCCESS in 1h 03m 20s","accounts_in_message":[],"_revision_number":3},{"id":"5583995ab1d02f770947eefe6ef61b39583ecced","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 21:00:29.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"4c9cdb0252dde813fbc5b1c457a4f3cb407ab98d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 22:12:27.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8-source https://zuul.opendev.org/t/openstack/build/38586ec5511e473eb11b51915d2748ce : SUCCESS in 30m 58s\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/38b1ebd1902648f3bbcc672a079a5692 : SUCCESS in 24m 44s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/34fc42743f2b4e918feb45b613a9e955 : SUCCESS in 58m 33s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/d4e6657e1cbb4510a68c5c1822288143 : SUCCESS in 1h 05m 23s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/d549ff0c71f6418c80591050a2e1ea56 : SUCCESS in 1h 03m 24s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/2aecee35bb8a4468867e2b5c0276e1f9 : SUCCESS in 1h 11m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c13a929cbe8f4023a40c4aff2771f0c8 : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e2b6685882344d02b2fdb68dde7dd123 : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a50c29fc388a445f9f9dc47119b9ce43 : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9ca2860139b4fb7822d299bd30f3ec8 : SUCCESS in 7m 36s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3cd3ffae71dc4afe8e024836094f7761 : SUCCESS in 37m 51s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b3e11d90322e4f74ad1bce7e3a214fd7 : SUCCESS in 1h 09m 34s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/313d647938034b29be6685d3f94c3122 : SUCCESS in 57m 02s","accounts_in_message":[],"_revision_number":3},{"id":"7ebbe52f2df7d755994dc38a25bbe99da8ad74c5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 22:12:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"77e3fd53defde0b600e9a3f3d1f0f8f9cfe66234","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 22:14:06.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/164c0508a687420086f7f99f99425cb9 : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":3},{"id":"6ff0b77f8a38dd4af6f93ed285e2e5692d6a760b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-02 22:34:25.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8-source-aarch64 https://zuul.opendev.org/t/openstack/build/89ec8f89b4e74e8dbfebd54c7e14235f : SUCCESS in 2h 51m 25s (non-voting)\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/270f157df4cc4f25bf44e82abb586d6f : SUCCESS in 2h 46m 07s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5ad7224f55c344fdaabf24427509715f : SUCCESS in 2h 11m 31s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a318291a2c6244519510666574a5ff47 : SUCCESS in 2h 48m 59s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/7449d445a08a437db74d58848adab5e4 : SUCCESS in 2h 13m 06s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d2e8ebe7f27237fd2e58ea9efaae816a64e73294","revisions":{"8251a3f5b3272c4f18c24b0587e9eb4e3781af5f":{"kind":"REWORK","_number":1,"created":"2021-09-02 16:11:15.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/77/807177/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/77/807177/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/807177/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/807177/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/807177/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/77/807177/1"}}},"commit":{"parents":[{"commit":"662a7503df6df223e1ae7fa0544a45cfb30b0d0e","subject":"Merge \"CentOS Stream: update kolla-ansible Zuul jobs\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/662a7503df6df223e1ae7fa0544a45cfb30b0d0e"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-02 16:10:20.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-02 16:10:20.000000000","tz":120},"subject":"Fix periodic builds of centos8 container images","message":"Fix periodic builds of centos8 container images\n\nkolla-publish-centos8 jobs (non-stream8) have been failing on\nstable/victoria with the following error: \u0027base_tag\u0027 is undefined. This\nis because the base_tag variable is only defined for centos8s jobs in\n.zuul.d/centos.yaml.\n\nFix by checking if base_tag is defined.\n\nChange-Id: Iad118c9885b2134a7d071a6629a2f3e3ae3d101e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8251a3f5b3272c4f18c24b0587e9eb4e3781af5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8251a3f5b3272c4f18c24b0587e9eb4e3781af5f"}]},"branch":"refs/heads/stable/victoria"},"221ed9de56dd66a27c0cc74ae2aa370d9a0f13f1":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-09-02 16:12:50.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/77/807177/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/77/807177/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/807177/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/807177/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/807177/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/77/807177/2"}}},"commit":{"parents":[{"commit":"662a7503df6df223e1ae7fa0544a45cfb30b0d0e","subject":"Merge \"CentOS Stream: update kolla-ansible Zuul jobs\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/662a7503df6df223e1ae7fa0544a45cfb30b0d0e"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-02 16:12:19.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-02 16:12:19.000000000","tz":120},"subject":"Fix periodic publish of centos8 container images","message":"Fix periodic publish of centos8 container images\n\nkolla-publish-centos8 jobs (non-stream8) have been failing on\nstable/victoria with the following error: \u0027base_tag\u0027 is undefined. This\nis because the base_tag variable is only defined for centos8s jobs in\n.zuul.d/centos.yaml.\n\nFix by checking if base_tag is defined.\n\nChange-Id: Iad118c9885b2134a7d071a6629a2f3e3ae3d101e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/221ed9de56dd66a27c0cc74ae2aa370d9a0f13f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/221ed9de56dd66a27c0cc74ae2aa370d9a0f13f1"}]},"branch":"refs/heads/stable/victoria"},"d2e8ebe7f27237fd2e58ea9efaae816a64e73294":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-09-02 16:19:33.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/77/807177/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/77/807177/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/807177/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/807177/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/77/807177/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/77/807177/3"}}},"commit":{"parents":[{"commit":"662a7503df6df223e1ae7fa0544a45cfb30b0d0e","subject":"Merge \"CentOS Stream: update kolla-ansible Zuul jobs\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/662a7503df6df223e1ae7fa0544a45cfb30b0d0e"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-02 16:18:50.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-09-02 16:18:50.000000000","tz":120},"subject":"[CI] Fix periodic publish of centos8 images","message":"[CI] Fix periodic publish of centos8 images\n\nkolla-publish-centos8 jobs (non-stream8) have been failing on\nstable/victoria with the following error: \u0027base_tag\u0027 is undefined. This\nis because the base_tag variable is only defined for centos8s jobs in\n.zuul.d/centos.yaml.\n\nFix by checking if base_tag is defined.\n\nChange-Id: Iad118c9885b2134a7d071a6629a2f3e3ae3d101e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d2e8ebe7f27237fd2e58ea9efaae816a64e73294"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d2e8ebe7f27237fd2e58ea9efaae816a64e73294"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
