)]}'
{"id":"openstack%2Fkolla~807273","triplet_id":"openstack%2Fkolla~master~Iee3e8f660eb0315969b7928461d2d6eded7b1fb2","project":"openstack/kolla","branch":"master","topic":"dnm-test-cinder-fix","hashtags":[],"change_id":"Iee3e8f660eb0315969b7928461d2d6eded7b1fb2","subject":"[DNM] Test Cinder fix","status":"ABANDONED","created":"2021-09-03 08:49:14.000000000","updated":"2021-10-18 08:40:32.000000000","total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":false,"meta_rev_id":"aa138124d88f9998d6356271e15616a1a4001859","_number":807273,"virtual_id_number":807273,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-09-03 10:09:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"CC":[{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2021-09-03 10:09:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-03 13:48:07.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"CC"}],"messages":[{"id":"7890b49120d500b073db8806c6e706221b00a5d7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-03 08:49:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a38cd7ee08b3c8bd4927797c77d68bf61942bbd4","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-03 08:54:39.000000000","message":"Patch Set 1:\n\ncheck experimental","accounts_in_message":[],"_revision_number":1},{"id":"bb894f7ef9e5637d64171c76d116027384cbb51c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 10:09:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/cf2c371b1b544668bc1e0032c2c6c780 : SUCCESS in 27m 56s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/7e2d21d392bd415cbf0f09580beff2c9 : SUCCESS in 33m 29s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/e7c7133fbd8e491ba12c3606cca3606d : SUCCESS in 1h 12m 48s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/b0940ec9425b49db8fe90c80d7db1de3 : SUCCESS in 59m 00s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/e228a59524e04a6dbb335c36982bc2dd : SUCCESS in 1h 18m 39s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/3ce690cc9cf9487fafeb5e216de3f06b : SUCCESS in 25m 43s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/71c29b74c2ff45fc90d74825c182dd58 : SUCCESS in 18m 47s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/607bd0693fb743b7a21328e22fb841bf : SUCCESS in 57m 12s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11411f2c579b467dbd4ab5e8050eedf9 : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e16f1347153746d7a6d54f9633e8d41f : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/64ed4e2d419f4dc786b513ffacdae5dc : SUCCESS in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a9e981c9a0fd4ad686b8edd5573fd07c : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1481e072818443b7a892d2eebb8c347e : SUCCESS in 6m 46s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a32799a7dfeb4378897bcf6c06a82fa5 : SUCCESS in 18m 25s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9778e2c053f0471581826c43b222c620 : SUCCESS in 34m 49s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/82fbae7ea5ee4d11a317a9d4d2086bef : SUCCESS in 59m 04s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/aaff74428d6b42faa1818b1a8819d4b0 : SUCCESS in 48m 37s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d7bde39cc1824711913322dfa85833ab : SUCCESS in 57m 47s","accounts_in_message":[],"_revision_number":1},{"id":"a3313e2fa7d0d58cd6023637505dbff04d900c14","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 10:55:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/4fcd5f3760044a89b6c91b20600ca955 : SUCCESS in 49m 10s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0e23479ec76b44ffa69d4725f711fe55 : SUCCESS in 53m 40s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/25520a15942749b99dd9478e8893fedf : SUCCESS in 2h 03m 53s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/31a5599094c34a54a1447f529f6a72a7 : SUCCESS in 6m 51s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4ccadbc4c2ee4c5eb140a23febf1cfa2 : SUCCESS in 5m 52s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/b662f0af6d5b4e9aa600a660f30489f0 : SUCCESS in 1h 13m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a1d0cb7244490ffdc0026abb02ffec583b9ca7c4","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 11:08:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-no-infra-wheels-centos8s-source https://zuul.opendev.org/t/openstack/build/c4584e92845744c390c876d569dc8ceb : SUCCESS in 35m 27s\n- kolla-ansible-centos8s-source-cephadm https://zuul.opendev.org/t/openstack/build/24a3217464684ee7b4fae4c447e6f658 : SUCCESS in 1h 31m 08s (non-voting)\n- kolla-ansible-centos8s-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/140f84015eb34fa0a0641f84c5c48e15 : FAILURE in 1h 27m 26s (non-voting)\n- kolla-ansible-centos8s-source-zun https://zuul.opendev.org/t/openstack/build/86c0cfc86ec4494ca74c799a3428783a : SUCCESS in 1h 42m 37s (non-voting)\n- kolla-build-no-infra-wheels-debian-source https://zuul.opendev.org/t/openstack/build/c635eb7568534e7db57836119c63299f : SUCCESS in 30m 56s\n- kolla-build-no-infra-wheels-ubuntu-source https://zuul.opendev.org/t/openstack/build/88558114d0ab4d56b158b15da9f90f9c : SUCCESS in 30m 13s\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/dc7d3642161246f8ba73682e46c5ceb9 : SUCCESS in 1h 11m 09s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/fa743ffe2dbf4b94a3091b4393da3fcf : FAILURE in 1h 06m 53s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/a1ffebb67ff843d3a690bcfd61a6fbbe : FAILURE in 1h 10m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b72c0966ffd046d03fd942f432d6778c52f58914","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-09-03 13:48:07.000000000","message":"Patch Set 1:\n\ncheck experimental","accounts_in_message":[],"_revision_number":1},{"id":"0c114ce84500ffa3cdcd843b4719ff82536610b3","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 15:27:06.000000000","message":"Patch Set 1:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"6096d6950929a4f5f3d5b69e7a869e64ac4b15c6","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-09-03 15:27:17.000000000","message":"Patch Set 1:\n\ncheck experimental","accounts_in_message":[],"_revision_number":1},{"id":"32c694b63a03993f198d7a16a1417c671a4665e5","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 17:28:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-no-infra-wheels-centos8s-source https://zuul.opendev.org/t/openstack/build/9510f46287d94130b7698a18c4d683f3 : SUCCESS in 39m 51s\n- kolla-ansible-centos8s-source-cephadm https://zuul.opendev.org/t/openstack/build/b0a2ebabfea5482ea9a00da585176a43 : FAILURE in 1h 26m 07s (non-voting)\n- kolla-ansible-centos8s-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/e74f1fe600ce443abaa93b774763e079 : SUCCESS in 1h 44m 49s (non-voting)\n- kolla-ansible-centos8s-source-zun https://zuul.opendev.org/t/openstack/build/bc04ac7ba6094daf81b597cb2988ff28 : SUCCESS in 1h 57m 04s (non-voting)\n- kolla-build-no-infra-wheels-debian-source https://zuul.opendev.org/t/openstack/build/44982985d9674edc8f14c5f2f104f8b1 : SUCCESS in 34m 54s\n- kolla-build-no-infra-wheels-ubuntu-source https://zuul.opendev.org/t/openstack/build/a78fa464fcb7467ca94eb55935121a23 : SUCCESS in 33m 00s\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/2d264c552eaa4d74b5415fa816a2a504 : SUCCESS in 1h 06m 36s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/83a7e723a5f94f8399d09e726f86d421 : SUCCESS in 1h 30m 17s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/3d3a4d5ca6114bf3882c13e93ba2142f : FAILURE in 58m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"026cb4d00da8d9b34457552ea1d11fad9a1bcb2c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-03 18:41:54.000000000","message":"Abandoned\n\nserved its purpose","accounts_in_message":[],"_revision_number":1},{"id":"5966f60d628a99b25cdb5019822ca6754c21902e","tag":"autogenerated:gerrit:restore","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-06 19:39:54.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"1eddac41b846a73e13a2def9c05bb3cb8abbd41f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-06 19:40:01.000000000","message":"Patch Set 1:\n\ncheck experimental","accounts_in_message":[],"_revision_number":1},{"id":"c7f26e43c4949e00eb3aa0f4ebeccfcfb7127b38","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-06 21:12:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/268fbef55c3345ebbdcd4efdf06d13ee : SUCCESS in 1h 30m 43s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/20fa6bd631874dfa8b6fcc703a7a4957 : SUCCESS in 36m 35s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/66ed945af849483fa2cbb7bf5b1f1c42 : SUCCESS in 1h 16m 49s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e3b8e03485314371bfd245d451544de1 : SUCCESS in 5m 48s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/79142b21b6f3467f8754b01ad9a99df0 : SUCCESS in 7m 57s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/cf1d1ceb83c04b8cb25de12471c6008d : SUCCESS in 39m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2d807e294aa00733e546f7eafe34340601f9da02","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-06 21:13:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-no-infra-wheels-centos8s-source https://zuul.opendev.org/t/openstack/build/54ec24564fa74f858cc54b0d43352621 : SUCCESS in 42m 24s\n- kolla-ansible-centos8s-source-cephadm https://zuul.opendev.org/t/openstack/build/a1c4e480a8504d31a4d40cb1cb335c73 : SUCCESS in 1h 26m 05s (non-voting)\n- kolla-ansible-centos8s-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/c1f6e6c3d69d4301ae1b624c076cd174 : FAILURE in 1h 27m 33s (non-voting)\n- kolla-ansible-centos8s-source-zun https://zuul.opendev.org/t/openstack/build/dbbe75de0d7d4af6a404c9b59879f657 : SUCCESS in 1h 32m 03s (non-voting)\n- kolla-build-no-infra-wheels-debian-source https://zuul.opendev.org/t/openstack/build/0c36d448c90948bfac8551ef88149a01 : SUCCESS in 33m 23s\n- kolla-build-no-infra-wheels-ubuntu-source https://zuul.opendev.org/t/openstack/build/bbfb1d02e2bf4db6a306ece07db393b5 : SUCCESS in 31m 31s\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/5163183df0084215951a9cc86b904726 : SUCCESS in 1h 10m 22s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/bb48e987a79149218dff56d6982a468e : FAILURE in 1h 09m 27s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/8186ec2453a04c20af84706b4860a4a5 : POST_FAILURE in 1h 25m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1f4cca3287f80721e2b8aef9d286de0bcba02a02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-07 01:45:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/1f64d6e92d12402bbad0fe2484a82d3f : TIMED_OUT in 6h 01m 48s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/203c372a7a4448f3b4beead1cf414749 : SUCCESS in 43m 59s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/3b5b3796eab84f589a7e0f457ad5b80a : SUCCESS in 1h 04m 13s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/85e8d771d22e438487ed82ae1cbb1e26 : SUCCESS in 1h 00m 59s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/21d9ede7490d4f9d9e3253f8fad65afd : SUCCESS in 1h 16m 49s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/1dd9f19a3bf24994ad2e3d0d639a1178 : SUCCESS in 29m 31s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/13b89d221bfc4687a1abbca501bc5b43 : SUCCESS in 32m 21s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ced3a3561108443a9d9cef14a26b9116 : SUCCESS in 53m 16s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16b7ad68a1be41ee8ec52bada72ffe37 : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d6def48dc0e34935ae60e764f54659cf : SUCCESS in 7m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6901bd4e1598413bb0d4cd082db69837 : SUCCESS in 4m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6d0d87a1ecae442ab19678619b5bb890 : SUCCESS in 7m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f79282e8f624fd3b0c4301c4fad45ac : SUCCESS in 7m 57s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a1dbdb55ea1847f7a68a2f4d00e67745 : SUCCESS in 20m 13s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/dea5a8fbb5f542959a395725e175b98e : SUCCESS in 27m 46s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c7eda2b746ae4e8f810b9349be9bc3b6 : SUCCESS in 44m 44s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7f10bf40e59b449e8f236d495151db4b : SUCCESS in 46m 25s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0d2d3dff420a4f5d834b91640954b137 : SUCCESS in 54m 56s","accounts_in_message":[],"_revision_number":1},{"id":"1410bbc3dfb2a8b89fd7a4390b36cdfa8c4ed06f","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-09-07 09:12:05.000000000","message":"Patch Set 1:\n\ncheck experimental","accounts_in_message":[],"_revision_number":1},{"id":"8fcc2cb499de9f9b58ecfa0dd0419d77999cee9d","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-07 12:11:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-no-infra-wheels-centos8s-source https://zuul.opendev.org/t/openstack/build/05d43791cf6942adb1ffc02862900e25 : SUCCESS in 44m 00s\n- kolla-ansible-centos8s-source-cephadm https://zuul.opendev.org/t/openstack/build/cf1798fba33647ad81bfa27dab04b804 : SUCCESS in 1h 38m 41s (non-voting)\n- kolla-ansible-centos8s-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/c5f49c976de7441791ac750e01c1e27b : SUCCESS in 1h 41m 09s (non-voting)\n- kolla-ansible-centos8s-source-zun https://zuul.opendev.org/t/openstack/build/b77d7bbd31c94bf8850a4d5202e8b5c3 : SUCCESS in 1h 33m 39s (non-voting)\n- kolla-build-no-infra-wheels-debian-source https://zuul.opendev.org/t/openstack/build/efa44f2a699849c1979fd005ef301253 : SUCCESS in 33m 39s\n- kolla-build-no-infra-wheels-ubuntu-source https://zuul.opendev.org/t/openstack/build/a06cf2862427424da19211a746cc938b : SUCCESS in 37m 01s\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/fe4a796928ec4f759ec03f5ba1e167b1 : SUCCESS in 1h 07m 02s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/e644ccef54c14e7d890e5922e6945ab9 : SUCCESS in 1h 31m 29s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/19373ea8a1744b12ae2481b2f1e5ee98 : POST_FAILURE in 1h 21m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"08dcaa5a0f617cdef14341ec59e73664df65fd09","tag":"autogenerated:gerrit:abandon","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-07 12:55:23.000000000","message":"Abandoned\n\nhopefully this will be enough :D","accounts_in_message":[],"_revision_number":1},{"id":"ae502bd2417897aa7cd2aa99a0ce92baa7d5210e","tag":"autogenerated:gerrit:restore","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-07 15:01:20.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"60c3aaa7cdacbcd9c277a5867ee7388a23d026b7","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-07 15:01:33.000000000","message":"Patch Set 1:\n\nmeh! 😄","accounts_in_message":[],"_revision_number":1},{"id":"cee66db72bf735f81068ff084374098c04ef2dec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-07 16:21:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/9d07006383c843e980496a7e9dccb037 : SUCCESS in 26m 17s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/d6428514f9ba49fd8c5545c73d3fbd84 : SUCCESS in 39m 14s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/5923abdcc1e04b5fb203017ff78b744e : SUCCESS in 1h 07m 18s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/00729e235b5040c8a070a549dd8fa83c : SUCCESS in 1h 08m 07s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/948f4d0c178b46febb28016dcf0f3b38 : SUCCESS in 1h 15m 48s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/d38130d669e544cb927dbcc72f506f25 : SUCCESS in 28m 59s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/9defe82172124e46b379b69405563f5b : SUCCESS in 17m 46s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e82b6faaa77c4a2dafaa839e74d027e0 : SUCCESS in 59m 05s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b29bacc35ed4b1ebf505d274ddd33c4 : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a5e3a743952472188dcce5c902a5e3a : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a6ebcbd53c6048188ed2450b484c9b30 : SUCCESS in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e84d3a303488410eb0bcdf22ffaa72cc : SUCCESS in 4m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e37778ab10b476b9bb3fd168a438d30 : SUCCESS in 6m 25s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9def4ad97a044eda8ea63d034dae2b61 : SUCCESS in 18m 32s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6c8feda1ee584a2fb51256bcb11f4214 : SUCCESS in 29m 26s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/221220ad0dbb4be487f18684a9ec0184 : SUCCESS in 52m 03s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5d8b03909fc34e91a30b257d3f18d99f : SUCCESS in 1h 02m 13s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f387604d7ce9498995046b5cbd0b290b : SUCCESS in 56m 56s","accounts_in_message":[],"_revision_number":1},{"id":"f68a54d795c669b51f0ae9b5e089aaa9ce9bd317","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-07 16:45:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/1739ececf005450880e5f11cc35e905d : SUCCESS in 1h 28m 33s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/04d1b648e4a944d486e90a86bf4bfff4 : SUCCESS in 50m 42s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8d38ac3bb5d64b66bebe75a2b54651ae : SUCCESS in 1h 41m 54s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b3460a76fff9495497514a17b7fc20ea : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/aa11dd11a22b45c0bc83ced070bc9c61 : SUCCESS in 7m 20s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/6dde91fa7d574522aa0a08671df80c59 : SUCCESS in 45m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3f19db45cb37654d6fc4bdd601c2a69947c0b8ab","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-09 10:43:36.000000000","message":"Patch Set 1:\n\ncheck experimental","accounts_in_message":[],"_revision_number":1},{"id":"38ccc75f124286dff662aaf382075dc1c765b34f","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-09 13:18:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-no-infra-wheels-centos8s-source https://zuul.opendev.org/t/openstack/build/4522750e3fdd40b79aef3d0a173ed307 : SUCCESS in 37m 58s\n- kolla-ansible-centos8s-source-cephadm https://zuul.opendev.org/t/openstack/build/a2baa23e95c04dc1a812deab8ffe05ca : SUCCESS in 1h 26m 34s (non-voting)\n- kolla-ansible-centos8s-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/db53561c86fd48c1b379d3c19ecaa19b : SUCCESS in 1h 42m 58s (non-voting)\n- kolla-ansible-centos8s-source-zun https://zuul.opendev.org/t/openstack/build/cf08c32745024e98bd1b3bc1edfc6ec1 : SUCCESS in 1h 38m 25s (non-voting)\n- kolla-build-no-infra-wheels-debian-source https://zuul.opendev.org/t/openstack/build/288c98a12e1f480a8bbef7c85c4ec3a4 : SUCCESS in 38m 16s\n- kolla-build-no-infra-wheels-ubuntu-source https://zuul.opendev.org/t/openstack/build/6916bb45156c4733868997123e96c4e3 : SUCCESS in 27m 03s\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/e3f4df5d407346f0926c6908bffddbdd : SUCCESS in 1h 15m 28s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/9d3f497fa0ef4651a4e5c7dd8bf48aad : SUCCESS in 1h 19m 45s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/2be995988adc46eeb84ab807cc764d07 : SUCCESS in 1h 10m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3ca8203d9a135f66d6228c98594c7d93d67e620d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-14 16:39:47.000000000","message":"Patch Set 1:\n\ncheck experimental","accounts_in_message":[],"_revision_number":1},{"id":"e979042ec57bf8b515cdcaac0786b8178d03a5e8","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-14 21:12:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (experimental pipeline).\n\n- kolla-build-no-infra-wheels-centos8s-source https://zuul.opendev.org/t/openstack/build/1e35f1cc29ff406bba62d5e77bbc64d5 : SUCCESS in 42m 29s\n- kolla-ansible-centos8s-source-cephadm https://zuul.opendev.org/t/openstack/build/83ad1cb7b3e8424e9e28de16395b871a : SUCCESS in 1h 35m 14s (non-voting)\n- kolla-ansible-centos8s-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/a923031944b041949a1486671821628c : SUCCESS in 1h 35m 43s (non-voting)\n- kolla-ansible-centos8s-source-zun https://zuul.opendev.org/t/openstack/build/94876a2d5cb44187994731ac6f683eb1 : SUCCESS in 1h 47m 36s (non-voting)\n- kolla-build-no-infra-wheels-debian-source https://zuul.opendev.org/t/openstack/build/774dcacdd62445f98464a5da08f85f02 : SUCCESS in 32m 04s\n- kolla-build-no-infra-wheels-ubuntu-source https://zuul.opendev.org/t/openstack/build/ce3fdcdb50224328bf95ea190028c35a : SUCCESS in 30m 48s\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/9cd86ba4bcca40b080a3eed055c9b106 : SUCCESS in 1h 08m 22s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/eda7163a51204942869cb4910b4bdaef : SUCCESS in 1h 33m 07s (non-voting)\n- kolla-ansible-ubuntu-source-zun https://zuul.opendev.org/t/openstack/build/7e43eb74c81942e29d2deea24461f35f : SUCCESS in 1h 06m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"aa138124d88f9998d6356271e15616a1a4001859","tag":"autogenerated:gerrit:abandon","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-18 08:40:32.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c4f9f7331b3aef80715e906ade0b53b1fc20a185","revisions":{"c4f9f7331b3aef80715e906ade0b53b1fc20a185":{"kind":"REWORK","_number":1,"created":"2021-09-03 08:49:14.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/73/807273/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/73/807273/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/807273/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/807273/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/807273/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/73/807273/1"}}},"commit":{"parents":[{"commit":"d03d5aaee5e8e42e1a050e46c3d681b74173646a","subject":"[doc] Ubuntu has Ceph Pacific in Wallaby+","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d03d5aaee5e8e42e1a050e46c3d681b74173646a"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-09-03 08:48:49.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-09-03 08:48:49.000000000","tz":0},"subject":"[DNM] Test Cinder fix","message":"[DNM] Test Cinder fix\n\nDepends-On: https://review.opendev.org/c/openstack/cinder/+/807083\nChange-Id: Iee3e8f660eb0315969b7928461d2d6eded7b1fb2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c4f9f7331b3aef80715e906ade0b53b1fc20a185"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c4f9f7331b3aef80715e906ade0b53b1fc20a185"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
