)]}'
{"id":"openstack%2Fkolla~807279","triplet_id":"openstack%2Fkolla~master~I55327cc3c686eae8f2da85ea957cae851e3041e4","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I55327cc3c686eae8f2da85ea957cae851e3041e4","subject":"toolbox: Move to ansible-core 2.11 and collections","status":"MERGED","created":"2021-09-03 09:38:34.000000000","updated":"2021-09-16 16:19:16.000000000","submitted":"2021-09-16 16:16:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":4,"has_review_started":true,"submission_id":"807279","meta_rev_id":"e1787cf76342eb1184f35829141f04a99a3abd48","_number":807279,"virtual_id_number":807279,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-16 16:16:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2021-09-16 14:21:44.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2021-09-16 12:18:45.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2021-09-16 16:19:16.000000000","post_submit":true,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2021-09-16 14:21:44.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-03 10:38:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-03 11:58:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-14 08:43:22.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-09-15 09:21:17.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2021-09-16 14:21:44.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"}],"messages":[{"id":"6270ba8213485892599234679edb578f6636fd61","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-09-03 09:38:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fda153732bf11e03a7f6ecfbb8bf8331c22a54da","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 10:38:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/68628d980ce2492bb1362b28f4c0bdbd : SUCCESS in 53m 34s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/31ae9ba674714119a45331cd96bc4160 : SUCCESS in 50m 47s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e35a9b0d7ea34742bb8918bebb1ac0ff : FAILURE in 43m 15s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a1a7c2b5a1634d0c9b7b5ab0588d21d5 : SUCCESS in 6m 26s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/992e3fcf4a0c4511ae02658105ac5be7 : SUCCESS in 6m 33s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/fc3cf7f85cd5462db491d5d18f41b5f4 : SUCCESS in 54m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bdf1ef046b33c5c45ee42ab69f9b72b8a9aff758","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-09-03 10:39:39.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c0d4ab092ffef16ba88ff366c7fcc4b295caecd4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 11:58:53.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/b25ef50cf08744abafcd5b658ddeb026 : SUCCESS in 23m 30s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/b86e5c3aaba34796bd3c517690272194 : SUCCESS in 36m 36s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/780be49ed94348879501ee5f05033ac6 : FAILURE in 40m 00s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/d0000185fcd141f7b8833dc80c235f1b : FAILURE in 42m 51s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/352fbe551a1248e2aaca8b06b893efc1 : SUCCESS in 1h 12m 09s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/f57ec54f1e334c44bf23f4b6314ce641 : SUCCESS in 30m 57s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/62294791e1e7491099e1d5c2420de92c : SUCCESS in 18m 37s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/74a68d0ee8d840a58ea408f3023016bf : SUCCESS in 59m 04s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbce1c54c55147f3a403a177d9472219 : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/634ef347b9504f299b6461e35ba258d2 : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4800851b9f9247e9aac4e487775e0cb2 : SUCCESS in 4m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/30dd556ad9474550884e4da0ace23987 : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f35d20462d764a5b8b2377044c9edfe6 : SUCCESS in 6m 29s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/20e1e95988fd4a1eb0142cd0d32ca566 : SUCCESS in 19m 43s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/779576ba692c4ba3a53643a62364707b : SUCCESS in 33m 49s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9ad38aef370042bdb2da7547ecdf880c : SUCCESS in 48m 36s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5ec62976645e48229a0f4eb839add877 : SUCCESS in 48m 58s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/409ce0eb1a1c44e1b09c710f33b4ebdb : SUCCESS in 58m 17s","accounts_in_message":[],"_revision_number":2},{"id":"ad5aad0ed06223098c8eb20c2131e2c9c8eb20d0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-03 12:29:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/dbd25a9b55c04d3fa476967cadeafd3f : SUCCESS in 1h 44m 07s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/453d44bdc7da4e9f965c3016081cc0cb : SUCCESS in 1h 46m 53s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ec652a12276a448a84ad76c079c8c9d6 : SUCCESS in 1h 19m 20s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cd6f37c9986e4f76ba0108054f4bba42 : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/070ad20399e6467ea49548a4cb35c995 : SUCCESS in 6m 37s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/304dd0e18dbc4a4ba98b6ab6f5802bcb : SUCCESS in 33m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"607a18caaf4c57442e254a2034048c11c53229fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-09-06 08:35:23.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"06361eebc5e9386a19da28d3c4a0161f345fa081","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-06 10:14:11.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/39b2d35d69a2458ebb0e8675e7c91ad1 : SUCCESS in 25m 53s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/97558a19a9744cd5ac28d99b71787a83 : SUCCESS in 42m 09s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/308ef6592f7646a8a7620f096aefad1d : FAILURE in 42m 55s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/7c31f150707a4e0e814ac68ff9faca93 : FAILURE in 41m 52s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/2500a408c17c4a069dba431de51d33c0 : SUCCESS in 1h 07m 40s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9a175d5c24ba45d9bb935df2d69b86c1 : SUCCESS in 26m 38s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/0e6328c3d6d24fed8dbdf281452407f7 : SUCCESS in 27m 35s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/5af44bce799d4adc993bd2387f23fd85 : SUCCESS in 1h 06m 04s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8049c1005724b19bfd863ea459c36e6 : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2d31990b2d934cc099fc2854e245b565 : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/30163a5d908f403fa19f518654d2dedb : SUCCESS in 5m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a7dd55237e2d477f9223101038b95d14 : SUCCESS in 7m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edc8a94491ee4edc847e10c1b6d231fd : SUCCESS in 6m 11s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b421a5162087448090746319f4c013c3 : SUCCESS in 31m 18s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a9ed9dd4518e4e6f8fcbdcad18bb7da4 : SUCCESS in 49m 40s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e8282343b8124b5cbd992604018344b7 : SUCCESS in 53m 30s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a38b44a60e384190afeae3924c17dcb7 : SUCCESS in 43m 13s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ed844896af9946289a8bc2de4437acac : SUCCESS in 1h 33m 15s","accounts_in_message":[],"_revision_number":3},{"id":"4db05b80790f074c91a99a2230a0d6045ad29018","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-09-06 10:24:16.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"3e0db187bee7c92300202609d61d7c49967d7ddc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-09-06 11:26:49.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"94bc015f0544282f37001fe866cdf0ded2d3870d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-09-06 11:45:29.000000000","message":"Patch Set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"9417fc256629714a8f28472cade7107731d3cf0c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-09-06 12:44:26.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"0b011ea9a1ee69244db171ffdfe816d1a9d2a717","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-09-06 13:25:23.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"13a2a8bcf2717e2618bd98083948cb37051db050","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-09-06 14:06:55.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"95cc515701a64c4d43fd7de6585b43d4d82b7d5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-06 15:17:31.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/29da345a1bb2438794a66a0c189476e2 : SUCCESS in 32m 35s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/b7ca8e3ee0a94c769eff17a29ad6c17e : SUCCESS in 41m 30s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/5a5b529c427b4178bf38ed42732dceb4 : FAILURE in 40m 30s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/f1bd713293dc4899b0f03f202fd45b9d : FAILURE in 44m 41s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/234a8ee4d4e547a09b7f950f8ea13ee2 : SUCCESS in 1h 09m 03s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/a849bbbfe3b04ff2a7c2645a5677d6e0 : SUCCESS in 29m 34s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/5d3ad184160041a8a6907ff64bcc6da6 : SUCCESS in 16m 28s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/85032a4990ff4c138db945c89999e424 : FAILURE in 37m 23s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59f9c3c6f1854dab9c0ca84656be6506 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc9f2a917aea47f2b383e141d5812db2 : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c72d7ec2954c4b42ba41d7ebc8150c8a : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/549beefc85024719b9642e746739857f : SUCCESS in 4m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/789e6c5f31294cc9a665220b7b010b97 : SUCCESS in 6m 48s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5743151bc985457db6604e85a172f19e : SUCCESS in 15m 49s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/db0d2ba1649b4dd29500172d9fca02f3 : SUCCESS in 28m 18s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f8c3defaf91e418cada0c41e6937d3fe : FAILURE in 29m 17s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c7e8bc9cf6be4c2d9c8e1d5365079b71 : FAILURE in 31m 45s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/42d08b08896b494e8779328579d83c82 : SUCCESS in 59m 48s","accounts_in_message":[],"_revision_number":9},{"id":"cc738e00b6502d00a6b42afeb1f64cc2225b07cd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-06 15:40:09.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/b49c5d8fa9a24f9491268b09998ea696 : SUCCESS in 1h 30m 48s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e72e781c493a4ba5a9f9ec238cb21e3a : SUCCESS in 39m 19s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/091562bdc4f24897aace2213ef6e1b1e : FAILURE in 1h 17m 53s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7e189e88b08d487f806c5053d26dfd8a : SUCCESS in 4m 42s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/918efa1a46634964aa5dce0f2ac2d535 : SUCCESS in 7m 52s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/3c864e0cb1c743d099d43442035aad8d : SUCCESS in 41m 38s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"7a52978303507e914468380398ad87a53e8935c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-09-07 07:50:46.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"edc0c5f6467315dd72efd0d0b1b177de17f9ed2a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-09-07 07:51:43.000000000","message":"Patch Set 11: Published edit on patch set 10.","accounts_in_message":[],"_revision_number":11},{"id":"4831868d3c5ab6aeabfefb58a833680e5a079302","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-09-07 07:52:38.000000000","message":"Patch Set 12: Published edit on patch set 11.","accounts_in_message":[],"_revision_number":12},{"id":"e4a657219fcdd4a635c4dff9928f73669b53e08d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-09-07 07:52:54.000000000","message":"Patch Set 13: Patch Set 12 was rebased","accounts_in_message":[],"_revision_number":13},{"id":"675669e60d567fae1df41eca8860b3f00e4477e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-09-07 07:53:34.000000000","message":"Patch Set 14: Published edit on patch set 13.","accounts_in_message":[],"_revision_number":14},{"id":"0632c4791346cd6d26f06263c2adbfd05930d981","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-07 09:11:18.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/e0e82c4580d84c6c857ab67560d29d08 : SUCCESS in 30m 21s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/d6c4b5dbc16747769b69cf9943ed41f7 : SUCCESS in 42m 12s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/1f0d6f0b94ac4c2b89ef2cb50b1234cc : FAILURE in 45m 51s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/ccc5a49786cf4db981e60bab4c38312c : FAILURE in 45m 04s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/42ad92c8bb4147d8936facc9617faf4c : SUCCESS in 1h 14m 37s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/00a024a6748544e7ab66aa0fbb88643a : SUCCESS in 26m 31s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/2deff46f958b498ca509684c50f3e78b : SUCCESS in 18m 53s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/79cebe50a3b84138bc7cc9441cb528cb : FAILURE in 35m 36s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3203e600185d42558cd3585fa6b58160 : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8c8824ee76ba474f85e194f6c0ab7d09 : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9c882639421243fda057942153efff39 : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fefbe84709e74d5a814b77f9ce65001d : SUCCESS in 4m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f75ae05dd194cd39fd3eeb13e2051c3 : SUCCESS in 7m 08s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/659a71b5c26e4120a8db171e3b769bed : SUCCESS in 19m 42s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6f0c12f5039d4b31a8a8cc05e868a3be : SUCCESS in 25m 29s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/97ceb64cd4344181a7cb9f392e175be1 : FAILURE in 34m 10s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1b1951afb05b4cdba2dd342e5df3f95a : FAILURE in 32m 16s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/97b33365e5cb4828830484bb5e248f70 : SUCCESS in 1h 00m 30s","accounts_in_message":[],"_revision_number":14},{"id":"8b2f922c6af6413ced227ce4836181c6ecc05f8f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-07 09:25:01.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/ee4cfe27999b4dcb99534a1aec50cb38 : SUCCESS in 1h 30m 46s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3b9141e5657f4198b0fcddc9a938ce74 : SUCCESS in 1h 25m 53s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ff43e235e00947ca983c25b9c30e4d11 : FAILURE in 1h 10m 30s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6c193a0babc04091807bb06ed1e6cd19 : SUCCESS in 7m 15s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f2856e86246c4752b08703c4fb9794d5 : SUCCESS in 7m 08s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/08745237f6d445a6a9feaa51ecefe2bb : SUCCESS in 1h 00m 57s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"d9f8d9f72ad2b8b04ec8c225ca4e2496734b5094","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-09-08 06:29:37.000000000","message":"Patch Set 15: Patch Set 14 was rebased","accounts_in_message":[],"_revision_number":15},{"id":"08c958ed5cb4caff6c34409bfd87adbde4e67d1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-08 07:53:51.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/6c9bf3dc31b448eabf9adf295d0805b0 : SUCCESS in 27m 29s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/46557b6d50f54c1b8d7f6ee46f5b38fc : SUCCESS in 41m 56s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/a32d6067c113419ebedac98682941be5 : SUCCESS in 1h 16m 39s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/d39dd02f590a44df85211308792d8111 : SUCCESS in 1h 13m 59s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/37387cfeb6e04bf6be1fa58663e85bb9 : SUCCESS in 1h 16m 57s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/064142460b4e482983993910a38034fc : SUCCESS in 38m 34s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/3cbe76eb7ab34060af249c91ad0228e6 : SUCCESS in 18m 43s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/109a9b6f02a74e439e10e221431d088f : SUCCESS in 1h 02m 40s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc748cb40357481a9086b753f2328fc3 : SUCCESS in 7m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6afa7533718d4540bb20b70a24d562ae : SUCCESS in 5m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/81c9ac2533e34938b7c3f3159b5f3ead : SUCCESS in 6m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/54bfdd73b8744a3db27acee03a36b3da : SUCCESS in 7m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6462527289c342c5b54d9f41d165e9c4 : SUCCESS in 8m 55s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/56b906d622e94a05a78db5201e3549a2 : SUCCESS in 24m 13s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/f417407640b04430adc6c04fa5b72107 : SUCCESS in 27m 03s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c865ece0c09f4eaa979b9ff94f98b547 : SUCCESS in 50m 03s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d5ed52d58c924fd4b086bbbced63a3ef : SUCCESS in 48m 59s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/13aac1525b2f4300adeacc818d509758 : SUCCESS in 58m 33s","accounts_in_message":[],"_revision_number":15},{"id":"dc8f0f00ebc3dc46a747123da8c9f71fde346ed9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-08 08:38:03.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/46ff8db324584d54957f76dc7a1b3d46 : SUCCESS in 1h 04m 09s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ab961465691d48a9aa367eeb03045edb : SUCCESS in 1h 07m 20s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/38a9a64c7bef455a8d5008dc8d0edb37 : SUCCESS in 2h 03m 08s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/15e86a9b5f474cd8a9d5a01b9dd58d1d : SUCCESS in 9m 06s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/58a53233df804b7a9411216263095e70 : SUCCESS in 9m 39s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/4098cf69be0245ab99fc64fb51769e6d : SUCCESS in 45m 30s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"3e9c42284e597815f62cb1a0be743adc3ee96c74","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-09-08 08:47:42.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"689affd9d8ac4b367af52fc06ad4cdfad206ab97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-08 10:08:34.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/4a36e176273f48d686dac769e2104167 : SUCCESS in 27m 29s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/31ca3b33d78b4bb68451f844b655ba55 : SUCCESS in 39m 24s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/830e94a3acbe468d838e8b71a1d0599e : SUCCESS in 1h 06m 27s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/f40da3a13ba44356b52e4d47d6edd12b : SUCCESS in 1h 05m 32s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/d9319ee3b0fa42ecb0f2de05ab7c74b9 : SUCCESS in 1h 19m 34s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/e276471e3f0c4530875e20a65fcc8ec4 : SUCCESS in 32m 57s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/facc1b091e604dd19e58e67291ca91e3 : SUCCESS in 20m 52s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b52f6bb5592f4be3a16fb9becbc9615a : SUCCESS in 54m 19s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ea14f37a0cc479ea462c14c9ad9d8e4 : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e4b33b88ea014e5abb65f367b20715bb : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8fe0c3ffc3da48d3b0942913f2f4e603 : SUCCESS in 4m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f1641a0df321422d8ccbdd01a3b35d43 : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9088d55b60164805bef0155a696eb028 : SUCCESS in 7m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5c05595089a84b3898fbf0228c51290a : SUCCESS in 5m 35s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/91a596075a9746d180a74fbc8a0b61bf : SUCCESS in 21m 39s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/257f13073a804b5f923ace9b3c08b4f9 : SUCCESS in 28m 45s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f4b1c5bbcb3a494caae310994d520c13 : SUCCESS in 53m 21s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/51f3c098299749088ff82a3a34d9e566 : SUCCESS in 50m 57s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/663fa848be724c8f92d29744c8c9259d : SUCCESS in 1h 03m 35s","accounts_in_message":[],"_revision_number":16},{"id":"79bd99646aab0ecbf6815d35a2d750b58177d1bd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-08 10:16:07.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/2f74f277224344b78a5526b4de174025 : SUCCESS in 1h 00m 19s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9d3711a6fbc840f4be03755ed7ee1de2 : SUCCESS in 39m 11s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a9d5f42c14ec4692bad57ee42498660c : SUCCESS in 1h 22m 35s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2a0f041d34974502bf3d6a0e92c0052b : SUCCESS in 4m 28s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a6fc9336614d482cb4a61e8792839cb2 : SUCCESS in 5m 36s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/17f91e8d8d6f4cf688adc40aca6f3109 : SUCCESS in 39m 13s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"33d1d7f4b0e36237dcf8f142629f6dd1161fcafa","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-09-14 08:43:22.000000000","message":"Patch Set 16: Code-Review+2","accounts_in_message":[],"_revision_number":16},{"id":"23bb8b4b6ea88f4478b5bc7683af13a0a2d6182c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-15 17:11:41.000000000","message":"Patch Set 16:\n\n(3 comments)","accounts_in_message":[],"_revision_number":16},{"id":"3857a8f5f51020420428ce2771476fd7b55e97dd","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-09-15 19:16:54.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"c6e3d9c0686a78a44d238e0804573bd61b9dc737","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-09-16 08:27:47.000000000","message":"Patch Set 16:\n\n(2 comments)","accounts_in_message":[],"_revision_number":16},{"id":"226d409e25d709e9812f38f0610201c429a7af68","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-16 09:57:18.000000000","message":"Patch Set 16:\n\n(2 comments)","accounts_in_message":[],"_revision_number":16},{"id":"91667e70aaffd270a8aff0a84ce7411f3e8b7561","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-09-16 10:19:57.000000000","message":"Patch Set 17: Published edit on patch set 16.","accounts_in_message":[],"_revision_number":17},{"id":"e979431d587264676febabcd6bfafd19ec46bd40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-16 11:43:02.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/f7403ea618f24aedb0f323f2590def91 : SUCCESS in 24m 21s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/9c39d4e880dc4714a91da01137b24ff0 : SUCCESS in 34m 08s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/055d12d51197470cafcf7685ac9896e0 : SUCCESS in 1h 17m 23s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/491b44242720430bb5b769a4485a7d3e : SUCCESS in 1h 14m 32s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/79f00555bb104eb6b5428af556c208fb : SUCCESS in 1h 13m 16s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b83014d99e804198848d99b4c1ba8d48 : SUCCESS in 36m 40s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/0988fd1dc55043c8b2821da222dab7a3 : SUCCESS in 17m 37s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/86b6e022ce5a4991ba90248d17479795 : SUCCESS in 59m 28s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e6206686b2447ba9fa92b095df34573 : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1dff45b689474119bdaeb20b6946cf29 : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0292de7345c6445aa5c9d029fd06e45c : SUCCESS in 4m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fd059699530448c1ab8961e79c57cea3 : SUCCESS in 4m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/934a7aca5b1043239ef8dd7c615a985e : SUCCESS in 6m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ec3716af5ffb458bb70cad7f7820b1c5 : SUCCESS in 4m 57s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9baf8d9cee6c4b9c8f10831a367d26b8 : SUCCESS in 18m 32s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/44bae69d64ed4ef497d68355f6c0f665 : SUCCESS in 27m 26s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ce2d75e0b39f4460a56bc586f78064c7 : SUCCESS in 50m 47s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/fc7d2a16c639421b8da01ad340336365 : SUCCESS in 53m 08s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/fa194a93c3f44ff8b6fca535c7202142 : SUCCESS in 57m 00s","accounts_in_message":[],"_revision_number":17},{"id":"ee5a7835ce9662b02df2a9047a8f600c2badde99","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-16 11:51:24.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/881ce213c81a4440bc44cbcf47005b67 : SUCCESS in 1h 29m 07s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/996b395ff0f4409c94a4d24e91f2b180 : SUCCESS in 48m 03s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8f10265a611148c2b86292965fe99863 : SUCCESS in 1h 21m 06s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6b3ab839e2074f509cdee5de595154a7 : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/472c2d0bc17f40b29fd4025dd9fb3cb5 : SUCCESS in 5m 34s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/abdcc88de98144358235e2b12774fb0e : SUCCESS in 45m 29s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"5c49be5b613d8e3c02e08fff701f60828f2a59ea","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-09-16 12:18:45.000000000","message":"Patch Set 17: Code-Review+2","accounts_in_message":[],"_revision_number":17},{"id":"7153b519e3c395118383bf90eeba7d64611a7325","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-09-16 14:21:44.000000000","message":"Patch Set 17: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"3e72c8810463ddd82f8d22d579cb61ab5cd2a21e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-16 14:23:17.000000000","message":"Patch Set 17: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":17},{"id":"b99201efcea9ba88363874cdac795dded077ad64","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-16 16:16:04.000000000","message":"Patch Set 17: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/7b7b27b0faa848b695c19919c0283d12 : SUCCESS in 36m 02s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/a9c45c2fe4654620a25a220e38934883 : SUCCESS in 1h 18m 23s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/75b7fa5c8c3545cb840863d5a18f7ff3 : SUCCESS in 1h 44m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20738f1800744a5a8eabe00613769d76 : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7f0e837e257d48128c859e8c7217995f : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ce4b501512444e32b684d210ee708636 : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a869ce56dbd4e229ede3c39991af317 : SUCCESS in 6m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/91b36f7fd959473f818b1c4ee02a70ce : SUCCESS in 5m 42s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/61a8ae0e8e54471db1f5af2cd40e3463 : SUCCESS in 32m 51s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/26bd1ec2127f449cb3453aec6c5531e9 : SUCCESS in 53m 37s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3e3c6e2918ee4086a12da01ed7baecc7 : SUCCESS in 1h 03m 20s","accounts_in_message":[],"_revision_number":17},{"id":"c2a0a12ac23a493971cfa6348fd6b43432d7ca94","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-16 16:16:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":17},{"id":"9414e3b277b0e901dfcc3727de699cd072c53ee5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-16 16:17:57.000000000","message":"Patch Set 17:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef1e69d20376464fbbeae7ece56da3e0 : SUCCESS in 1m 13s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bc1911793f7f47e68a48af929665a388 : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":17},{"id":"e1787cf76342eb1184f35829141f04a99a3abd48","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-16 16:19:16.000000000","message":"Patch Set 17: Code-Review+2\n\n(1 comment)\n\nfollow up on that whitespace...","accounts_in_message":[],"_revision_number":17}],"current_revision_number":17,"current_revision":"6eeb2cbdd106ec92f99b500acd8aaeea23278216","revisions":{"50af95dce5178a55f773606f379eb75cef1648e3":{"kind":"REWORK","_number":1,"created":"2021-09-03 09:38:34.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/807279/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/807279/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/807279/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/807279/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/807279/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/807279/1"}}},"commit":{"parents":[{"commit":"d03d5aaee5e8e42e1a050e46c3d681b74173646a","subject":"[doc] Ubuntu has Ceph Pacific in Wallaby+","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d03d5aaee5e8e42e1a050e46c3d681b74173646a"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-09-03 09:38:14.000000000","tz":120},"committer":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-09-03 09:38:32.000000000","tz":120},"subject":"WIP: Move to ansible-core and collections","message":"WIP: Move to ansible-core and collections\n\nChange-Id: I55327cc3c686eae8f2da85ea957cae851e3041e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/50af95dce5178a55f773606f379eb75cef1648e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/50af95dce5178a55f773606f379eb75cef1648e3"}]},"branch":"refs/heads/master"},"0f0940b8a1c7ec02d77719148aaedeee8e9f38fe":{"kind":"REWORK","_number":2,"created":"2021-09-03 10:39:39.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/807279/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/807279/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/807279/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/807279/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/807279/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/807279/2"}}},"commit":{"parents":[{"commit":"d03d5aaee5e8e42e1a050e46c3d681b74173646a","subject":"[doc] Ubuntu has Ceph Pacific in Wallaby+","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d03d5aaee5e8e42e1a050e46c3d681b74173646a"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-09-03 09:38:14.000000000","tz":120},"committer":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-09-03 10:39:31.000000000","tz":120},"subject":"WIP: Move to ansible-core and collections","message":"WIP: Move to ansible-core and collections\n\nChange-Id: I55327cc3c686eae8f2da85ea957cae851e3041e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0f0940b8a1c7ec02d77719148aaedeee8e9f38fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0f0940b8a1c7ec02d77719148aaedeee8e9f38fe"}]},"branch":"refs/heads/master"},"fc3506a4a549b221f15c73139c2466650cf1bca4":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-09-06 08:35:23.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/807279/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/807279/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/807279/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/807279/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/807279/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/807279/3"}}},"commit":{"parents":[{"commit":"d03d5aaee5e8e42e1a050e46c3d681b74173646a","subject":"[doc] Ubuntu has Ceph Pacific in Wallaby+","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d03d5aaee5e8e42e1a050e46c3d681b74173646a"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-09-03 09:38:14.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-09-06 08:35:20.000000000","tz":0},"subject":"WIP: Move to ansible-core and collections","message":"WIP: Move to ansible-core and collections\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/807546\nChange-Id: I55327cc3c686eae8f2da85ea957cae851e3041e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fc3506a4a549b221f15c73139c2466650cf1bca4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fc3506a4a549b221f15c73139c2466650cf1bca4"}]},"branch":"refs/heads/master","description":"Edit commit message"},"acb95be8f28ad3295d5728fe28669e7a775267b7":{"kind":"REWORK","_number":4,"created":"2021-09-06 10:24:16.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/807279/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/807279/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/807279/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/807279/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/807279/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/807279/4"}}},"commit":{"parents":[{"commit":"d03d5aaee5e8e42e1a050e46c3d681b74173646a","subject":"[doc] Ubuntu has Ceph Pacific in Wallaby+","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d03d5aaee5e8e42e1a050e46c3d681b74173646a"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-09-03 09:38:14.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-09-06 10:23:56.000000000","tz":0},"subject":"WIP: Move to ansible-core and collections","message":"WIP: Move to ansible-core and collections\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/807546\nChange-Id: I55327cc3c686eae8f2da85ea957cae851e3041e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/acb95be8f28ad3295d5728fe28669e7a775267b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/acb95be8f28ad3295d5728fe28669e7a775267b7"}]},"branch":"refs/heads/master"},"03b2c74a17b3f2ac8aed99d59017ca24ec164aa3":{"kind":"REWORK","_number":5,"created":"2021-09-06 11:26:49.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/807279/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/807279/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/807279/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/807279/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/807279/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/807279/5"}}},"commit":{"parents":[{"commit":"d03d5aaee5e8e42e1a050e46c3d681b74173646a","subject":"[doc] Ubuntu has Ceph Pacific in Wallaby+","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d03d5aaee5e8e42e1a050e46c3d681b74173646a"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-09-03 09:38:14.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-09-06 11:26:45.000000000","tz":0},"subject":"WIP: Move to ansible-core and collections","message":"WIP: Move to ansible-core and collections\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/807546\nChange-Id: I55327cc3c686eae8f2da85ea957cae851e3041e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/03b2c74a17b3f2ac8aed99d59017ca24ec164aa3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/03b2c74a17b3f2ac8aed99d59017ca24ec164aa3"}]},"branch":"refs/heads/master"},"9c683f3ab5574dfa911344c42b9f56f44595149c":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2021-09-06 11:45:29.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/807279/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/807279/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/807279/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/807279/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/807279/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/807279/6"}}},"commit":{"parents":[{"commit":"d03d5aaee5e8e42e1a050e46c3d681b74173646a","subject":"[doc] Ubuntu has Ceph Pacific in Wallaby+","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d03d5aaee5e8e42e1a050e46c3d681b74173646a"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-09-03 09:38:14.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-09-06 11:45:26.000000000","tz":0},"subject":"WIP: Move to ansible-core and collections","message":"WIP: Move to ansible-core and collections\n\nChange-Id: I55327cc3c686eae8f2da85ea957cae851e3041e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9c683f3ab5574dfa911344c42b9f56f44595149c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9c683f3ab5574dfa911344c42b9f56f44595149c"}]},"branch":"refs/heads/master","description":"Edit commit message"},"8da2f513f0cce5dfb0660c8d50b601d1f91a24d4":{"kind":"REWORK","_number":7,"created":"2021-09-06 12:44:26.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/807279/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/807279/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/807279/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/807279/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/807279/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/807279/7"}}},"commit":{"parents":[{"commit":"d03d5aaee5e8e42e1a050e46c3d681b74173646a","subject":"[doc] Ubuntu has Ceph Pacific in Wallaby+","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d03d5aaee5e8e42e1a050e46c3d681b74173646a"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-09-03 09:38:14.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-09-06 12:44:13.000000000","tz":0},"subject":"WIP: Move to ansible-core and collections","message":"WIP: Move to ansible-core and collections\n\nChange-Id: I55327cc3c686eae8f2da85ea957cae851e3041e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8da2f513f0cce5dfb0660c8d50b601d1f91a24d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8da2f513f0cce5dfb0660c8d50b601d1f91a24d4"}]},"branch":"refs/heads/master"},"70a6c1b16ff97b834787327af618996b5ae80c89":{"kind":"REWORK","_number":8,"created":"2021-09-06 13:25:23.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/807279/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/807279/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/807279/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/807279/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/807279/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/807279/8"}}},"commit":{"parents":[{"commit":"d03d5aaee5e8e42e1a050e46c3d681b74173646a","subject":"[doc] Ubuntu has Ceph Pacific in Wallaby+","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d03d5aaee5e8e42e1a050e46c3d681b74173646a"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-09-03 09:38:14.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-09-06 13:22:47.000000000","tz":0},"subject":"WIP: Move to ansible-core and collections","message":"WIP: Move to ansible-core and collections\n\nChange-Id: I55327cc3c686eae8f2da85ea957cae851e3041e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/70a6c1b16ff97b834787327af618996b5ae80c89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/70a6c1b16ff97b834787327af618996b5ae80c89"}]},"branch":"refs/heads/master"},"f4f297f6daf00996204c925ab57ed5c8f92932f1":{"kind":"REWORK","_number":9,"created":"2021-09-06 14:06:55.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/807279/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/807279/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/807279/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/807279/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/807279/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/807279/9"}}},"commit":{"parents":[{"commit":"d03d5aaee5e8e42e1a050e46c3d681b74173646a","subject":"[doc] Ubuntu has Ceph Pacific in Wallaby+","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d03d5aaee5e8e42e1a050e46c3d681b74173646a"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-09-03 09:38:14.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-09-06 14:06:50.000000000","tz":0},"subject":"WIP: Move to ansible-core and collections","message":"WIP: Move to ansible-core and collections\n\nChange-Id: I55327cc3c686eae8f2da85ea957cae851e3041e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f4f297f6daf00996204c925ab57ed5c8f92932f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f4f297f6daf00996204c925ab57ed5c8f92932f1"}]},"branch":"refs/heads/master"},"9345edaec6cf6343f0dd287024c43b3e9e70b9fb":{"kind":"REWORK","_number":10,"created":"2021-09-07 07:50:46.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/807279/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/807279/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/807279/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/807279/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/807279/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/807279/10"}}},"commit":{"parents":[{"commit":"d03d5aaee5e8e42e1a050e46c3d681b74173646a","subject":"[doc] Ubuntu has Ceph Pacific in Wallaby+","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d03d5aaee5e8e42e1a050e46c3d681b74173646a"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-09-03 09:38:14.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-09-07 07:50:42.000000000","tz":0},"subject":"WIP: Move to ansible-core and collections","message":"WIP: Move to ansible-core and collections\n\nChange-Id: I55327cc3c686eae8f2da85ea957cae851e3041e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9345edaec6cf6343f0dd287024c43b3e9e70b9fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9345edaec6cf6343f0dd287024c43b3e9e70b9fb"}]},"branch":"refs/heads/master"},"ec9801cddff7883c01deae34068e29079d9eb6c7":{"kind":"REWORK","_number":11,"created":"2021-09-07 07:51:43.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/807279/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/807279/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/807279/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/807279/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/807279/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/807279/11"}}},"commit":{"parents":[{"commit":"d03d5aaee5e8e42e1a050e46c3d681b74173646a","subject":"[doc] Ubuntu has Ceph Pacific in Wallaby+","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d03d5aaee5e8e42e1a050e46c3d681b74173646a"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-09-03 09:38:14.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-09-07 07:51:40.000000000","tz":0},"subject":"WIP: Move to ansible-core and collections","message":"WIP: Move to ansible-core and collections\n\nChange-Id: I55327cc3c686eae8f2da85ea957cae851e3041e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ec9801cddff7883c01deae34068e29079d9eb6c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ec9801cddff7883c01deae34068e29079d9eb6c7"}]},"branch":"refs/heads/master"},"dc4778fa5429e1ea08eafc0662011c67e8821d32":{"kind":"REWORK","_number":12,"created":"2021-09-07 07:52:38.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/807279/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/807279/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/807279/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/807279/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/807279/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/807279/12"}}},"commit":{"parents":[{"commit":"d03d5aaee5e8e42e1a050e46c3d681b74173646a","subject":"[doc] Ubuntu has Ceph Pacific in Wallaby+","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d03d5aaee5e8e42e1a050e46c3d681b74173646a"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-09-03 09:38:14.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-09-07 07:52:35.000000000","tz":0},"subject":"WIP: Move to ansible-core and collections","message":"WIP: Move to ansible-core and collections\n\nChange-Id: I55327cc3c686eae8f2da85ea957cae851e3041e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dc4778fa5429e1ea08eafc0662011c67e8821d32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dc4778fa5429e1ea08eafc0662011c67e8821d32"}]},"branch":"refs/heads/master"},"000142e374eada751cd631fe811ad4722d22f33c":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2021-09-07 07:52:54.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/807279/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/807279/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/807279/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/807279/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/807279/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/807279/13"}}},"commit":{"parents":[{"commit":"1ec83fde74815413260b09a2b1c50b99dbc9fc24","subject":"toolbox: Move custom Ansible config to global location","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1ec83fde74815413260b09a2b1c50b99dbc9fc24"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-09-03 09:38:14.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-09-07 07:52:54.000000000","tz":0},"subject":"WIP: Move to ansible-core and collections","message":"WIP: Move to ansible-core and collections\n\nChange-Id: I55327cc3c686eae8f2da85ea957cae851e3041e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/000142e374eada751cd631fe811ad4722d22f33c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/000142e374eada751cd631fe811ad4722d22f33c"}]},"branch":"refs/heads/master","description":"Rebase"},"1ba9ee7b8437aa7a56c866ffb827c9f9cd222927":{"kind":"REWORK","_number":14,"created":"2021-09-07 07:53:34.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/807279/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/807279/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/807279/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/807279/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/807279/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/807279/14"}}},"commit":{"parents":[{"commit":"1ec83fde74815413260b09a2b1c50b99dbc9fc24","subject":"toolbox: Move custom Ansible config to global location","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1ec83fde74815413260b09a2b1c50b99dbc9fc24"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-09-03 09:38:14.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-09-07 07:53:28.000000000","tz":0},"subject":"WIP: Move to ansible-core and collections","message":"WIP: Move to ansible-core and collections\n\nChange-Id: I55327cc3c686eae8f2da85ea957cae851e3041e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1ba9ee7b8437aa7a56c866ffb827c9f9cd222927"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1ba9ee7b8437aa7a56c866ffb827c9f9cd222927"}]},"branch":"refs/heads/master"},"05b4660803070cf732b7888a9f7496d7705cea38":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2021-09-08 06:29:37.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/807279/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/807279/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/807279/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/807279/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/807279/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/807279/15"}}},"commit":{"parents":[{"commit":"88f1e32aa997b2b8c3627cd4b2128402262fdef4","subject":"toolbox: Move custom Ansible config to global location","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/88f1e32aa997b2b8c3627cd4b2128402262fdef4"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-09-03 09:38:14.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-09-08 06:29:37.000000000","tz":0},"subject":"WIP: Move to ansible-core and collections","message":"WIP: Move to ansible-core and collections\n\nChange-Id: I55327cc3c686eae8f2da85ea957cae851e3041e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/05b4660803070cf732b7888a9f7496d7705cea38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/05b4660803070cf732b7888a9f7496d7705cea38"}]},"branch":"refs/heads/master","description":"Rebase"},"2b07cebb91b08c26e0f505c73f602d3d937abaee":{"kind":"REWORK","_number":16,"created":"2021-09-08 08:47:42.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/807279/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/807279/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/807279/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/807279/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/807279/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/807279/16"}}},"commit":{"parents":[{"commit":"88f1e32aa997b2b8c3627cd4b2128402262fdef4","subject":"toolbox: Move custom Ansible config to global location","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/88f1e32aa997b2b8c3627cd4b2128402262fdef4"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-09-03 09:38:14.000000000","tz":120},"committer":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-09-08 08:34:16.000000000","tz":120},"subject":"toolbox: Move to ansible-core 2.11 and collections","message":"toolbox: Move to ansible-core 2.11 and collections\n\nChange-Id: I55327cc3c686eae8f2da85ea957cae851e3041e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2b07cebb91b08c26e0f505c73f602d3d937abaee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2b07cebb91b08c26e0f505c73f602d3d937abaee"}]},"branch":"refs/heads/master"},"6eeb2cbdd106ec92f99b500acd8aaeea23278216":{"kind":"REWORK","_number":17,"created":"2021-09-16 10:19:57.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/807279/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/807279/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/807279/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/807279/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/807279/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/807279/17"}}},"commit":{"parents":[{"commit":"88f1e32aa997b2b8c3627cd4b2128402262fdef4","subject":"toolbox: Move custom Ansible config to global location","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/88f1e32aa997b2b8c3627cd4b2128402262fdef4"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-09-03 09:38:14.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-09-16 10:19:54.000000000","tz":0},"subject":"toolbox: Move to ansible-core 2.11 and collections","message":"toolbox: Move to ansible-core 2.11 and collections\n\nChange-Id: I55327cc3c686eae8f2da85ea957cae851e3041e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6eeb2cbdd106ec92f99b500acd8aaeea23278216"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6eeb2cbdd106ec92f99b500acd8aaeea23278216"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
