)]}'
{"id":"openstack%2Fkolla~807382","triplet_id":"openstack%2Fkolla~stable%2Fwallaby~Id5938697b62ba040062475fcf5215d64e1ba8c6f","project":"openstack/kolla","branch":"stable/wallaby","topic":"bug/1930867-debian-stable/wallaby","hashtags":[],"change_id":"Id5938697b62ba040062475fcf5215d64e1ba8c6f","subject":"Pin td-agent to 4.0.* also on Debian","status":"MERGED","created":"2021-09-06 09:30:48.000000000","updated":"2021-09-07 08:29:08.000000000","submitted":"2021-09-07 08:26:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"807382-bug/1930867-debian-stable/wallaby","cherry_pick_of_change":807475,"cherry_pick_of_patch_set":1,"meta_rev_id":"f2c814e04e2ca04925834c11b2775c3507290a0b","_number":807382,"virtual_id_number":807382,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-07 08:26:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-09-07 06:56:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-09-07 06:55:17.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-09-07 06:56:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-06 10:48:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-07 06:56:52.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"61a4c9a7c8049b7dd95e0e4c62b84954abceb07f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-06 09:30:48.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"5f65efc744a8146223e775643ab5dfe4be7f35ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-06 10:48:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/9554cffe101843ce88c8fa78d98cf252 : SUCCESS in 20m 43s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/117acab93b124b1188b173c3a81a7018 : SUCCESS in 36m 46s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/7921c2eabc874f11bf2e445e44598d0b : SUCCESS in 1h 00m 15s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/911823fb1ad34f8d939b751c7ce0cc8c : SUCCESS in 1h 02m 22s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/48df4587eece4d259376bb665ef465f1 : SUCCESS in 1h 13m 56s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/d606cab019ec45b6a27ceef5f41d0162 : SUCCESS in 30m 38s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/216235fa18be4f24b5c730c675435935 : SUCCESS in 17m 45s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/95f57660ab6e4d59a966bd047afce464 : SUCCESS in 52m 42s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/001bbbf7127d4495a33b36fa7cdd6fb3 : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9402aad41bf34f9ca1bc775f60e16d22 : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/426674831f914e2cbc2cf0ed6b32f69c : SUCCESS in 4m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/411587cfae4f4969b8470973cf5fece8 : SUCCESS in 4m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a339340e93d84a1aa83685f9f37ae08a : SUCCESS in 6m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3adcec00c44d4a9a87d3cf1fcc56bf10 : SUCCESS in 4m 31s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cf191f17182a453094a06d910aee3dfe : SUCCESS in 20m 15s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/89b58dbc5ab5470ab378fc22c31247f3 : SUCCESS in 32m 13s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/59e2e9f8cb104becb30b3248e442f220 : SUCCESS in 49m 01s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/da68b8423e4a402388e1c1f49499f0fd : SUCCESS in 47m 49s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/bdd4797d470d4f65844078b175d45b4e : SUCCESS in 57m 59s","accounts_in_message":[],"_revision_number":1},{"id":"aec1284ad5507384333195e28ddee9c3fa6d18cf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-06 11:53:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/b422149bdfd0448886dd6a3795377820 : SUCCESS in 2h 19m 45s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e82b981cbb894f0784503e0d6448afa4 : SUCCESS in 41m 19s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d5444ab61aee499aa8690dcfaa6c1a66 : SUCCESS in 1h 21m 29s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/2e4b05783cb14af9a59f9861f889efaf : SUCCESS in 40m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"341ef600c3ab600578fa1359b18576c2d4250c63","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-07 06:55:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ee9964e0e2203b50f1d775cf55738201ca5682fc","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-09-07 06:56:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a6264e82524060d02da2765956b768d674dffb28","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-07 06:57:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e05bbf3ae53132643904485a6d0e176d43a8d401","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-07 08:26:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/4645fbaff8564441bd2d5a742c1f630d : SUCCESS in 41m 46s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/bfd72474a7b94b299672538ac5552c8d : SUCCESS in 1h 10m 55s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/39f1c7e6639c461cb3fd5441e2680261 : SUCCESS in 1h 27m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa66945f0a1749ceaf6b0e8ec48a0e8c : SUCCESS in 7m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1066e6ff4c924fed8583c3d2153844eb : SUCCESS in 6m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e6adb73f708b4d709d8318441b0b7d1f : SUCCESS in 6m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4dca26fefc7e45818f81c3bc11323cec : SUCCESS in 9m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9e8d102d7d964c7e94bada0d468e73e0 : SUCCESS in 5m 53s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3abd372d340e4f11a1542fc0aefc39f0 : SUCCESS in 30m 36s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a1e05be2650341e69003a5de6e5a2048 : SUCCESS in 48m 57s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7fef8f70ba1e45599a3557b895b5e472 : SUCCESS in 1h 07m 27s","accounts_in_message":[],"_revision_number":1},{"id":"3c1a9af842134f4579ca5d6bdf5e21b36d888878","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-07 08:26:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f2c814e04e2ca04925834c11b2775c3507290a0b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-07 08:29:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab5d6412672f49e4be4a17c075f38ece : SUCCESS in 1m 17s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7e78e7c685ab43c88635d573c7a7c8e7 : SUCCESS in 1m 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1ec1d8b467f8eca44182f2e0e366a3245b2dd1db","revisions":{"1ec1d8b467f8eca44182f2e0e366a3245b2dd1db":{"kind":"REWORK","_number":1,"created":"2021-09-06 09:30:48.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/82/807382/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/82/807382/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/807382/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/807382/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/807382/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/82/807382/1"}}},"commit":{"parents":[{"commit":"333f02a0725adfeb000e2fe99b2c07749c9186ad","subject":"Allow build debian/ubuntu-binary cloudkitty images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/333f02a0725adfeb000e2fe99b2c07749c9186ad"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-09-05 08:50:52.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-09-06 09:30:48.000000000","tz":0},"subject":"Pin td-agent to 4.0.* also on Debian","message":"Pin td-agent to 4.0.* also on Debian\n\nDebian was missed in the original patch. [1]\n\nBackport to Victoria.\n\n[1] 6d0882ef799bb531d007c3e6839ef318fd512d5c\naka Iefcdd3100b7e3c5320bc5f1286a18251bdeab885\n\nCloses-Bug: #1930867\nChange-Id: Id5938697b62ba040062475fcf5215d64e1ba8c6f\n(cherry picked from commit 5253e031c6e62e89b03a3756d44c58da007f1cbd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1ec1d8b467f8eca44182f2e0e366a3245b2dd1db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1ec1d8b467f8eca44182f2e0e366a3245b2dd1db"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
