)]}'
{"id":"openstack%2Fkolla~807387","triplet_id":"openstack%2Fkolla~stable%2Fwallaby~Ic5d90459b6b95b69c7df7382ca74e1420f4cdb35","project":"openstack/kolla","branch":"stable/wallaby","hashtags":[],"change_id":"Ic5d90459b6b95b69c7df7382ca74e1420f4cdb35","subject":"debian: Change influxdb, rabbitmq and td-agent repos to bullseye","status":"MERGED","created":"2021-09-06 11:36:25.000000000","updated":"2021-09-07 07:46:04.000000000","submitted":"2021-09-07 07:44:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"807387","cherry_pick_of_change":807536,"cherry_pick_of_patch_set":1,"meta_rev_id":"86a50f1533b9ddd418c6a5fd83eeae7e236f39da","_number":807387,"virtual_id_number":807387,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-07 07:44:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-09-07 06:28:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-09-06 12:07:40.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-09-07 06:28:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-06 12:07:40.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2021-09-06 12:57:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4f39b0237cb1fc0857bbba8e91482183e9001cd2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-09-06 11:36:25.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* kolla/template/repos.yaml","accounts_in_message":[],"_revision_number":1},{"id":"953dfeba51ff509caaba4b3585c54c68c8d427fe","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-09-06 11:38:11.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"3d6253ca78c557a284a7a85343ea4b3500e09081","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-09-06 11:38:21.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"b0b2a35daa16e72d4df14a9db464ab1b16f0e375","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-06 12:07:40.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"1ebf0ed33908cf3e8054584f56b09c1a60871e15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-06 12:57:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/a16911b450d348f1936b7fc60d419b13 : SUCCESS in 32m 01s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/8b17acf388404fdfa5db790e4abcb03b : SUCCESS in 39m 26s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/af9123b199b54c3fbe3550618e60ddbb : SUCCESS in 1h 06m 44s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/0674aac4f7e34196afa7eec0fb54cbe6 : SUCCESS in 1h 05m 31s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/409953f45d6248aabe663bd8e46152a4 : SUCCESS in 1h 17m 53s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/cbfaaceeb82e4587a8fe1064d4cc3dcb : SUCCESS in 29m 31s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/89f6ce4c18004b0b91f1055df04c3979 : SUCCESS in 15m 53s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/5bd134fa31c74fb498704145f1e8456a : SUCCESS in 1h 04m 12s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2d8ff02c47141ed858f3d8fb233d6e8 : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f0473e656b2643d8ace4893e4ac6fc12 : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/13519da2b05e45c18ac6c503d02377cd : SUCCESS in 6m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a9816faa282f46e4aa560645997f8ceb : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7dc0ba3361044603b35079201e520949 : SUCCESS in 8m 38s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/11d7f26257694cdab236e83b6181bd14 : SUCCESS in 19m 28s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/b5da84cba3274b6eb7fb5cf916212a3b : SUCCESS in 29m 21s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e002630bbb5e4ee9a6907e93bde297f5 : SUCCESS in 54m 44s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/53b22819055840d59fc4e93d9371d63c : SUCCESS in 44m 52s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7a7ba33e17214cb8898d5aa8ce80d4c8 : SUCCESS in 1h 02m 12s","accounts_in_message":[],"_revision_number":2},{"id":"6961e1ed6c0999f309aecd185b1c4f3ffa19eeda","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-06 13:56:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/6d4ea97ad6be44cd817fd593758ef2d7 : SUCCESS in 1h 28m 12s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1f1d415bbf4b4be78b49e1bcc0c5487f : SUCCESS in 1h 15m 45s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/fee7faf8b41a45c4b140f88d73acdd4d : SUCCESS in 2h 12m 10s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/fc13872663304b049b569db48259d43d : SUCCESS in 1h 27m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"78821f0253c95082bb5e4e5b4a2ad2d9198dc3af","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-09-07 06:28:01.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b8a3484ffae277ec585c9a0a367d2cba22725618","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-07 06:28:22.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a18ff04d2d1bd6e08394fa2c3f262cabee0c23bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-07 07:44:14.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/b90eb8338a174f728a4d73a551125179 : SUCCESS in 37m 51s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/8fad27d2aa5449a8a5e0dc539ec6eb56 : SUCCESS in 1h 03m 06s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/9e02e4fbf1ba4519bfddc76620814e5d : SUCCESS in 1h 12m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d368914291734be4a2c24a95043a10c4 : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bbef96ef9db8439895827ad6d4f4f78b : SUCCESS in 6m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/795fd495254642da9a4d8ceba5419d26 : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/907d732f68124391891826edbcc2251d : SUCCESS in 7m 07s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/2e072ed4945c4f6ca1bc497ab6391fb2 : SUCCESS in 41m 38s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/512caa5be5b54c1b853fe59f118f4e5f : SUCCESS in 56m 17s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/69bbc591987441f397a0598fa6e659b0 : SUCCESS in 1h 05m 57s","accounts_in_message":[],"_revision_number":2},{"id":"63d68e8f333b5a0aff744d3e60bbd784e1e68104","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-07 07:44:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"86a50f1533b9ddd418c6a5fd83eeae7e236f39da","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-07 07:46:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec7e89d96a554835afd275ba09a5529a : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f0ffe4050fb4dfd276234b7c208ecb66f1e95e49","revisions":{"250f98a9cf8992ee7a34f6a88e723c3d52daf7d5":{"kind":"REWORK","_number":1,"created":"2021-09-06 11:36:25.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/87/807387/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/87/807387/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/807387/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/807387/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/807387/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/87/807387/1"}}},"commit":{"parents":[{"commit":"af76e670375d3208c3b2391eee0ff18706eef46c","subject":"[doc] Ubuntu has Ceph Pacific in Wallaby+","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/af76e670375d3208c3b2391eee0ff18706eef46c"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-09-06 07:48:17.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-09-06 11:36:25.000000000","tz":0},"subject":"debian: Change influxdb, rabbitmq and td-agent repos to bullseye","message":"debian: Change influxdb, rabbitmq and td-agent repos to bullseye\n\nChange-Id: Ic5d90459b6b95b69c7df7382ca74e1420f4cdb35\n(cherry picked from commit 006c19fdbfba881fb1fb98fbb429a5053cbbd1a3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/250f98a9cf8992ee7a34f6a88e723c3d52daf7d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/250f98a9cf8992ee7a34f6a88e723c3d52daf7d5"}]},"branch":"refs/heads/stable/wallaby"},"f0ffe4050fb4dfd276234b7c208ecb66f1e95e49":{"kind":"REWORK","_number":2,"created":"2021-09-06 11:38:11.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/87/807387/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/87/807387/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/807387/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/807387/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/87/807387/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/87/807387/2"}}},"commit":{"parents":[{"commit":"af76e670375d3208c3b2391eee0ff18706eef46c","subject":"[doc] Ubuntu has Ceph Pacific in Wallaby+","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/af76e670375d3208c3b2391eee0ff18706eef46c"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-09-06 07:48:17.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-09-06 11:38:07.000000000","tz":0},"subject":"debian: Change influxdb, rabbitmq and td-agent repos to bullseye","message":"debian: Change influxdb, rabbitmq and td-agent repos to bullseye\n\nChange-Id: Ic5d90459b6b95b69c7df7382ca74e1420f4cdb35\n(cherry picked from commit 006c19fdbfba881fb1fb98fbb429a5053cbbd1a3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f0ffe4050fb4dfd276234b7c208ecb66f1e95e49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f0ffe4050fb4dfd276234b7c208ecb66f1e95e49"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
