)]}'
{"id":"openstack%2Fkolla~807475","triplet_id":"openstack%2Fkolla~master~Id5938697b62ba040062475fcf5215d64e1ba8c6f","project":"openstack/kolla","branch":"master","topic":"bug/1930867-debian","hashtags":[],"change_id":"Id5938697b62ba040062475fcf5215d64e1ba8c6f","subject":"Pin td-agent to 4.0.* also on Debian","status":"MERGED","created":"2021-09-05 08:52:23.000000000","updated":"2021-09-06 09:22:55.000000000","submitted":"2021-09-06 09:21:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"807475-bug/1930867-debian","meta_rev_id":"d97160558944c6320313ed945468c3cd03379cbd","_number":807475,"virtual_id_number":807475,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-06 09:21:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-09-06 07:29:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-09-06 08:06:12.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2021-09-06 08:06:12.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2021-09-05 08:52:46.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-05 08:52:46.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2021-09-05 08:52:46.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-09-05 08:52:46.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-09-05 10:17:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8271bc43e9580e59a45fa7bd761b2fbfc7cc4669","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-05 08:52:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b9b75119bc6935c0eac71e8a444f386f3db9a3d6","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-05 08:52:46.000000000","message":"Patch Set 1: Backport-Candidate+2","accounts_in_message":[],"_revision_number":1},{"id":"4b93b1cba2821eb20842f67ae8700d50f2054ddc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-05 10:17:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/d6bdee81ab5f4dcf8b0cc762e5aba155 : SUCCESS in 27m 15s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/6daae0f5946d40edb8514d0b72c52258 : SUCCESS in 40m 37s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/1afa8c1d303646f1a19fd713ea9b7bef : SUCCESS in 1h 09m 55s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/138bf65e4f84414898b372e5ff207c34 : SUCCESS in 1h 05m 50s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/d377347e21794456add832dde4edb8dc : SUCCESS in 1h 20m 08s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b2e2b428c61e4e9e8bc8af48c04a61b5 : SUCCESS in 28m 35s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/22711d41d5e74e348396ddf6a159dfc1 : SUCCESS in 19m 24s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/657dcd4578f7460bb87391280de48acb : SUCCESS in 1h 01m 15s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6abf2f21b5544b3b285f9cc1a2328a0 : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/30d69a022f4c4743a4b9b5712bd61e9c : SUCCESS in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/634991c8b38e4650ac6fbfde62cc8061 : SUCCESS in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/25b8a466943948dbb84603e670b3ec5b : SUCCESS in 4m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6738b077544b43dca58076c3d08c9f6c : SUCCESS in 7m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a39997548742440493ff5aafd902d9fd : SUCCESS in 4m 08s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ec5e3c4737ee40dfb377874b8ca8f32c : SUCCESS in 22m 38s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4e85853d302a42208ff17b16d8f17d47 : SUCCESS in 30m 17s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7751ffdc009f42b18d00a91df8487270 : SUCCESS in 46m 30s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/18374e4b90ea4b1fb53104b86039d2b7 : SUCCESS in 51m 07s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2b7513c2ed804cc5a6ce4d374a19caa7 : SUCCESS in 1h 03m 05s","accounts_in_message":[],"_revision_number":1},{"id":"5232c8d7eea7a3a1488c7f63c17e141f377bab82","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-05 10:42:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/e16986a79aad406db6200f20d0f430e8 : SUCCESS in 45m 37s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ebfa786863ad4f1f8ea8407ef770f2a7 : SUCCESS in 1h 31m 30s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5e41a8d8d45249a7840311eb5280e258 : SUCCESS in 1h 47m 45s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a5a26e20265d4fa4ae3cb04466e009d7 : SUCCESS in 6m 01s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f674a37e46974bdb97338e73a4da26a3 : SUCCESS in 6m 22s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/d78a28ce942040818cef030b6f37e70d : SUCCESS in 1h 00m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"81f017ed83e3a8ead0719852cfef56fda34dd242","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-09-06 07:29:23.000000000","message":"Patch Set 1: Code-Review+2\n\nI guess we had that issue so it\u0027s good to have it fixed in Debian as well, but since then we have like td-agent 4.2.0 - are we sure the issue persists? is there any bug raised anywhere? It can\u0027t be persistently pinned ;-)","accounts_in_message":[],"_revision_number":1},{"id":"84faa8008d1aaf38e54bbede4bc291592ee50ca9","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-06 07:42:31.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1: Code-Review+2\n\u003e \n\u003e I guess we had that issue so it\u0027s good to have it fixed in Debian as well, but since then we have like td-agent 4.2.0 - are we sure the issue persists? is there any bug raised anywhere? It can\u0027t be persistently pinned ;-)\n\nAFAIU, Debian still fails with 4.2.0","accounts_in_message":[],"_revision_number":1},{"id":"a0539d4589dfca2961282912fb5dab7c92d78d88","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-09-06 08:06:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8319255d94b29ce498ef1547fb875fc9425e2555","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-06 08:06:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"653a7449ded1342477de5ea10302785e3cd04811","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-06 09:21:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/aab54e45b1fb486f96a9f17c0fd1456b : SUCCESS in 43m 28s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/50d7eb6d3ead4182bd34ebf774f0eb94 : SUCCESS in 1h 03m 04s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/c739e35b944a407c9331e197bc574c81 : SUCCESS in 1h 13m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1d860ee915b4c64a85f3b497740f3ac : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0672bc0c8d044104975230028f76498a : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a9a1d2d8530a483986a8907d53a0b7bf : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f80ba52ab6b24cd9b2fdd7f0f932811e : SUCCESS in 7m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/623afc9b55eb4e93b172d76c4d489cba : SUCCESS in 5m 00s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/3e59e5dcd1a343d8bb93c3f1efa2037f : SUCCESS in 32m 40s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3693f552f49746c4970e8935c8405f8b : SUCCESS in 47m 04s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4c49de72032d452fa695a47eac315827 : SUCCESS in 57m 14s","accounts_in_message":[],"_revision_number":1},{"id":"36b1f598c13a0164ab1ce9825a9189190dda2222","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-06 09:21:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d97160558944c6320313ed945468c3cd03379cbd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-06 09:22:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62e55e156a534b289d690cc5c3bc5c47 : SUCCESS in 1m 22s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/37897522bdfa4ddeb0f665a12bf5618f : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5253e031c6e62e89b03a3756d44c58da007f1cbd","revisions":{"5253e031c6e62e89b03a3756d44c58da007f1cbd":{"kind":"REWORK","_number":1,"created":"2021-09-05 08:52:23.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/75/807475/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/75/807475/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/807475/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/807475/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/807475/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/75/807475/1"}}},"commit":{"parents":[{"commit":"d03d5aaee5e8e42e1a050e46c3d681b74173646a","subject":"[doc] Ubuntu has Ceph Pacific in Wallaby+","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d03d5aaee5e8e42e1a050e46c3d681b74173646a"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-09-05 08:50:52.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-09-05 08:50:52.000000000","tz":0},"subject":"Pin td-agent to 4.0.* also on Debian","message":"Pin td-agent to 4.0.* also on Debian\n\nDebian was missed in the original patch. [1]\n\nBackport to Victoria.\n\n[1] 6d0882ef799bb531d007c3e6839ef318fd512d5c\naka Iefcdd3100b7e3c5320bc5f1286a18251bdeab885\n\nCloses-Bug: #1930867\nChange-Id: Id5938697b62ba040062475fcf5215d64e1ba8c6f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5253e031c6e62e89b03a3756d44c58da007f1cbd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5253e031c6e62e89b03a3756d44c58da007f1cbd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
