)]}'
{"id":"openstack%2Fkolla~807536","triplet_id":"openstack%2Fkolla~master~Ic5d90459b6b95b69c7df7382ca74e1420f4cdb35","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"Ic5d90459b6b95b69c7df7382ca74e1420f4cdb35","subject":"debian: Change influxdb, rabbitmq and td-agent repos to bullseye","status":"MERGED","created":"2021-09-06 07:49:03.000000000","updated":"2021-09-06 10:30:49.000000000","submitted":"2021-09-06 10:28:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"807536","meta_rev_id":"930ee1da119991bb0ac8101cd2f38535b9fa8521","_number":807536,"virtual_id_number":807536,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-06 10:28:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-09-06 09:11:26.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2021-09-06 08:12:01.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2021-09-06 09:11:26.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2021-09-06 08:12:07.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-06 08:12:01.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2021-09-06 09:03:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-06 09:11:26.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"6f09ed2d27a4a3faf2a021fd02b5a873787ccd1e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-09-06 07:49:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"495a3d4ddd3064923a4637b792589a3486d07422","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-06 08:12:01.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"86e702aa516d1645ed32072cf1a00b70824a4258","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-06 08:12:07.000000000","message":"Patch Set 1: Backport-Candidate+2","accounts_in_message":[],"_revision_number":1},{"id":"ef8c932830ee300f351fa7e8619d7cf4cdd8de33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-06 09:03:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/48b44c011bd2465aa30d4faaa1be8ae3 : SUCCESS in 22m 14s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/b763ed8e9db94da0947906ac4186c7ad : SUCCESS in 32m 57s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/baad559257d94b53b5f179c5b613bb45 : SUCCESS in 1h 08m 40s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/c694ae46d09d4de4be53dddc34c13c6d : SUCCESS in 1h 01m 46s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/198c2b7066ba4871986ef924dd3063c4 : SUCCESS in 1h 09m 35s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/edc2a8042bf34ec8b73835fedd180c21 : SUCCESS in 28m 19s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/250cf7fcb6af44c68d601446a3c38e3d : SUCCESS in 17m 08s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c7463d970a944b38bf16c304e57f23a3 : SUCCESS in 1h 00m 24s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f5bfef481324701992a593cdc0e90ca : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b90cb640704743f1ae36fa8155ee2ca2 : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1b1343c64de74cd6989f81baa709d55b : SUCCESS in 4m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6ba5d8373ca448268208a3cc9e450113 : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6394b439d9574afdaca50965460a91e8 : SUCCESS in 7m 09s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/06be8e7595684af991d5761f681c07e4 : SUCCESS in 18m 33s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/c44805e84ea043c7b088cc3893788835 : SUCCESS in 30m 51s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ca1297b0149648f5bdf0a8a62876cdb7 : SUCCESS in 51m 32s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/782a468c410540338178a17928af673d : SUCCESS in 50m 49s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7d52b19ad6f54f79b67cfaaeec2a899a : SUCCESS in 57m 54s","accounts_in_message":[],"_revision_number":1},{"id":"59eef5a3008fb12f7f659a06cfea477c487665aa","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-09-06 09:11:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b1a797c343aa6d9a70c0b308c98079746f775d94","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-06 09:11:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cbc3e7346e257d68462730eb4920441e4efe36f5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-06 10:15:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/6d91c67de24c442db1e860c4a1918e6c : SUCCESS in 2h 02m 27s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2852c8df13bd4ecfb4d6db8367ea706a : SUCCESS in 46m 22s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7bc46e235b294e0088cd3056188d4f96 : SUCCESS in 2h 24m 34s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4e472f75a4aa425b97b5eecf1cf5a0eb : SUCCESS in 4m 56s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8184c15b8d8f4297a6d4022d60d9520c : SUCCESS in 7m 28s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/78a19e35df394205b850d0a22f285793 : SUCCESS in 47m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c619c18944b29a7b9914d807753752fc21ed013e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-06 10:28:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/d75dfc5088894efbaa6ce95a8fb632d1 : SUCCESS in 38m 20s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/a4be7557c7b24a0da4ee9a22b622f656 : SUCCESS in 1h 09m 30s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/0f2ee224e967427ca488745745343d20 : SUCCESS in 1h 15m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e80317a54344480a3233ecb4295de42 : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f6fc67b44954d2c8163f73cfb759af8 : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4fdc01064dd44765a496b107903b558f : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d99ae74b29cf487f85b8bbe7017e80b3 : SUCCESS in 7m 30s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/01e910a718664c17bc78c115d18912f5 : SUCCESS in 31m 49s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4dbf9b3200d548f9a97cab19ad0b48ac : SUCCESS in 51m 36s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/383c21cf8fa24f42a02c719015e77f9a : SUCCESS in 1h 01m 47s","accounts_in_message":[],"_revision_number":1},{"id":"a1050a613ba9ffc316e30df545911c1f4a423b97","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-06 10:28:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"930ee1da119991bb0ac8101cd2f38535b9fa8521","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-06 10:30:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f3dc4cb9dfc4c81a27d70f3b0d3a75c : SUCCESS in 1m 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"006c19fdbfba881fb1fb98fbb429a5053cbbd1a3","revisions":{"006c19fdbfba881fb1fb98fbb429a5053cbbd1a3":{"kind":"REWORK","_number":1,"created":"2021-09-06 07:49:03.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/36/807536/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/36/807536/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/807536/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/807536/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/36/807536/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/36/807536/1"}}},"commit":{"parents":[{"commit":"d03d5aaee5e8e42e1a050e46c3d681b74173646a","subject":"[doc] Ubuntu has Ceph Pacific in Wallaby+","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d03d5aaee5e8e42e1a050e46c3d681b74173646a"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-09-06 07:48:17.000000000","tz":120},"committer":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-09-06 07:49:00.000000000","tz":120},"subject":"debian: Change influxdb, rabbitmq and td-agent repos to bullseye","message":"debian: Change influxdb, rabbitmq and td-agent repos to bullseye\n\nChange-Id: Ic5d90459b6b95b69c7df7382ca74e1420f4cdb35\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/006c19fdbfba881fb1fb98fbb429a5053cbbd1a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/006c19fdbfba881fb1fb98fbb429a5053cbbd1a3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
