)]}'
{"id":"openstack%2Fkolla~807681","triplet_id":"openstack%2Fkolla~master~I4325fe031647a04766516afa528639d6ff0b5ae6","project":"openstack/kolla","branch":"master","topic":"bug-1942846","hashtags":[],"change_id":"I4325fe031647a04766516afa528639d6ff0b5ae6","subject":"toolbox: Move custom Ansible config to global location","status":"MERGED","created":"2021-09-07 07:49:38.000000000","updated":"2021-09-10 19:50:13.000000000","submitted":"2021-09-10 19:48:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"807681-bug-1942846","meta_rev_id":"96ec4f00c615515bbebfaa67aa54a2e7bb5b1127","_number":807681,"virtual_id_number":807681,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-10 19:48:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-09-08 12:27:55.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2021-09-09 14:47:47.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2021-09-09 14:47:47.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2021-09-09 14:47:47.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-07 07:55:32.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"CC"},{"updated":"2021-09-07 08:33:16.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2021-09-07 09:06:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-09-07 09:19:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-08 12:27:55.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-09-09 14:47:47.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"dcdafe86ab272255deb30ff768db643f03bd6843","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-09-07 07:49:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"247bed2a9c01db8caa972e287ed67a56130ffdd1","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-07 07:55:32.000000000","message":"Patch Set 1:\n\nreno, reno?","accounts_in_message":[],"_revision_number":1},{"id":"8c643f77eb96275f8a58b421c2a2758ba4a78d61","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-09-07 07:58:03.000000000","message":"Topic set to bug-1942846","accounts_in_message":[],"_revision_number":1},{"id":"ef0c404a57c86246e57641dffc541949e6704b02","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-09-07 08:00:46.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"17698a3987d2efc6dc3c711b85d7a5b85ca7bbc7","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-09-07 08:33:16.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"10df0d9c0a009e1d00dfc3a58ac04e4167d930f8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-07 09:06:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/efe8c2fd74ac4eba94b177807bbf1f1f : SUCCESS in 59m 18s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5b7adfd4517f474d8a2352d1ee7f0f45 : SUCCESS in 39m 35s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7f57b07208d9412da64120ded5b4170b : FAILURE in 49m 23s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a301cf7a11bb4647a839ff8fcddff9c4 : SUCCESS in 7m 31s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1e3b8053504448768c483302776d3e89 : SUCCESS in 6m 10s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/b39f00d638be4fa7bbfa8df8a149e9fc : SUCCESS in 41m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"774dbe2725e9706348376e8033c8e74689857730","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-07 09:19:15.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/f2983cf559d34b769bbc0cef60cb9db8 : SUCCESS in 42m 18s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/54a4d5e908bf420d8b635fd0f02eaa69 : SUCCESS in 47m 22s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/df95cf7811654578b0e305fb50c589fd : FAILURE in 50m 23s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/cf7a68ad7ea847cdab355605ebb3919b : FAILURE in 44m 24s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/a68f8dc6dcb14235b6482fd8ad4f5010 : SUCCESS in 1h 14m 54s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/974c9aff79d84008bbd11aa89c17aa31 : SUCCESS in 45m 00s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/d6a02cd248d043269d5dac50dd79f312 : SUCCESS in 23m 25s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/24fdb171b01f42afb7ff62f1facd622f : FAILURE in 35m 25s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9d61928995f4b5190a312bfdcaf2339 : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b7bfdbefa1504db69f7463c7444143b3 : SUCCESS in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ec74494175ab4f669237d9e15ab918d6 : SUCCESS in 5m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5c8b0b95bf7f4e08b6b1e6803aeedb3c : SUCCESS in 6m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e49fda5eb8b49439c536e4baaa06b34 : SUCCESS in 8m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7a0eec7afa9c4090b8b3431739f7b79b : SUCCESS in 5m 55s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/49d6f164adc1426f8174812b810a656d : SUCCESS in 35m 35s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/10c6e3fcac184c4aba3b753cf6bf59d6 : SUCCESS in 33m 27s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0559221693f9460aa15cbbf9ecee465c : FAILURE in 31m 22s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1d11873069c640e7bf1b3349a748480c : FAILURE in 34m 34s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6371bb0d92f5493594ab6cf632e94825 : SUCCESS in 1h 00m 40s","accounts_in_message":[],"_revision_number":2},{"id":"b7e11d28aaec6be0017585f59b5d4709c6a83406","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-09-07 12:09:40.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d37ebe3d3ae2394bf800887e63d2d8ada9bcbb5e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-09-07 12:11:46.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"2d4d793a3184082340fa92878398da50b17454a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-07 12:12:38.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"1bd32476474919689476be8a8d8fc0c8263469bb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-07 12:12:42.000000000","message":"Patch Set 3:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"c546ce4a04568d7a3b0328ac8595eabf1263aacf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-09-07 12:19:38.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"f63e4a3010bae86ee720a705a3f1864bd1dd636e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-07 13:38:10.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/9caae0deefb34756b9cf6cf9d10c4d32 : SUCCESS in 1h 05m 41s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/df6a59233e5b453b8dd08385b16b3bec : SUCCESS in 1h 14m 41s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c8c55b0dc72d4408937fa440260adde9 : FAILURE in 55m 58s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/74c38860b99345bba4fddffe0f5b61e8 : SUCCESS in 5m 25s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/19e158994cd34d369527d1478806b6e2 : SUCCESS in 6m 04s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/dbbc413f22f143ae9c5aae7ab922c8e6 : SUCCESS in 48m 57s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3af8b637ea87578d88d77a392087c79854d74607","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-07 13:41:19.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/d5ba774d77654515b88be3466879d39b : SUCCESS in 31m 36s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/76fd10ba7aad477da180f3e7dff639f6 : SUCCESS in 37m 20s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/58d1af10887048e89cddf81fb41277bd : FAILURE in 46m 28s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/07f909fe9dad40d1bf5168519f916965 : FAILURE in 39m 31s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/a89801ea54c947629e63f26507999609 : SUCCESS in 1h 21m 08s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/c24addfe358b45c583b53f1d62deb0e7 : SUCCESS in 29m 24s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/4fa4c5d7b6264fd5ab4eba67314c5107 : SUCCESS in 18m 48s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7c28e56317a94e2cb9a6d2c3db6a04db : FAILURE in 39m 59s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d5e4d04b9844c7d89c32b36421a2640 : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bfbfd431093c4cd38abc32f80ae7bbf8 : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2b9dc9ec8d054ef29823b75a21251f83 : SUCCESS in 4m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/270edd5df1b845a3a37798cef28f6676 : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55d8c954c9014257870a84f892e15c81 : SUCCESS in 5m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3f081c1865194ba19f8622b57c4adb10 : SUCCESS in 4m 51s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6977f2c5aee94097968b0b2ecd80e9e8 : SUCCESS in 19m 30s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/42bd9cce601e43aa8690267bac4778e9 : SUCCESS in 24m 36s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/6cf3387a4acd484dbdce8c53fbe76299 : FAILURE in 30m 42s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/40a29905d98e4354bfce8118f528c488 : FAILURE in 31m 00s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/77c22d6a18e444cd80c864236225a314 : SUCCESS in 57m 21s","accounts_in_message":[],"_revision_number":4},{"id":"b0e18a64450c4a00f3a2d260419b6cd5891d7229","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-09-07 13:48:41.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"221062f7307d00a7065901e444ea66bed313dfe4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-07 14:51:32.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/6baae079d2e040a2a3c261195ac18df3 : SUCCESS in 56m 42s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4937c95b77d84f5892aab584e7feeed1 : SUCCESS in 45m 05s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/47d188cb717d41b69b9ef2989d41ef9f : FAILURE in 48m 55s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/42e2429e09a54ddfa75af120bf46b482 : SUCCESS in 7m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/da1f8f5026884fa6a363687b7be64d7f : SUCCESS in 5m 44s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/e21556ac75ac42e29997b14afb5a40d1 : SUCCESS in 1h 00m 39s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"e3714e76d4241d52e3435789fd94ff19df65efc6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-07 15:11:16.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/2e7a081c4063448dbe057d13c459ae5d : SUCCESS in 30m 42s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/77054d9c718148118845c1ed7407c01c : SUCCESS in 41m 17s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/9d318aa5f7e148ed80e3f2da9ab54716 : FAILURE in 34m 59s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/d0d398216cb440578081b19eac2cb53b : FAILURE in 33m 02s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/dc0c2c1ef7bb488cbb51f17ab084eec0 : SUCCESS in 1h 21m 00s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/ed8bd70d3cff433e8ad41aa20f9557a2 : SUCCESS in 35m 49s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/62ec9ed4d5a548348f7ccd4f7a965fb4 : SUCCESS in 23m 09s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/94b1719c063e4a2c9906f22e0ba87bd8 : FAILURE in 28m 25s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8905e1388f264bfa8fc47ad896245f68 : FAILURE in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/826eede8379f4cfeadc5a97059a88171 : SUCCESS in 5m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a2f51ca21fe246f682e8a704aa542a12 : SUCCESS in 5m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3f83d19e9a4241a8a91109d7af6f8fed : SUCCESS in 4m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62eecee79e354a03ac7bbe647980adee : SUCCESS in 8m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f6827d8a275e45bdab7ef1d263ce51b1 : SUCCESS in 5m 11s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/612f670bb8de4727a30c0110f5096d40 : SUCCESS in 26m 04s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/8c5c8090e9be4b3b9fae2fdca173de42 : SUCCESS in 26m 25s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7b3d9c0761b14f79a2456c7daf9a4653 : FAILURE in 33m 28s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/65734c0d8bce428bbdfac0ef81491e26 : FAILURE in 23m 57s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/cbf02e6b43fc4b07a7723c2b0e420ebe : SUCCESS in 1h 02m 00s","accounts_in_message":[],"_revision_number":5},{"id":"14fbaa3a17bd8c919d3cbb3c93b4f1339516846b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-09-07 16:00:40.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"454ca7762e6d19b1b8f6c9300eff81927acd6836","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-09-07 16:45:58.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"60e96ebd90791ab4b2642fd2a99fbccad2e548a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-09-07 18:00:51.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"16487c5574d6ee9e009ccd86046a0a1ca7015a53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-07 19:24:57.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/4c826e1860534c088f75a6b6c661b5e0 : SUCCESS in 23m 09s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/e282ecd3b4ab4950ac19492651b051b0 : SUCCESS in 41m 21s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/a4bbc4a949194b8c96afe1c33b7ea849 : SUCCESS in 1h 13m 45s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/8702734b8f744453bd87181240cad8e0 : SUCCESS in 59m 43s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/a0f69474a3394d1ca4592a9d9dc34ba8 : SUCCESS in 1h 18m 02s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/4e5496d5187849ea9a4cd00c7dff9587 : SUCCESS in 29m 47s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/85d6250d698e4b1bab72d269ff9bf1fe : SUCCESS in 19m 47s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/93764dd7236f4c24800533e290ff52f8 : SUCCESS in 1h 06m 19s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0d7830bd8504b86800db97996080859 : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/173c00ae11064c7fb220289d0a4d790f : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/50d5ff9d43fa4a1ab401938804ae9121 : SUCCESS in 4m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dd19947cc7e24d1ea05d2071065331f6 : SUCCESS in 5m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e8eb9e7db5d49ce9133ac59fce9615f : SUCCESS in 7m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/887e886883f348c0ad1bc38c416f66b5 : SUCCESS in 4m 55s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/469a64f93b7346b185528e9ac13ac310 : SUCCESS in 29m 06s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/0e4b3f2994f64df1a97033fa5f3134a9 : SUCCESS in 32m 05s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/973a532e226e4cfb9e0265653d4bbd1a : SUCCESS in 54m 06s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c55fb1d91efe4b51bb1b3c6fb5f0cafe : SUCCESS in 58m 23s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/10f756cf2e0a4607b2b9b42ed9419cdf : SUCCESS in 1h 07m 34s","accounts_in_message":[],"_revision_number":8},{"id":"a04840d0b08b023d094e9a7598f5c96c4b347bf9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-07 19:40:08.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/959eadb37a7342c39efabea47d8925ce : SUCCESS in 1h 36m 43s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7f64458a2dc2417cbcfc1844e65035e0 : SUCCESS in 1h 13m 08s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0b5c58ff603848cebd5efd450e3ccbf8 : FAILURE in 35m 09s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9f36ec2004a24eb0b9f20bd02da84b64 : SUCCESS in 4m 49s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5b1dd387d17f4d4a8352fdcdc3ebc327 : SUCCESS in 4m 48s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/9b6e802aa77c4586acbc26046ee0c358 : SUCCESS in 35m 14s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"e7ff6c724aa413f8d78108f113d2ec78dcf69d9c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-09-08 09:01:14.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"ba839a66a23c397dc6aaa81103878c1625805b6f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-09-08 09:08:11.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"62913504e166eaea65c2750c5d9768a2a466c714","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-09-08 12:27:55.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"bf8035d7c04e3e7acea1c0b7861d050cf70066cc","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-09-08 12:28:21.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"d2c8cd04a0321b8cff1ee5c06e5d3b795650a509","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-09-08 13:26:02.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"d320eb44a64dfd17e1eaf36994e76e136b411143","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-09 14:47:47.000000000","message":"Patch Set 8: Backport-Candidate+2 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"530209f1b162a7261f72407938e6e9ef2fc32806","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-09 14:52:02.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"73cac0ab5516bebf2db05be526bd7642be9219db","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-09 15:55:30.000000000","message":"Patch Set 8: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/9e157c58f44b45e7948b13f5de9d3f96 : SUCCESS in 39m 44s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/02661b47118a4c4894ce6219e05285b3 : FAILURE in 35m 55s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/5b527a7ce4c44c108739969f69f5eb4e : FAILURE in 11m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed96df0050cb462097ca6fe7e54a2aec : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/93aac4cc2f4a48449eb3ecbd7935f33c : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ad58a1c045b4410baa099071a250c6a9 : SUCCESS in 6m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a906cd2a7fb74357b7dfda4e5c09024a : SUCCESS in 7m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1f37648b0b014f119184416a3ba36fe4 : SUCCESS in 4m 48s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/4bdbe0d3bd5146b485e262f1156e0c73 : SUCCESS in 30m 25s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f2bb4792b8bd447a92c0b77cd2cb0398 : FAILURE in 26m 43s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/60c8746c951e4a17a9c1021c925e7bd2 : FAILURE in 9m 18s","accounts_in_message":[],"_revision_number":8},{"id":"eb6b286dbbf93921c33bc22c037abfb6c7fbd627","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-09 15:59:05.000000000","message":"Patch Set 8:\n\nCI broken by https://review.opendev.org/c/openstack/kolla-ansible/+/805449","accounts_in_message":[],"_revision_number":8},{"id":"32a24af70e91e2ed0f55a626e3b0017bd9a9acec","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-09 15:59:45.000000000","message":"Patch Set 8:\n\nneed to adapt it","accounts_in_message":[],"_revision_number":8},{"id":"1a3a904d2d6acedde9177cf1734b6d550f1d745f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-09 16:45:14.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/ed86322a575d4df8a7fcb5709ae53538 : SUCCESS in 27m 07s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/005357dec3e541d79cedd5f527d3e43d : SUCCESS in 40m 49s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/6bf1c5c67ca148c78ee317f3999f74a0 : FAILURE in 32m 14s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/7193a61d0933430db7320e2b479869b3 : FAILURE in 33m 43s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/0c0d8c85e39149a29ae9b8f957ce57cc : FAILURE in 10m 52s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/b1a064eee3174fe0a438d30d18c4f09b : SUCCESS in 36m 44s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/35b6909d0dec40968b7e0926aeb941ea : SUCCESS in 24m 12s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e5a37ba01fbd43d3a3762286f70120c9 : FAILURE in 29m 21s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd2742b295a74de29f6f5b9463ecfd20 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f8e6134c4f954ebca2d3d6b43e57bc6c : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6576dff34426431d90b05150df775f36 : SUCCESS in 6m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8def58c3861b460199ebe107349282c8 : SUCCESS in 5m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fb1a7d38da5408ebe4a088dea81be25 : SUCCESS in 6m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/99504c678d034881ae28b1d15ad82661 : SUCCESS in 4m 29s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3271a62b061542798ca5f042a5d154fa : SUCCESS in 20m 24s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e7fd6b5fac5445aabe7d1379fd5248ce : SUCCESS in 31m 26s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/919ea652c6ef47f995d8f49ede10203a : FAILURE in 19m 40s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1496d8a8d11d4921950c7118da8efb7c : FAILURE in 22m 56s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a9094af798434df5b02ac73a324748b5 : FAILURE in 9m 41s","accounts_in_message":[],"_revision_number":8},{"id":"ae3c308698f87d59542cc75de36f464abeadf01c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-09-10 08:38:59.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"c89fbf8fd5bd931167ad8f3fa785d60377dd2cca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 09:54:28.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/4d706e723b9c40d7bdf6c00ce7bdfa05 : SUCCESS in 31m 13s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/d2d121461286494b9a7e13b8c40549cd : SUCCESS in 42m 12s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/4c7035bd05ad4816b7ac111d41e63dcd : SUCCESS in 1h 09m 39s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/d0d8b5592a4041e2aef68a8e6e24ca00 : SUCCESS in 1h 01m 47s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/7a2b9ceeda2b4a9ca050671537e8f248 : FAILURE in 10m 32s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/be54a8ff24f24dc28c803aa1aee2f166 : SUCCESS in 31m 23s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/ab29d37958d84274b3964af18fc1cb74 : SUCCESS in 19m 35s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/1628bfe6dd35467d96d755b0894fed74 : SUCCESS in 59m 36s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aeea7d933043413ca61b5fd1b980255b : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c70d7409339140228f86026c8eeff513 : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ad99aff53eb8477789a72221772064f5 : SUCCESS in 5m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/38c219a0a67847bf83d25b22559a2231 : SUCCESS in 5m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb33363fa6384ed49dabf20401213294 : SUCCESS in 6m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7b23105231484cb5a1ce5e40b6e5c36a : SUCCESS in 6m 19s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7a2aa8f7a5dd497689136901649e6bbc : SUCCESS in 26m 45s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/a8fa4c54226c47ab92f34e2fab8168e8 : SUCCESS in 39m 17s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8ec5b2ba3e264c78a0000d05cf2a5685 : SUCCESS in 51m 22s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/16c80915241c48d28364302ee3889092 : SUCCESS in 51m 14s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7286baa3bca3486485acc39a5be651eb : FAILURE in 9m 33s","accounts_in_message":[],"_revision_number":8},{"id":"0bb7462a4c6cea4bd83f8f25388c3fde07abca20","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 10:30:12.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/399366e86757417eb6919b4c3b3fa573 : SUCCESS in 1h 47m 37s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b8ac33ffe5314673a6133eb11a905a5e : SUCCESS in 1h 34m 17s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c59dc9cd10544af88b572842fe519f94 : SUCCESS in 1h 30m 06s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0175e4630db943feae1f3441437b323d : SUCCESS in 6m 30s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4945398272134cb9b94a15d9768257fd : SUCCESS in 7m 50s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/4de0682c40de487988f83715ad70e206 : SUCCESS in 56m 37s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"309f91c8490d59c41f6b9d456a2eb3880468eed2","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-10 16:57:59.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"a231ace6e57ce092f29fd7657978c103ae8e9916","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 18:31:01.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/22c8316362e34810925e454dce1d1619 : SUCCESS in 24m 02s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/84769ac6b30645bdb15da35fe6917bcb : SUCCESS in 39m 03s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/a500cc95b1784857a56853706400617c : SUCCESS in 59m 15s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/659775a683634d508ab15acfea5ee5c2 : SUCCESS in 1h 10m 20s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/22967f1c293844aeb5afe2ccf03b003b : SUCCESS in 1h 23m 16s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/56eaae1b0f4f4f1fa799b698e6576078 : SUCCESS in 33m 09s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/853dacf571d740e98b184dcdda9e5bc8 : SUCCESS in 20m 35s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/8103339d02324af5be0c97c8cb160799 : SUCCESS in 1h 06m 05s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8bd58a7b6a024938aacfe8fbccac041b : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/27fbd5916ba64af28cf20898f4b5e940 : SUCCESS in 6m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b907104c7c944f9bb31d6758e67f7637 : SUCCESS in 6m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d7ccd1ce78e043c286934df7527ca087 : SUCCESS in 6m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cd245e416ec4b8180f01810d1eb8a90 : SUCCESS in 9m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/83f6c8cb2e404ea297945bda0a5ccb3f : SUCCESS in 6m 03s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8c3ef2cf7f9e436395695cb92018ad43 : SUCCESS in 24m 57s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d9e8d59dd4e944b7880eb0b032772d19 : SUCCESS in 37m 03s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8b1a405b382749f29894a4a6bcd9d177 : SUCCESS in 51m 51s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1296a6b486024537aef9011336a46b7b : SUCCESS in 56m 19s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e0f8874a12f04396ab9c6957629993ee : SUCCESS in 1h 02m 02s","accounts_in_message":[],"_revision_number":8},{"id":"90ceb88fbf2aecd527875d9c863797ab00d304e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 18:31:24.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"448d9185a3a5755f43ac7dc4724b262b4f934e34","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 19:11:16.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/b68b68764e0b456281bc813ff61a3bd5 : SUCCESS in 1h 52m 07s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/dc576b5247fa49a7a40eb51023e2b39e : SUCCESS in 1h 24m 20s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a78c016457234f079cd691d8f2681325 : SUCCESS in 2h 03m 15s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/72c2a0499db84159b1c5e6c924bdfc8d : SUCCESS in 6m 35s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7c6cd9bc03d24ca59d895a266c07e224 : SUCCESS in 8m 05s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/79ebae89ccb0493c866cece9ff1d485f : SUCCESS in 49m 36s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"f6305028db7e10f6e86cd6540867eef85d926a64","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 19:48:48.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/0c2fdda5b3b94f24bf9b1d526675792a : SUCCESS in 37m 01s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/8fc8425dc4f442cfaacd0058b4a8e9c6 : SUCCESS in 1h 09m 20s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/ed9668bfe5a54fd29ac2acf3d98a96e3 : SUCCESS in 1h 13m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c50b26e76e5043d4900377444fa26b1b : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e0691fed256241b6b7e3465436743d48 : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/21417262dd9c4208964d3f2befe4db1f : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6ffdd849b924d75882348959f65e0bf : SUCCESS in 6m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/95fc1e7531214c2f936404eb12831fad : SUCCESS in 5m 05s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ba3439225b9443758b5b856b462963df : SUCCESS in 34m 27s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a659edb5f84c4ad9b09d084e92bf06f8 : SUCCESS in 49m 40s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/05dcb4ac7ad54dbf95ca1249ec12e543 : SUCCESS in 1h 02m 41s","accounts_in_message":[],"_revision_number":8},{"id":"f467002d6f7b026b7654da4d601ccbb6b65ed778","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 19:48:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"96ec4f00c615515bbebfaa67aa54a2e7bb5b1127","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 19:50:13.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f14d8dd29f884fdc8dc0cd7a47eff066 : SUCCESS in 1m 00s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c44fd4b19ad44cfb836f521f6f736497 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"88f1e32aa997b2b8c3627cd4b2128402262fdef4","revisions":{"1ec83fde74815413260b09a2b1c50b99dbc9fc24":{"kind":"REWORK","_number":1,"created":"2021-09-07 07:49:38.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/81/807681/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/807681/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/807681/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/807681/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/807681/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/807681/1"}}},"commit":{"parents":[{"commit":"18b3eea03ee1c23b770808e57036f1ee6f0b7346","subject":"Merge \"debian: Change influxdb, rabbitmq and td-agent repos to bullseye\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/18b3eea03ee1c23b770808e57036f1ee6f0b7346"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-09-07 07:47:41.000000000","tz":120},"committer":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-09-07 07:49:36.000000000","tz":120},"subject":"toolbox: Move custom Ansible config to global location","message":"toolbox: Move custom Ansible config to global location\n\nIn order to achieve logging to /var/log/kolla/ansible.log from different\nusers than default ansible - the config needs to be moved to global location\nand referenced via ANSIBLE_CONFIG variable.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/807679\nCloses-Bug: #1942846\nChange-Id: I4325fe031647a04766516afa528639d6ff0b5ae6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1ec83fde74815413260b09a2b1c50b99dbc9fc24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1ec83fde74815413260b09a2b1c50b99dbc9fc24"}]},"branch":"refs/heads/master"},"6103790d5b03e04e3b14a309903e7ef800c52818":{"kind":"REWORK","_number":2,"created":"2021-09-07 08:00:46.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/81/807681/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/807681/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/807681/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/807681/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/807681/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/807681/2"}}},"commit":{"parents":[{"commit":"18b3eea03ee1c23b770808e57036f1ee6f0b7346","subject":"Merge \"debian: Change influxdb, rabbitmq and td-agent repos to bullseye\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/18b3eea03ee1c23b770808e57036f1ee6f0b7346"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-09-07 07:47:41.000000000","tz":120},"committer":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-09-07 08:00:37.000000000","tz":120},"subject":"toolbox: Move custom Ansible config to global location","message":"toolbox: Move custom Ansible config to global location\n\nIn order to achieve logging to /var/log/kolla/ansible.log from different\nusers than default ansible - the config needs to be moved to global location\nand referenced via ANSIBLE_CONFIG variable.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/807679\nCloses-Bug: #1942846\nChange-Id: I4325fe031647a04766516afa528639d6ff0b5ae6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6103790d5b03e04e3b14a309903e7ef800c52818"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6103790d5b03e04e3b14a309903e7ef800c52818"}]},"branch":"refs/heads/master"},"c24ebb4db06edf5bbb3b4433c6b28b8ad0a5ccb3":{"kind":"REWORK","_number":3,"created":"2021-09-07 12:11:46.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/81/807681/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/807681/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/807681/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/807681/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/807681/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/807681/3"}}},"commit":{"parents":[{"commit":"18b3eea03ee1c23b770808e57036f1ee6f0b7346","subject":"Merge \"debian: Change influxdb, rabbitmq and td-agent repos to bullseye\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/18b3eea03ee1c23b770808e57036f1ee6f0b7346"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-09-07 07:47:41.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-09-07 12:11:41.000000000","tz":0},"subject":"toolbox: Move custom Ansible config to global location","message":"toolbox: Move custom Ansible config to global location\n\nIn order to achieve logging to /var/log/kolla/ansible.log from different\nusers than default ansible - the config needs to be moved to global location\nand referenced via ANSIBLE_CONFIG variable.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/807679\nCloses-Bug: #1942846\nChange-Id: I4325fe031647a04766516afa528639d6ff0b5ae6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c24ebb4db06edf5bbb3b4433c6b28b8ad0a5ccb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c24ebb4db06edf5bbb3b4433c6b28b8ad0a5ccb3"}]},"branch":"refs/heads/master"},"9bfb4b9dc15a9c9d2cbf615ec438595f3fb1dbd4":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2021-09-07 12:19:38.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/81/807681/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/807681/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/807681/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/807681/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/807681/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/807681/4"}}},"commit":{"parents":[{"commit":"18b3eea03ee1c23b770808e57036f1ee6f0b7346","subject":"Merge \"debian: Change influxdb, rabbitmq and td-agent repos to bullseye\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/18b3eea03ee1c23b770808e57036f1ee6f0b7346"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-09-07 07:47:41.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-09-07 12:19:34.000000000","tz":0},"subject":"toolbox: Move custom Ansible config to global location","message":"toolbox: Move custom Ansible config to global location\n\nIn order to achieve logging to /var/log/kolla/ansible.log from different\nusers than default ansible - the config needs to be moved to global\nlocation.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/807679\nCloses-Bug: #1942846\nChange-Id: I4325fe031647a04766516afa528639d6ff0b5ae6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9bfb4b9dc15a9c9d2cbf615ec438595f3fb1dbd4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9bfb4b9dc15a9c9d2cbf615ec438595f3fb1dbd4"}]},"branch":"refs/heads/master","description":"Edit commit message"},"e1c280b9cdd6ed230cc6a434be2e748933404429":{"kind":"REWORK","_number":5,"created":"2021-09-07 13:48:41.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/81/807681/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/807681/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/807681/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/807681/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/807681/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/807681/5"}}},"commit":{"parents":[{"commit":"18b3eea03ee1c23b770808e57036f1ee6f0b7346","subject":"Merge \"debian: Change influxdb, rabbitmq and td-agent repos to bullseye\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/18b3eea03ee1c23b770808e57036f1ee6f0b7346"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-09-07 07:47:41.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-09-07 13:48:33.000000000","tz":0},"subject":"toolbox: Move custom Ansible config to global location","message":"toolbox: Move custom Ansible config to global location\n\nIn order to achieve logging to /var/log/kolla/ansible.log from different\nusers than default ansible - the config needs to be moved to global\nlocation.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/807679\nCloses-Bug: #1942846\nChange-Id: I4325fe031647a04766516afa528639d6ff0b5ae6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e1c280b9cdd6ed230cc6a434be2e748933404429"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e1c280b9cdd6ed230cc6a434be2e748933404429"}]},"branch":"refs/heads/master"},"e96618af779b8f4cbe3872473284e5f6723bb915":{"kind":"REWORK","_number":6,"created":"2021-09-07 16:00:40.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/81/807681/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/807681/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/807681/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/807681/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/807681/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/807681/6"}}},"commit":{"parents":[{"commit":"18b3eea03ee1c23b770808e57036f1ee6f0b7346","subject":"Merge \"debian: Change influxdb, rabbitmq and td-agent repos to bullseye\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/18b3eea03ee1c23b770808e57036f1ee6f0b7346"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-09-07 07:47:41.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-09-07 16:00:37.000000000","tz":0},"subject":"toolbox: Move custom Ansible config to global location","message":"toolbox: Move custom Ansible config to global location\n\nIn order to achieve logging to /var/log/kolla/ansible.log from different\nusers than default ansible - the config needs to be moved to global\nlocation.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/807679\nCloses-Bug: #1942846\nChange-Id: I4325fe031647a04766516afa528639d6ff0b5ae6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e96618af779b8f4cbe3872473284e5f6723bb915"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e96618af779b8f4cbe3872473284e5f6723bb915"}]},"branch":"refs/heads/master"},"e900421ab5bb9c57850f7723340dbcde9f3adafb":{"kind":"REWORK","_number":7,"created":"2021-09-07 16:45:58.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/81/807681/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/807681/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/807681/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/807681/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/807681/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/807681/7"}}},"commit":{"parents":[{"commit":"18b3eea03ee1c23b770808e57036f1ee6f0b7346","subject":"Merge \"debian: Change influxdb, rabbitmq and td-agent repos to bullseye\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/18b3eea03ee1c23b770808e57036f1ee6f0b7346"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-09-07 07:47:41.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-09-07 16:45:49.000000000","tz":0},"subject":"toolbox: Move custom Ansible config to global location","message":"toolbox: Move custom Ansible config to global location\n\nIn order to achieve logging to /var/log/kolla/ansible.log from different\nusers than default ansible - the config needs to be moved to global\nlocation.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/807679\nCloses-Bug: #1942846\nChange-Id: I4325fe031647a04766516afa528639d6ff0b5ae6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e900421ab5bb9c57850f7723340dbcde9f3adafb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e900421ab5bb9c57850f7723340dbcde9f3adafb"}]},"branch":"refs/heads/master"},"88f1e32aa997b2b8c3627cd4b2128402262fdef4":{"kind":"REWORK","_number":8,"created":"2021-09-07 18:00:51.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/81/807681/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/807681/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/807681/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/807681/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/807681/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/807681/8"}}},"commit":{"parents":[{"commit":"18b3eea03ee1c23b770808e57036f1ee6f0b7346","subject":"Merge \"debian: Change influxdb, rabbitmq and td-agent repos to bullseye\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/18b3eea03ee1c23b770808e57036f1ee6f0b7346"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-09-07 07:47:41.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-09-07 18:00:46.000000000","tz":0},"subject":"toolbox: Move custom Ansible config to global location","message":"toolbox: Move custom Ansible config to global location\n\nIn order to achieve logging to /var/log/kolla/ansible.log from different\nusers than default ansible - the config needs to be moved to global\nlocation.\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/807679\nCloses-Bug: #1942846\nChange-Id: I4325fe031647a04766516afa528639d6ff0b5ae6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/88f1e32aa997b2b8c3627cd4b2128402262fdef4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/88f1e32aa997b2b8c3627cd4b2128402262fdef4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
