)]}'
{"id":"openstack%2Fkolla~810016","triplet_id":"openstack%2Fkolla~master~Iee451db823a15b40f9e893c1febfd910c336ddec","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2021-11-01 10:41:21.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2021-11-01 09:26:26.000000000","reason":"Mark Goddard replied on the change"},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2021-10-11 13:01:14.000000000","reason":"Radosław Piliszek replied on the change"}},"hashtags":[],"change_id":"Iee451db823a15b40f9e893c1febfd910c336ddec","subject":"cinder-volume: Install binary python libs only in binary","status":"MERGED","created":"2021-09-20 13:43:23.000000000","updated":"2021-11-01 10:42:42.000000000","submitted":"2021-11-01 10:41:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":12,"has_review_started":true,"submission_id":"810016","meta_rev_id":"fbe95a9d6ea545ae4649aa25d373563ebb7eb3e0","_number":810016,"virtual_id_number":810016,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-01 10:41:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2021-11-01 09:26:26.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2021-10-28 08:59:35.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2021-11-01 09:26:26.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-20 13:45:29.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2021-09-20 15:09:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-21 06:20:20.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2021-10-06 13:19:48.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"dfde970a5c0f38828c62713b4a5914b5901500a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-09-20 13:43:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"16b92fbefe67d3d64fc0f4fab6d49850a2a0be00","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-09-20 13:45:29.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b6e2285fa6eb063f3178b9162db39e7d652704f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 15:09:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/563b4f183af24e469733df3c5dcba014 : SUCCESS in 21m 25s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/1402dcf2deda4c6a87d9ef1a94e374d4 : SUCCESS in 38m 36s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/3421bd3f29ec4fb58c565903db00653e : SUCCESS in 1h 19m 03s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/449fe98463f44ee6ac293d5de05b6ec2 : SUCCESS in 1h 00m 18s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/77eea23e717f40d1814703d7a2d5ce55 : SUCCESS in 1h 12m 14s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/9021e684db8f4a829d11290a47ba837a : SUCCESS in 28m 40s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/9fcabd0839ff45e8906fc4f7e3ec695a : SUCCESS in 18m 44s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/4b8b198127af4f159ca4eb2e99d48875 : SUCCESS in 58m 55s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8fa0f78869844c3946f55a6da31881f : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/22287e95879444fca16d9e684a383e31 : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/79d2e11bcff64c97ae91ab3ec1418024 : SUCCESS in 4m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/de35e0c6ddaf4c5a922f0195f84f740e : SUCCESS in 4m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c941e6c3e0f04139a4c59f4296cdc96e : SUCCESS in 7m 17s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/fbc1b6c9f4d14eac92aab62fd288a825 : SUCCESS in 18m 36s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/6cdf19c6bd8d4c77b46b95e8a81277bb : SUCCESS in 42m 44s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f5da93bbc4ba4bdfa53304dd07810643 : SUCCESS in 50m 35s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f74a15657d604e39a0895b504566c2b9 : SUCCESS in 52m 38s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f9f3faf0692446ad921d84a4bd727eec : SUCCESS in 58m 26s","accounts_in_message":[],"_revision_number":1},{"id":"26a65f22f7dcec6e0beb17f41d1d62f979069132","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-20 16:20:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/516a60d112ca45ebaeb529918eea2b42 : SUCCESS in 2h 19m 02s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/55ce71da4e904bd1b7b14e7d4d614396 : SUCCESS in 1h 55m 51s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ed817ae8ba4c42219ea1195b91d2a8c5 : SUCCESS in 2h 32m 51s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e57e3827ab05459eb3a1734971fc26b4 : SUCCESS in 7m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8eb13928887f4f92adfa4be798bab201 : SUCCESS in 7m 35s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/6e265d22b41540c0b30574067b5e1412 : SUCCESS in 1h 49m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"158579ebf4c15f6d760106b429c9fbb8e73ac802","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-21 06:20:20.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"ff2d6fcd37ba410e3cf528f1e3c0ab3af55b8390","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-10-01 08:38:09.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"2562f718d4e17b336aa5ef1ef07242e06a9d1f5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-01 10:07:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/94e74d5a096645a6b04b2f4a63409bf7 : SUCCESS in 40m 55s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/cc17b20d3f63471b8cfdfc7ef2eb81c8 : SUCCESS in 30m 08s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/c4081deacd5c48bd88a499fda81c18fa : SUCCESS in 1h 12m 09s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/7de93aaa783e47058ac1c890cb70e6dd : SUCCESS in 1h 04m 45s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/d02e2d6283884456a4c4ec6e9980d681 : SUCCESS in 1h 27m 13s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/49ca4b88bcd4440eb7d9648eb8bb236e : SUCCESS in 30m 42s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/c09901b6c9f14530b02f8aac16ce08e1 : SUCCESS in 19m 17s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f356805a4c6b4a189fa6cb354081f5e1 : SUCCESS in 59m 52s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2129c1a2883a4d9e8c9c548ee0561dd3 : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/63faa13a051649d4ab1e38ebe68380d7 : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4a35c56e772e40c08df573e69f8d172a : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/50fa1c6b2831411f9322503df8131a55 : SUCCESS in 4m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c56b49479be409b939d96df2fe61c5e : SUCCESS in 6m 18s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/43ca046db5d442d6bc833e4b14b46cb6 : SUCCESS in 18m 41s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/da584df43798401d80894aee77ee6120 : SUCCESS in 23m 32s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/76704a3130534cc5ad3d14f383e73114 : SUCCESS in 48m 15s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9d4a08f91f6c4639be6002b67b1a312a : SUCCESS in 51m 36s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b12327dbcdc64f98ab73417de3ab2206 : SUCCESS in 1h 06m 23s","accounts_in_message":[],"_revision_number":2},{"id":"a4f4a1cda63667ef504cdae9aece17d0515b4171","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-01 11:13:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/1037b6327aad4075bcf390894298c7d4 : SUCCESS in 1h 51m 34s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a66dd434632c49618951dc8493eb6b60 : SUCCESS in 1h 40m 52s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/386c5481b9684b4cad5dcf6b2ffe0568 : SUCCESS in 2h 16m 47s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ff1d7f25914241ad9d47392b3d6efc4a : SUCCESS in 6m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e2864d73a4fc429c9a386c277347a7b5 : SUCCESS in 6m 18s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/47e4a1e40bb74ed4a0375bdb2507f560 : SUCCESS in 1h 47m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c656c8170d7334a9be6fcb760999d09f435de4f6","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-04 15:04:26.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)\n\nI have figured we don\u0027t need cinderlib at all. That\u0027s the reason why it was not installed in deb-based distros - it\u0027s not used. Cinderlib is actually an implementation for users wishing to use cinder functionality without OpenStack. 😊","accounts_in_message":[],"_revision_number":2},{"id":"1c4dc2b5228ac97bf70f0f1781776bde7c338451","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-10-05 08:51:38.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"03dfaf1dca7886f2c5822328870ad445dd880d17","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-05 08:54:06.000000000","message":"Patch Set 3: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"e4af26282505ad4b7a9a76f1eb970191662bd2bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-10-05 09:07:59.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"d51d571aac4bdf77409c13c7a7cfad8d6b9842a0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-05 09:18:43.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"cbad6fe9a2536f686074738579f7fcacd9e2b869","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-05 09:18:57.000000000","message":"Patch Set 4:\n\nI\u0027ll drop for you","accounts_in_message":[],"_revision_number":4},{"id":"f4e7bb56b9815a2cb161e68df6430fdfa7db1146","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-05 09:19:05.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"5a4378549154b902977d2b2f54d587afb7c79bf6","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-05 09:19:14.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"6594972007cf859871528848438633d3d4f1b04d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-05 09:20:46.000000000","message":"Patch Set 5:\n\nso, all in all, of used libs we only installed rtslib which does not change much hence had close to no effect whatsoever","accounts_in_message":[],"_revision_number":5},{"id":"e4285545b54b4fecab8fa6c34b849b4f3e68a42d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-05 10:49:53.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/4229713e18b94fdf9a77976708d8d44c : SUCCESS in 25m 11s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/122b8dda1a8c4236b36a0f9879f527be : SUCCESS in 39m 52s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/8fd1c7634a894fec90a7cbac5c66b0d5 : SUCCESS in 1h 05m 39s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/cb327f63dc6840318271d34798a5b9e2 : SUCCESS in 1h 05m 53s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/f2d3415bcef54d31ba3f00e7960ec80e : SUCCESS in 1h 25m 16s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/a5829f58d0514c19975fd99a9575d8ce : SUCCESS in 28m 41s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/0a27591190d64459b087159bd286d0c6 : SUCCESS in 18m 57s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/52b929650f4946049b3b3a9aba2a93cb : SUCCESS in 1h 01m 20s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f2ada36644741dab77a9e5fcdd88dab : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/728133e6de384eb3bfacd0af15d99241 : SUCCESS in 6m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f5705b45405742909e0b8ca6cc5ef5d4 : SUCCESS in 5m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/76db29c571214e8888a322043b64942f : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acfaf36763964e4a96108f898bcf7960 : SUCCESS in 7m 32s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/02fdffbd394440ba803c41b4a9bd8dd5 : SUCCESS in 22m 07s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/ddb4b3ebfe3d4e398d7b01e7aa9e697a : SUCCESS in 41m 49s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/05ec21cb70d4435487c1e95670e53548 : SUCCESS in 58m 11s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/12c6f4299d46456d9765260d8749449f : SUCCESS in 59m 10s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5dfffa438b794bd49a05cf94ddee6556 : SUCCESS in 1h 04m 33s","accounts_in_message":[],"_revision_number":5},{"id":"eaadab0262ef4b0b7a2761b523f77081ebd5d4a1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-05 13:30:16.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/9727124b45ac47df94a38a276d38ad09 : SUCCESS in 2h 20m 03s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d75d0bdc8ffe45d28132ac934e0051fe : SUCCESS in 1h 42m 55s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3dfbd9e4b3d94f0f9e68881ea5d787f2 : SUCCESS in 2h 52m 32s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/525d616a210640e6ac8b0a220acde6e7 : SUCCESS in 6m 07s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7ba4533330f5443eab2fdf8aa7f78bb1 : SUCCESS in 6m 23s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/137d90ace02e45ca9e00aaa2fddeb42c : SUCCESS in 1h 49m 33s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a2b07e25053e0eba020ec7320d321bc043c9f6de","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-10-11 09:27:38.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"f9a0188052521da83601456ac9a26f325fba60d4","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-10-11 09:31:21.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b40604de764703c366bb9a9f78b16811374bafd9","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-11 13:01:14.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e0fc91637b92f205ad669a379f738f2477bcd7e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-10-28 06:12:23.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"062df1deb83207eb1ce17a36cf02995ffd4180a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-28 07:47:22.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/48eccbfad5714969b6f0bf3e92b355a3 : SUCCESS in 27m 02s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/e04178b274d646ab89eb05037fd970ea : SUCCESS in 34m 25s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/4c73b29b0d814c07a2c57e41e1678fe8 : SUCCESS in 1h 29m 04s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/aa937bb5141d485485dc5a981a5349e1 : SUCCESS in 1h 25m 43s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/d3cef4e96df54afdab11481a2e07f47a : SUCCESS in 1h 30m 33s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/0dab580abac045b9bb7aa936b149ad2b : SUCCESS in 23m 33s\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/02999559342b4f2b98f5762a261d1314 : SUCCESS in 18m 34s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ec1b92e00a3240268ab1270b10e075d2 : SUCCESS in 56m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20a12b00d99747b68b4da8102402bd9d : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cc13bf1c2c0740019f91009d65b0925c : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/abab1ef69af44491806a10a8e9cbd643 : SUCCESS in 4m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f3ac2dea2af044f2a7378e37ba570a43 : SUCCESS in 4m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c88c22e4d0ed4501b90fc941200b52b9 : SUCCESS in 5m 58s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/dd1552cba1ba4d1aa13993a2bfab7098 : SUCCESS in 21m 59s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/d940b06659184209b258c3ff102bb4f7 : SUCCESS in 28m 08s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d4373d7717aa4bfea36c0510b5a3032f : SUCCESS in 48m 55s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6d89cfc7e32c4d6ab03abff7ab3f9082 : SUCCESS in 49m 25s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1ddf21ba9ffd4e77b3fbc059c12a8472 : SUCCESS in 1h 02m 11s","accounts_in_message":[],"_revision_number":6},{"id":"a1410674cf42116be862d2bb90b2a9a8589a4187","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-28 07:53:30.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/9c11d00ca1e148edaed7f9d5dd2c5480 : SUCCESS in 59m 01s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c774c0afe680462083fe26e8b3685bb0 : SUCCESS in 49m 45s\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/23ea091ebd244382937981d00b1461aa : SUCCESS in 1h 34m 32s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c47987969a744c2cac302ce455d68ff5 : SUCCESS in 5m 04s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ffe6ceae80784e9d92beb52c9b7d0317 : SUCCESS in 5m 19s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/d44935c92af842328d5c38d85d416ef2 : SUCCESS in 45m 22s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"a4adbacdd118f9f6c3a054e4d0fbe93ad6d78f5d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-10-28 08:59:35.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"209ff957400b788297559ec0d76926ff90a57676","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-11-01 09:26:26.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"2074986f27d3da1f55a15b5aca992837b1a51598","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-01 09:26:49.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"03a8cfa175ec3723c3351bc3cec72d9133c987ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-01 10:41:21.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/e2710438c0b84988b6664f700932b4a1 : SUCCESS in 35m 28s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/48a4464e5b54416b919b6a684a52f9d5 : SUCCESS in 1h 01m 44s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/8a3d7593a72a419498f5dbec88729a31 : SUCCESS in 1h 13m 25s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/bada371b4ed94847bfd99d6fcd2b0ee9 : SUCCESS in 28m 51s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/762e8fa81f484064a5933b9ae3f4ef41 : SUCCESS in 50m 11s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/8d28873c828c4cf9b796b5b959c430d9 : SUCCESS in 1h 02m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25548eca0fe64d7190df4c248c22ce65 : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0691f88d8e884f1495d2bed1924ec80d : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b6306cbc1fac4877ace7971bc525017b : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/151f43af84874908afcf649254a204ab : SUCCESS in 7m 34s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/e7fac5a4a3414ae1836556f177427594 : SUCCESS in 27m 31s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/bad4e7aacbbd434996f09cc26f21cc5d : SUCCESS in 40m 14s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6c3d8f6a9d1948de916283f4ec9ea73d : SUCCESS in 57m 46s","accounts_in_message":[],"_revision_number":6},{"id":"a5ac00fd00a27666daa5ce3285097a960ab0c3f6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-01 10:41:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"fbe95a9d6ea545ae4649aa25d373563ebb7eb3e0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-01 10:42:42.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/daeada8e0964483d8688cbd6661c1638 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"3d55f995a9945bf5e12050f6ba55edcfc17bc158","revisions":{"0538c35027a443f1de3835190e584408863b3b8b":{"kind":"REWORK","_number":1,"created":"2021-09-20 13:43:23.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/810016/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/810016/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/810016/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/810016/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/810016/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/810016/1"}}},"commit":{"parents":[{"commit":"e998831b9bbcb808effb0c92be5568cbcaf6142f","subject":"Merge \"Upgrade prometheus elasticsearch exporter to 1.2.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e998831b9bbcb808effb0c92be5568cbcaf6142f"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-09-20 13:42:14.000000000","tz":120},"committer":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-09-20 13:43:20.000000000","tz":120},"subject":"cinder-volume: Install source cinderlib in source","message":"cinder-volume: Install source cinderlib in source\n\nCurrently we install binary cinderlib package in source\ncinder-volume container images.\n\nChange-Id: Iee451db823a15b40f9e893c1febfd910c336ddec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0538c35027a443f1de3835190e584408863b3b8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0538c35027a443f1de3835190e584408863b3b8b"}]},"branch":"refs/heads/master"},"581570620a9894a2575747861e927adadebeba64":{"kind":"REWORK","_number":2,"created":"2021-10-01 08:38:09.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/810016/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/810016/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/810016/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/810016/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/810016/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/810016/2"}}},"commit":{"parents":[{"commit":"e998831b9bbcb808effb0c92be5568cbcaf6142f","subject":"Merge \"Upgrade prometheus elasticsearch exporter to 1.2.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e998831b9bbcb808effb0c92be5568cbcaf6142f"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-09-20 13:42:14.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-10-01 08:38:00.000000000","tz":0},"subject":"cinder-volume: Install source cinderlib in source","message":"cinder-volume: Install source cinderlib in source\n\nCurrently we install binary cinderlib package in source\ncinder-volume container images.\n\nChange-Id: Iee451db823a15b40f9e893c1febfd910c336ddec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/581570620a9894a2575747861e927adadebeba64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/581570620a9894a2575747861e927adadebeba64"}]},"branch":"refs/heads/master"},"23a498a2f85d3082eedeb76c0080d30ced1bc470":{"kind":"REWORK","_number":3,"created":"2021-10-05 08:51:38.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/810016/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/810016/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/810016/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/810016/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/810016/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/810016/3"}}},"commit":{"parents":[{"commit":"e998831b9bbcb808effb0c92be5568cbcaf6142f","subject":"Merge \"Upgrade prometheus elasticsearch exporter to 1.2.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e998831b9bbcb808effb0c92be5568cbcaf6142f"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-09-20 13:42:14.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-10-05 08:51:35.000000000","tz":0},"subject":"cinder-volume: Install binary libraries only in binary","message":"cinder-volume: Install binary libraries only in binary\n\nChange-Id: Iee451db823a15b40f9e893c1febfd910c336ddec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/23a498a2f85d3082eedeb76c0080d30ced1bc470"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/23a498a2f85d3082eedeb76c0080d30ced1bc470"}]},"branch":"refs/heads/master"},"1e0f8c67e80d1f6e0a4d1255159e23aaaab4d365":{"kind":"REWORK","_number":4,"created":"2021-10-05 09:07:59.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/810016/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/810016/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/810016/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/810016/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/810016/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/810016/4"}}},"commit":{"parents":[{"commit":"e998831b9bbcb808effb0c92be5568cbcaf6142f","subject":"Merge \"Upgrade prometheus elasticsearch exporter to 1.2.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e998831b9bbcb808effb0c92be5568cbcaf6142f"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-09-20 13:42:14.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-10-05 09:07:41.000000000","tz":0},"subject":"cinder-volume: Install binary python libs only in binary","message":"cinder-volume: Install binary python libs only in binary\n\nChange-Id: Iee451db823a15b40f9e893c1febfd910c336ddec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1e0f8c67e80d1f6e0a4d1255159e23aaaab4d365"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1e0f8c67e80d1f6e0a4d1255159e23aaaab4d365"}]},"branch":"refs/heads/master"},"eb2d4ccac636543aaa28c7cbbe7de9bcca968ac2":{"kind":"REWORK","_number":5,"created":"2021-10-05 09:19:05.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/16/810016/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/810016/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/810016/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/810016/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/810016/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/810016/5"}}},"commit":{"parents":[{"commit":"e998831b9bbcb808effb0c92be5568cbcaf6142f","subject":"Merge \"Upgrade prometheus elasticsearch exporter to 1.2.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e998831b9bbcb808effb0c92be5568cbcaf6142f"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-09-20 13:42:14.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-10-05 09:19:03.000000000","tz":0},"subject":"cinder-volume: Install binary python libs only in binary","message":"cinder-volume: Install binary python libs only in binary\n\nChange-Id: Iee451db823a15b40f9e893c1febfd910c336ddec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eb2d4ccac636543aaa28c7cbbe7de9bcca968ac2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eb2d4ccac636543aaa28c7cbbe7de9bcca968ac2"}]},"branch":"refs/heads/master"},"3d55f995a9945bf5e12050f6ba55edcfc17bc158":{"kind":"REWORK","_number":6,"created":"2021-10-28 06:12:23.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/810016/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/810016/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/810016/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/810016/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/810016/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/810016/6"}}},"commit":{"parents":[{"commit":"e998831b9bbcb808effb0c92be5568cbcaf6142f","subject":"Merge \"Upgrade prometheus elasticsearch exporter to 1.2.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e998831b9bbcb808effb0c92be5568cbcaf6142f"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-09-20 13:42:14.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-10-28 06:12:20.000000000","tz":0},"subject":"cinder-volume: Install binary python libs only in binary","message":"cinder-volume: Install binary python libs only in binary\n\nChange-Id: Iee451db823a15b40f9e893c1febfd910c336ddec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3d55f995a9945bf5e12050f6ba55edcfc17bc158"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3d55f995a9945bf5e12050f6ba55edcfc17bc158"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
