)]}'
{"id":"openstack%2Fkolla~810161","triplet_id":"openstack%2Fkolla~master~I9704e5371a410cb1f69ec0367174e2b368961dc1","project":"openstack/kolla","branch":"master","topic":"debian-mariadb-upstream","hashtags":["bugfix-in-stable"],"change_id":"I9704e5371a410cb1f69ec0367174e2b368961dc1","subject":"[debian] Use upstream MariaDB","status":"MERGED","created":"2021-09-21 07:43:17.000000000","updated":"2021-09-23 19:49:26.000000000","submitted":"2021-09-23 19:47:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"810161-debian-mariadb-upstream","meta_rev_id":"2e4094ff12da4932853435417b9a30d45e82bc90","_number":810161,"virtual_id_number":810161,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-23 19:47:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"recommended":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":1,"date":"2021-09-21 11:38:26.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":2,"date":"2021-09-23 18:13:56.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":1,"date":"2021-09-23 18:13:56.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":1,"date":"2021-09-21 09:59:52.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":1,"date":"2021-09-21 11:38:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-21 07:44:28.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2021-09-21 07:44:28.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-09-21 07:44:28.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-09-21 07:47:30.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2021-09-21 09:13:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-23 18:13:56.000000000","updated_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"reviewer":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"state":"REVIEWER"}],"messages":[{"id":"cd5fe43fd411afde46ca1faf6a71d6d1385804d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-21 07:43:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eb8a545f4a2fd4f30ed8abb17d716b73b8f302b0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-21 07:44:28.000000000","message":"Patch Set 1:\n\n\"other\" reno rather than \"upgrade\" as there is no required action of operators (validated).","accounts_in_message":[],"_revision_number":1},{"id":"5e649812d42de0631c433a81b0ca5b640711e95c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-21 07:45:40.000000000","message":"Patch Set 1:\n\nWell, unless they use local mirrors. I will amend.","accounts_in_message":[],"_revision_number":1},{"id":"ba4fec8f67925d3b03afba80cb9088cae352a11f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-21 07:47:07.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"16e629e6be6dafd3f9079cdcddd7e2f9f8a581e8","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-21 07:52:35.000000000","message":"Hashtag added: bugfix-in-stable","accounts_in_message":[],"_revision_number":2},{"id":"2d2f3bc30894425a3f591d534fa6a346bfd05836","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-21 09:13:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/9da2f636a3d54cce94d58d867bded9bf : SUCCESS in 28m 13s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/854cb64d6b594a38b5c7413c2c7b5394 : SUCCESS in 40m 53s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/46967ff01ed74191a308926a88351e08 : SUCCESS in 1h 10m 31s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/42a5de838430472083a0540ebf42c9f3 : SUCCESS in 1h 08m 33s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/86e184885f65476ea4930b614ba69c17 : SUCCESS in 1h 23m 25s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/19031df12ac141d0839d017ae8151771 : SUCCESS in 34m 11s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/16c7b074417a482aa4689c942cd88dba : SUCCESS in 19m 41s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/99655164c5244a078a43132747545e1a : SUCCESS in 58m 44s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f9cdc7e32984665ba707d554877fab7 : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6c8a7a2f009948cd9451d61adbf2e9b8 : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f1bdc05a2714a01a5e8c5bd1826dac1 : SUCCESS in 4m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4bf175802c3f470087e83f8f0592b898 : SUCCESS in 4m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70f89ede61014cee8bdccf39de692ca6 : SUCCESS in 7m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/079092bd75a3416e91e082ed6692e7cf : SUCCESS in 4m 54s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0bf298d0d3334af7a4bd588bf56c38cc : SUCCESS in 21m 19s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9cec39a014784c67b55cad4567c1719b : SUCCESS in 31m 05s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/640252dbfa6e46f8827561a09b906001 : SUCCESS in 1h 00m 47s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/30a505b2fd5d46db9a0d71c32d5b7765 : SUCCESS in 54m 03s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0638e9cf424e4d35a7c4bb5549b6c055 : SUCCESS in 1h 02m 45s","accounts_in_message":[],"_revision_number":2},{"id":"de5b77e4a8d0310d04c7ff5dd01c54c17ce6522f","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-09-21 09:39:21.000000000","message":"Patch Set 2:\n\nWhere is patch bumping Ubuntu to use 10.5 as well? Or will we keep those distributions on different versions (again)?","accounts_in_message":[],"_revision_number":2},{"id":"3bd898ac1613fcf6cf6d3e96ba49020f8b1d813d","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-09-21 09:40:13.000000000","message":"Patch Set 2: Backport-Candidate+1 Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"9a33a022f4fbaea7a1c7d67c86fb055dd1a6c7e0","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-09-21 09:55:19.000000000","message":"Patch Set 2: -Backport-Candidate -Code-Review\n\n\u003e Patch Set 2:\n\u003e \n\u003e Where is patch bumping Ubuntu to use 10.5 as well? Or will we keep those distributions on different versions (again)?\n\nOh, good catch, I also vote for same versions in both dists.","accounts_in_message":[],"_revision_number":2},{"id":"73340e68b273def55d12daee67bf52376deb89f8","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-21 09:59:52.000000000","message":"Patch Set 2: Backport-Candidate+2\n\n\u003e Patch Set 2:\n\u003e \n\u003e Where is patch bumping Ubuntu to use 10.5 as well? Or will we keep those distributions on different versions (again)?\n\nThey are different already and always were. 😊 It\u0027s a topic of the meeting tomorrow (also syncing CentOS). Now let\u0027s fix Debian on all branches.","accounts_in_message":[],"_revision_number":2},{"id":"5c020526c7f559c2ef4d4dda7cd1b2ce5dfbc193","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-09-21 10:05:24.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2: Backport-Candidate+2\n\u003e \n\u003e \u003e Patch Set 2:\n\u003e \u003e \n\u003e \u003e Where is patch bumping Ubuntu to use 10.5 as well? Or will we keep those distributions on different versions (again)?\n\u003e \n\u003e They are different already and always were. 😊 It\u0027s a topic of the meeting tomorrow (also syncing CentOS). Now let\u0027s fix Debian on all branches.\n\nWell, you mentioned in bugfix that it used to fail on Ubuntu too .. So, shouldn\u0027t it be changed also for ubuntu ? (Not because not synced - but because it is in mentioned in bug-report also ?)","accounts_in_message":[],"_revision_number":2},{"id":"e2872b8106af5d4412f8cb59167dc33d89a9e6ee","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-21 10:13:09.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2:\n\u003e \n\u003e \u003e Patch Set 2: Backport-Candidate+2\n\u003e \u003e \n\u003e \u003e \u003e Patch Set 2:\n\u003e \u003e \u003e \n\u003e \u003e \u003e Where is patch bumping Ubuntu to use 10.5 as well? Or will we keep those distributions on different versions (again)?\n\u003e \u003e \n\u003e \u003e They are different already and always were. 😊 It\u0027s a topic of the meeting tomorrow (also syncing CentOS). Now let\u0027s fix Debian on all branches.\n\u003e \n\u003e Well, you mentioned in bugfix that it used to fail on Ubuntu too .. So, shouldn\u0027t it be changed also for ubuntu ? (Not because not synced - but because it is in mentioned in bug-report also ?)\n\nUbuntu already uses upstream.","accounts_in_message":[],"_revision_number":2},{"id":"c632fe43fb8309ddc0fe28fa646b706248dba2ad","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-21 10:19:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/e523a0c9ff4043298b5e4a5208d40d0b : SUCCESS in 1h 03m 27s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/98435372a4814f1899567f83e240ba4b : SUCCESS in 1h 39m 02s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/cc3291e552294635916206b28047ed50 : SUCCESS in 1h 37m 37s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/50cfe8bea4f34c15b239cf4229c0a5c5 : SUCCESS in 6m 37s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3e807bec97b74072be4cf7de940c30e7 : SUCCESS in 6m 50s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/49f22492d0b04395a6cc1457ccc71dd3 : SUCCESS in 1h 02m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"235feee2a40854496d21fbbf1386f548f9f81e27","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-09-21 11:38:26.000000000","message":"Patch Set 2: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"6a2c9eae49ac9f8c3ed787eeca2ce25c20a81402","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2021-09-23 18:13:56.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"826d80f27c2d04a6d18a5e6d001fafb7af41f273","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-23 18:22:41.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"af52335e28cae84abba8855a8a9c9d0dd84c82ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-23 19:47:56.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/cda1e5d133f148e9b8ce921347a11f87 : SUCCESS in 40m 40s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/d6c55206907044ffa7d9cb0ed483cf9f : SUCCESS in 1h 19m 08s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/2bf6e281276a4a1d9e86aeef710c3c0e : SUCCESS in 1h 23m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d78fecfef748442ebd66fdcb70623477 : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/27b22a029eb647df829dd5fa815807bd : SUCCESS in 6m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/51a10158375340a9ae315140cba40608 : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6c7afc12a394e56929a76234effada3 : SUCCESS in 8m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6a824b4498a041e8ab2712db1d7c0a65 : SUCCESS in 6m 14s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/cf14eb6ad27f43a5acd7100a14c7b5d1 : SUCCESS in 34m 57s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/45cc1810f7a8489695f5e85abb3f424d : SUCCESS in 50m 30s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/637a3bafe71a4ae98367a6c563137345 : SUCCESS in 1h 05m 45s","accounts_in_message":[],"_revision_number":2},{"id":"5af640ceea795988b2b64dd62461425f0ec30380","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-23 19:47:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"2e4094ff12da4932853435417b9a30d45e82bc90","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-23 19:49:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d849ca1df2348d1af9dc92df5c12c68 : SUCCESS in 1m 06s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/efb027d8748e45bf9846838d3a9bea5a : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b5422a6bc27797144da1ea973cec24a5cf46d2eb","revisions":{"4460a9b73a33c5f52d2a374debb3c5b0cb548c1c":{"kind":"REWORK","_number":1,"created":"2021-09-21 07:43:17.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/61/810161/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/61/810161/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/810161/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/810161/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/810161/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/61/810161/1"}}},"commit":{"parents":[{"commit":"e998831b9bbcb808effb0c92be5568cbcaf6142f","subject":"Merge \"Upgrade prometheus elasticsearch exporter to 1.2.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e998831b9bbcb808effb0c92be5568cbcaf6142f"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-09-21 07:40:58.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-09-21 07:40:58.000000000","tz":0},"subject":"[debian] Use upstream MariaDB","message":"[debian] Use upstream MariaDB\n\nThis preserves the Debian-provided minor version - 10.5 in case\nof Bullseye in Wallaby and Xena - and will be backported as\na fix to Victoria and Ussuri (where it will follow Buster and its\n10.3).\n\nRelated-Bug: #1944410\nChange-Id: I9704e5371a410cb1f69ec0367174e2b368961dc1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4460a9b73a33c5f52d2a374debb3c5b0cb548c1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4460a9b73a33c5f52d2a374debb3c5b0cb548c1c"}]},"branch":"refs/heads/master"},"b5422a6bc27797144da1ea973cec24a5cf46d2eb":{"kind":"REWORK","_number":2,"created":"2021-09-21 07:47:07.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/61/810161/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/61/810161/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/810161/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/810161/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/61/810161/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/61/810161/2"}}},"commit":{"parents":[{"commit":"e998831b9bbcb808effb0c92be5568cbcaf6142f","subject":"Merge \"Upgrade prometheus elasticsearch exporter to 1.2.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e998831b9bbcb808effb0c92be5568cbcaf6142f"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-09-21 07:40:58.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-09-21 07:46:51.000000000","tz":0},"subject":"[debian] Use upstream MariaDB","message":"[debian] Use upstream MariaDB\n\nThis preserves the Debian-provided minor version - 10.5 in case\nof Bullseye in Wallaby and Xena - and will be backported as\na fix to Victoria and Ussuri (where it will follow Buster and its\n10.3).\n\nRelated-Bug: #1944410\nChange-Id: I9704e5371a410cb1f69ec0367174e2b368961dc1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b5422a6bc27797144da1ea973cec24a5cf46d2eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b5422a6bc27797144da1ea973cec24a5cf46d2eb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
