)]}'
{"id":"openstack%2Fkolla~810212","triplet_id":"openstack%2Fkolla~master~I67cb6037ff13de23ad2eb55e6174e9a338525670","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"I67cb6037ff13de23ad2eb55e6174e9a338525670","subject":"Debian: use Xena binary packages","status":"MERGED","created":"2021-09-21 11:54:53.000000000","updated":"2021-09-27 21:05:17.000000000","submitted":"2021-09-27 20:07:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"810212","meta_rev_id":"b7c4e9c7cbba271da6e1e5e801876f4886dbdf6c","_number":810212,"virtual_id_number":810212,"owner":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-27 20:07:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"recommended":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"all":[{"value":0,"date":"2021-09-27 09:57:56.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2021-09-27 12:01:33.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":2,"date":"2021-09-27 13:03:42.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2021-09-27 18:42:50.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":1,"date":"2021-09-27 18:42:50.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-21 13:13:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-27 11:45:22.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"CC"},{"updated":"2021-09-27 12:01:33.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2021-09-27 13:03:42.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2021-09-27 18:42:50.000000000","updated_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"reviewer":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"state":"REVIEWER"}],"messages":[{"id":"60eff05cd881d5f798d1206c2d68b8ddd3fec198","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-09-21 11:54:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"394f5ddfc2f31e4b27b63bed9522847bdb81ceea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-21 13:13:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/f5b6818d13a34676b29f71462f68a0ac : SUCCESS in 28m 10s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/dff712ace2c642a3867125c52ca33431 : SUCCESS in 43m 43s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/8157af2d8ba645b7a63201ee88b3b9e8 : SUCCESS in 1h 09m 28s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/8c82f13fc5d64d208e01015cebf79f59 : SUCCESS in 1h 05m 56s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/80f88f5d1e0249e7ae15ee01e92830a3 : SUCCESS in 1h 14m 33s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/a63cf0f9ee034f4f9b61df2b94640f5e : SUCCESS in 28m 06s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/29c6872cffeb41f1b1028feb4c8036c0 : FAILURE in 19m 31s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d54c06b9875b4a86a0b80a16d703851f : SUCCESS in 1h 07m 43s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/163fa65f4ed24e7a8e2b6c5a49d5c5ef : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9bb867250c4547fda3693cd982850c9e : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/099607902c0544b18f64091c296b017e : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9272c77931614a138ca244897cffcef1 : SUCCESS in 4m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10bdbbb9fb594e259c31e074ccb5c288 : SUCCESS in 6m 05s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1d0bac6929894b59a3067f06cb21aa17 : SUCCESS in 20m 13s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/059260bf5dfa42548cb3ccbb010760e0 : SUCCESS in 30m 39s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/63b730ffe8ca4da7b3a283c4f3714808 : SUCCESS in 51m 02s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/85d6e3032af84340afdaecd7089815cd : SUCCESS in 50m 13s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/52fe45df8fc547baa70bb64075e7dfdf : SUCCESS in 1h 00m 19s","accounts_in_message":[],"_revision_number":1},{"id":"fe3356833d170f42a68eaad7c6af9e809b08f251","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-21 13:52:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/d7aa2566889549cdbcef3c95e3c6cae5 : SUCCESS in 55m 29s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/efb6217202124468bf1b658c9b8b7e7a : SUCCESS in 1h 03m 18s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b5bf94dac001498a8f21bb7ad9995f46 : SUCCESS in 1h 50m 22s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1b5df56dcecb44039c453349c65d5577 : SUCCESS in 5m 00s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4155028b512b4a95bf0509fc98f8f274 : SUCCESS in 4m 29s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/5469567296754e5cac214dda54d3a20f : SUCCESS in 34m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"64b53b11bf94c0b79b80ab4c784dc3432adbc62b","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-09-21 14:54:33.000000000","message":"Patch Set 1: Code-Review-1\n\nneutron fix needs to land in repo","accounts_in_message":[],"_revision_number":1},{"id":"0da62aeb59f6fcd98094a5566ebb60539d4ac6b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-09-27 09:37:21.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"4cd67a7e0c4efc4b01d28497fea5e2338c3d6f57","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2021-09-27 09:57:56.000000000","message":"Removed Code-Review-1 by Marcin Juszkiewicz \u003cmarcin.juszkiewicz@linaro.org\u003e\n","accounts_in_message":[],"_revision_number":2},{"id":"8afe163ecbbee3947cbd8be893e94e1c6ca1ddc0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-27 11:06:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/21a06ade364b4286ab30fd9acb33c0f1 : SUCCESS in 1h 01m 10s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5d06024b5c3345b7877e3767ebf86d69 : SUCCESS in 45m 26s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/34a5e11f0291474cb8a2832e72d6e7c1 : SUCCESS in 1h 22m 11s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0573764733bd48bea9e5e4e2317d2517 : SUCCESS in 4m 51s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/809b81cecae449ac94685f54cf4833b7 : SUCCESS in 5m 31s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/e65c46a24ab64093be61224eb7d1766a : SUCCESS in 51m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c21cf085a3a3f2d52080fb5ed3c41a95b6f88d69","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-09-27 11:45:22.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 1: Code-Review-1\n\u003e \n\u003e neutron fix needs to land in repo\n\nThis is already fixed in repo, everything is working locally.","accounts_in_message":[],"_revision_number":2},{"id":"744f66a14f3dc2ab2888fcbc739a4372fd268707","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-09-27 11:45:53.000000000","message":"Patch Set 2:\n\nLet\u0027s wait for CI status for kolla-build-debian-binary","accounts_in_message":[],"_revision_number":2},{"id":"6edc22be55e6da40ae74363256e645f46ca7776f","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-09-27 11:45:59.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"aa2ed4ceb5f86637fd98268468c81c62b8ae0d23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-27 12:01:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kolla-build-centos8s-binary https://zuul.opendev.org/t/openstack/build/88a48db3e3c444f095482f537668b6a3 : SUCCESS in 24m 27s (non-voting)\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/01260effac4943aa8db923f77148bd71 : SUCCESS in 37m 13s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/00748932daf84c7bad248baf093e47b5 : SUCCESS in 1h 15m 03s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/9b9e4ae8bc6640f387fb23f394eee882 : SUCCESS in 1h 50m 54s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/aa6ce88faa8b4ddd9d87341b36c2cff5 : SUCCESS in 2h 22m 25s\n- kolla-build-debian-source https://zuul.opendev.org/t/openstack/build/4b2cd0ad84e14d5d8381f6bc852831b9 : SUCCESS in 48m 05s (non-voting)\n- kolla-build-debian-binary https://zuul.opendev.org/t/openstack/build/9be224d370ec4979941b64bad0c3245a : SUCCESS in 39m 14s (non-voting)\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c5c7211dfb7f4902b8c5310b930a233c : SUCCESS in 1h 07m 20s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d35e0fa1a0c04568b68fba256872d2d9 : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b09ad3a5746b4592aa1a9b4ca056611f : SUCCESS in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a8d1be933ce64c39a7d609912a75d7aa : SUCCESS in 5m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0f59fe121c2545ee908a8b4da9234426 : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33363b41825f43a1ba5298641e0cbbc3 : SUCCESS in 8m 53s\n- kolla-build-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1d150a9a867747c8892a723cee1cd7cb : SUCCESS in 26m 18s (non-voting)\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/548b978d7f0046748a6e1b8c94659115 : SUCCESS in 40m 07s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0d891da86cd0492abe75597e60f75a80 : SUCCESS in 54m 54s\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/01bdfc9ea86f41ca97e72f5fe0d8e1b5 : SUCCESS in 52m 08s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4d96915037d94358b8bdca15d53bf882 : SUCCESS in 1h 09m 06s","accounts_in_message":[],"_revision_number":2},{"id":"eaf441798bacada31d0e102b5506b9000ca5495a","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-09-27 12:01:33.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"788c213357bbff7e56d74729f5b2eb9fc277f8e6","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-09-27 13:03:42.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"641bc07c69f55e01bab9ac7318ad426c3e086121","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-27 13:46:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-build-centos8s-source-aarch64 https://zuul.opendev.org/t/openstack/build/ec8d6730a5484612aa8fe0bd331a8929 : SUCCESS in 50m 03s (non-voting)\n- kolla-build-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f3485da84156467294e60a3cad0e05d0 : SUCCESS in 1h 25m 35s (non-voting)\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2efa0ddc566e496e95ecbc4a0e012959 : SUCCESS in 1h 58m 22s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0006b565de9a4c3eabfcbcb00b0a1dec : SUCCESS in 7m 24s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/352ba520a5ac430695b91c169de52007 : SUCCESS in 7m 47s (non-voting)\n- kolla-build-ubuntu-source-aarch64 https://zuul.opendev.org/t/openstack/build/e7f3a60be06f4f57b365d750cefd348a : SUCCESS in 1h 24m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"71149b4a4e5c726d0c5d3541fa7b6b248c53899a","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2021-09-27 18:42:50.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"11a00a44cfd5d929fb8b9b959fdc81a0bc7cb48a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-27 18:44:59.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"f04d3ac2fdb914b941c9c9e88a9e72036166fd3c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-27 20:07:23.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- kolla-build-centos8s-source https://zuul.opendev.org/t/openstack/build/2dee81e908eb48b1a201b77959ec6e62 : SUCCESS in 32m 45s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/3d41c434b2ea4027a6eaaf434d373c83 : SUCCESS in 1h 05m 03s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/eb0c30a2e94f421ba41cb11fb603961a : SUCCESS in 1h 19m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25ed650cb4524bcda2534cdd72b6bbdf : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5f18cdeb5b2b4a6e996f38847016edb3 : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/69a2c465bcac47f2bc8137565ba7871c : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/269f257b3d5e49d5ba3ad08d0d884138 : SUCCESS in 9m 22s\n- kolla-build-ubuntu-source https://zuul.opendev.org/t/openstack/build/9fa8c0de954948518d9a09e71742168a : SUCCESS in 29m 00s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7de7db28439e43cb9147296acc7d7136 : SUCCESS in 46m 07s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/736ade3a3814404d9ea190905fd756d7 : SUCCESS in 1h 01m 34s","accounts_in_message":[],"_revision_number":2},{"id":"3e1f866d499510061787209383a091da3012d393","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-27 20:07:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"b7c4e9c7cbba271da6e1e5e801876f4886dbdf6c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-27 21:05:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/877f615e0c3c40f88dc705d8a1bd8062 : SUCCESS in 1m 34s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"623f375b5828e6792c93593501e6b7eb85f1f53a","revisions":{"bb1a9b7fd1907d44817f94e5e3b045b33e5e1585":{"kind":"REWORK","_number":1,"created":"2021-09-21 11:54:53.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/12/810212/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/12/810212/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/810212/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/810212/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/810212/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/12/810212/1"}}},"commit":{"parents":[{"commit":"d2bc56031941d3d46d091c6d8535d025bc30bfdd","subject":"Merge \"Fix stale commented-out repo entry\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d2bc56031941d3d46d091c6d8535d025bc30bfdd"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-09-21 11:54:23.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-09-21 11:54:23.000000000","tz":120},"subject":"Debian: use Xena binary packages","message":"Debian: use Xena binary packages\n\nDebian OpenStack team provides Xena packages now.\n\nChange-Id: I67cb6037ff13de23ad2eb55e6174e9a338525670\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bb1a9b7fd1907d44817f94e5e3b045b33e5e1585"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bb1a9b7fd1907d44817f94e5e3b045b33e5e1585"}]},"branch":"refs/heads/master"},"623f375b5828e6792c93593501e6b7eb85f1f53a":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-09-27 09:37:21.000000000","uploader":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"ref":"refs/changes/12/810212/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/12/810212/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/810212/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/810212/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/12/810212/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/12/810212/2"}}},"commit":{"parents":[{"commit":"1dc1041cb6b0fe8f11bfe852a840dea6fdebb13f","subject":"Merge \"[debian] Use upstream MariaDB\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1dc1041cb6b0fe8f11bfe852a840dea6fdebb13f"}]}],"author":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-09-21 11:54:23.000000000","tz":120},"committer":{"name":"Marcin Juszkiewicz","email":"marcin.juszkiewicz@linaro.org","date":"2021-09-27 09:37:21.000000000","tz":0},"subject":"Debian: use Xena binary packages","message":"Debian: use Xena binary packages\n\nDebian OpenStack team provides Xena packages now.\n\nChange-Id: I67cb6037ff13de23ad2eb55e6174e9a338525670\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/623f375b5828e6792c93593501e6b7eb85f1f53a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/623f375b5828e6792c93593501e6b7eb85f1f53a"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
